)]}'
{"id":"openstack%2Fdesignate~754998","triplet_id":"openstack%2Fdesignate~master~I8196e545af977780a0b9967873b35483523644cc","project":"openstack/designate","branch":"master","topic":"update_requirements","hashtags":[],"change_id":"I8196e545af977780a0b9967873b35483523644cc","subject":"Update requirements","status":"ABANDONED","created":"2020-09-29 13:08:49.000000000","updated":"2020-09-29 14:41:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"479817762ff3d1da512592dfd3bb3246cd0159e9","_number":754998,"virtual_id_number":754998,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-29 14:16:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":-2,"date":"2020-09-29 14:41:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 14:16:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b117bb8f79ce92098bd628fbf5053a3b794ef16b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 13:08:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"742913e5e757b1de5a4484ba80233382e9c4af23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 14:16:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e110f0a59c0942f9bef4b31b9402342b : SUCCESS in 4m 47s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1d61c8c6167b4d0e85794c182fd532a9 : SUCCESS in 2m 37s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1585a2b1f7ba45f9bd9b6361f17902f0 : SUCCESS in 53m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5421545293b44e0daa80731d3110405b : SUCCESS in 47m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d3e0949e07f2451da2732c9632f51af7 : SUCCESS in 1h 04m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/462167bab78a4224ab821ed9471b4e20 : SUCCESS in 40m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9912b51c31bd4c9d94b315fa102313ba : SUCCESS in 43m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbe0a177c59e4bc6b1b85b85687d0879 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc43e12b74544eeebca98e5404d00fbd : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/313dfae734e04c74bc6ccb861bd7eb4a : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e52f5747c6d84efaa81ae1d42b99d254 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09c46b6f6b644c439a3a8d26fbb97b20 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ec32036d9f43c8a3b92ca000f0176b : SUCCESS in 7m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c1200e341784f33b30ce586c9688846 : FAILURE in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23dbeedf92464039af0de10b5fc0ffc8 : SUCCESS in 3m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/52f530c8d3464035804697dd6b2783e1 : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e178d5cc2ed2f5c7fad104c95adcfb7e6ee5859c","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 14:41:22.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"e8981767919269aa7a1a80abc13b547235c34bcb","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 14:41:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c38d9595d6c0f13eac68c70f575edfb862192e0","revisions":{"2c38d9595d6c0f13eac68c70f575edfb862192e0":{"kind":"REWORK","_number":1,"created":"2020-09-29 13:08:49.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/98/754998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/754998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/754998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/754998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/754998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/754998/1"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-29 13:08:29.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-29 13:08:29.000000000","tz":-360},"subject":"Update requirements","message":"Update requirements\n\nChange-Id: I8196e545af977780a0b9967873b35483523644cc\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c38d9595d6c0f13eac68c70f575edfb862192e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c38d9595d6c0f13eac68c70f575edfb862192e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
