)]}'
{"id":"openstack%2Fdesignate~755126","triplet_id":"openstack%2Fdesignate~stable%2Ftrain~I500c932a502db24332aad95c54713b7056bc1df4","project":"openstack/designate","branch":"stable/train","hashtags":[],"change_id":"I500c932a502db24332aad95c54713b7056bc1df4","subject":"fix database sync always get the wrong  value","status":"MERGED","created":"2020-09-29 20:54:17.000000000","updated":"2021-01-14 14:28:57.000000000","submitted":"2021-01-14 14:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755126","meta_rev_id":"0a14680601c5e4761a22eb1e702327acdf526027","_number":755126,"virtual_id_number":755126,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 14:27:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-14 13:27:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-11-29 14:44:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2021-01-08 06:00:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2020-11-19 06:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-01-14 13:27:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-29 20:54:17.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 20:54:17.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-09-29 22:12:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 06:25:15.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-11-26 09:50:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-01-06 14:00:54.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-01-08 06:00:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-01-14 13:27:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"bcff36f58d4249fd6be9a9a222464d4367f5b584","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 20:54:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d33e1331a1a6a394ba97913018482ce8781b1704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:12:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8fe0ebdb703d4c9d9018d746d805f772 : SUCCESS in 45m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c17c419c0abe46889e031591e8896467 : SUCCESS in 46m 16s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/f866f3959d0d4935b6f509c2ca9e2f99 : SUCCESS in 42m 50s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/a2824b08b43c4ddbbf9d1b6e9e1387bb : SUCCESS in 40m 27s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8a862cecd8b147efb8459082c1841f8d : FAILURE in 24m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0fc2d4635e244e209e85ef1920c72ba3 : SUCCESS in 1h 03m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b12b86306aba4e4793caebd77dda2fdd : SUCCESS in 43m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a8da65ef875d4c81a90e5b7d33984f8a : SUCCESS in 41m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fae2477a00114e558f1960ba7670eb14 : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bc1194c87a74e36a9a74f2d774ebda0 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/478be0c6ef264e9180f4492dfd70a82c : SUCCESS in 6m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3cb30b9cd66944cf8c2ba78b0e128656 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e19f89bd4184e60b83b3c4a57d5f691 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/746ced60fd5c4411acae20fa0331a04e : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac082f393e449a3b2208b6bc995c0a6 : SUCCESS in 6m 43s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/67ec19f2a88a424abcfa7b70b8ba4dc7 : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bafba8c8aa32652a9894d4bdf2022d5667216867","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-19 06:25:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b144c8579923b888805a2bfc284d6782a46c65c0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-11-26 09:50:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b3e996093978fd66f0071299435fdf73a3d56849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 10:46:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f5074fb6d60c4cf4a7440b0cf574098a : SUCCESS in 46m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b23d19f0fb0f4aca94ab6234cc7bfa23 : SUCCESS in 38m 19s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/4d937b2acb2845abba40b306d8e346c5 : SUCCESS in 44m 25s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/53a5a4c148744aa19ba13a0c39f6b7ef : SUCCESS in 40m 22s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/961dbfb9a3e249219eaf03030aa74677 : FAILURE in 23m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3e1927e030c84ad896823f92ef668054 : SUCCESS in 51m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b0c1ff39d7754008be34f40be8c13da2 : SUCCESS in 43m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1e8338b5834b4cfaac868c45f2e62b0c : SUCCESS in 36m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a701a30083614343bca998e87c267970 : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57fc17bac42b46a2bccaa97bca44a21a : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b79f9d9e6b45a7894f8aab549969de : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96683e5b80a144ee84a95e242e8eb7b6 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b58aeb54ff1d403981e609768e2ad8cc : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/880cc1d251544fde9448920703526d65 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631d2cb2904d4f16aea4fea9afbf28ba : SUCCESS in 8m 00s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/800dfbbb53f74d4a90a7a26be5935eef : SUCCESS in 39m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7bb87117ce50ecf08b755d8cc97e03d6d38d18b3","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-29 14:44:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"40448d11a1973ea3b0347932cc45586eb94c6172","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-06 14:00:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a78f5d684d050afc047ff5273f9fb864769fb87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 15:14:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/54eda867413a474db74c44f37120cb8a : SUCCESS in 32m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0394ac00697544e38084819a5b1fb98a : SUCCESS in 55m 26s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/dbc8a792312e4dff8813e05672a8325e : SUCCESS in 40m 33s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/8c26d0d8890b4a738737573ea6dc30e2 : SUCCESS in 54m 18s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/2c4bb426ff9e43c88c7bb3b3b114c776 : FAILURE in 22m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0d2d701408124b6b9ceb6b574bc4a2a8 : SUCCESS in 1h 10m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/19604adc20b04b9a86bb38486830971d : SUCCESS in 36m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2b0eb887f3774cb9a23977a7c1fdf119 : SUCCESS in 50m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bcedd132a334c6b88ed3ccd59f3491a : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8abab82c17040f5a7aa7f15a7fc2a37 : FAILURE in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527938bdd6ce44fabf8430864dae508d : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/529962156bfe442aa50b8de447dce704 : FAILURE in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b11846b12d2c4090b3c8685d7df4efe7 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80f4ec7d94d94f47a1badbd77e3af07c : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab7571c269b4133974f860c8ba31563 : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/80d9685914a840979a1e9292e6799cc5 : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d5113456aa0f8d3cde6a6493c2a4e2a806ca29d2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-07 18:15:05.000000000","message":"Patch Set 1:\n\nrecheck - as the gate is now fixed","accounts_in_message":[],"_revision_number":1},{"id":"b4e95d479cee80dfb13d050136f2837942658445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 19:27:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/53594e64d0e547d283acedf42ee5362c : SUCCESS in 32m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/226a2de864b141ad801fbc613face09a : SUCCESS in 32m 50s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/5130c6e16819470e9049de116552cb38 : SUCCESS in 41m 04s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/e4a22aee6772443b995bec8b8afa788b : SUCCESS in 32m 27s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/12adbfd94e6646379db08b16854862b3 : FAILURE in 27m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d80df327c82d43879428e26cc57ac12f : SUCCESS in 1h 02m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1c25ed6735524a7ea25cac6ff18e3301 : SUCCESS in 36m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e47295f8625b45698f6e49116a3f7c08 : SUCCESS in 44m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80284aa039074a26acac66e8d81a8060 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1837978ba9b4892845715ccc73d2532 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a82fe5b58f54c8089cc5b5efb0ed15c : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e454fa1b57b24b07938fd93879e3210e : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06a2e4c73acd492dab4c1379d4608895 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d58e811dd0e412388996c12b3390eb2 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c2d05b609b6492fbace0e03a2f40fdd : SUCCESS in 8m 38s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/c2ca91f4d25545ea9fe40b2b437706f4 : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":1},{"id":"06a11f4a79ff49369290688b4670b3dd9dd534dd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-01-08 06:00:52.000000000","message":"Patch Set 1: Code-Review+1\n\nI\u0027m actually not sure if this patch should be backported and would like to see feedback from the stable team. The fix is changing the behavior in an incompatible way, but on the other hand unless I\u0027m mistaken it only affects the initial deployment, and it doesn\u0027t make sense for anyone to be relying on the previous, broken behaviour.","accounts_in_message":[],"_revision_number":1},{"id":"8d093b4cff4313f9048d18fc58f1121f1cbf0270","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-14 13:27:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n\u003e I\u0027m actually not sure if this patch should be backported and would like to see feedback from the stable team. The fix is changing the behavior in an incompatible way, but on the other hand unless I\u0027m mistaken it only affects the initial deployment, and it doesn\u0027t make sense for anyone to be relying on the previous, broken behaviour.\n\nGood question. That is definitely a concern with a stable backport. I agree though, since this is really only going to impact initial deployment, I think in this case it is safe to backport it.\n\nIt\u0027s also one of those areas where it is actually fixing an issue, so the choice is between strictly following stable guidelines, and leaving things broken, or backporting and introducing a potential operator concern.\n\nIt probably would have been better if this included a release note or some type of docs update. But overall, I think this looks safe enough and addresses an issue, so worth getting backported.","accounts_in_message":[],"_revision_number":1},{"id":"41a6a8463717310679fb4dd0df51a6c28385a7a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:28:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02f7681ee44b3fd0ffee1dc1e67cf8c0c8bc2138","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 14:27:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/334bd8497ac04085be1e626c7f949daa : SUCCESS in 49m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/91ea86db3f0e481d94cf7d465ee64932 : SUCCESS in 35m 37s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/2d7f19bf33524f9f9343e43542de3bd3 : SUCCESS in 38m 51s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/2142d9c0e92f4c3580b0df375d4a4250 : SUCCESS in 37m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/925e14488e5348a480f8988f12f9e3e1 : SUCCESS in 54m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cf9ccbbb54484b9bbf1cdf229eee419b : SUCCESS in 31m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/96568bf82e0b457ab4bad664e8de40fe : SUCCESS in 38m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6377519e831645dc9d4701af64f0ccfc : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/846a7a61bd8648e2b96ad5a3f697f6de : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95aef2fe21c049fd87a76bd69295963c : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31d4ed0148aa4b779a0d897208fb5277 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a059446536544029f7f9f9c8f159aa6 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782fa50285e542bd8250cd3e297da29a : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b212ab9c768f2a21e5fa8527f12cc618289b3046","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 14:27:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0a14680601c5e4761a22eb1e702327acdf526027","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 14:28:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b1cede404a418dbcc8539ac9a475d0 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b47d4306a0979b2a7801d2be03e566b1e1efaf80","revisions":{"b47d4306a0979b2a7801d2be03e566b1e1efaf80":{"kind":"REWORK","_number":1,"created":"2020-09-29 20:54:17.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/26/755126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/755126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/755126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/755126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/755126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/755126/1"}}},"commit":{"parents":[{"commit":"4fc04ee416d13f7e4e27c6b522b42a3c69c271b5","subject":"Merge \"Adding distributed locking to central\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fc04ee416d13f7e4e27c6b522b42a3c69c271b5"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-09-11 10:57:49.000000000","tz":480},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-29 20:54:17.000000000","tz":0},"subject":"fix database sync always get the wrong  value","message":"fix database sync always get the wrong  value\n\nwe create a default pool  when exec database sync, and get the\ndefault_pool_id from designate.conf, but we did not register this\nopt before using  it, so we always get the default value which is\nwrong\n\nChange-Id: I500c932a502db24332aad95c54713b7056bc1df4\nCloses-Bug: 1881277\nRelated-Bug: https://bugs.launchpad.net/kolla-ansible/+bug/1855877\n(cherry picked from commit 609437e73deff55e9f5548538f093d01f0addd1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b47d4306a0979b2a7801d2be03e566b1e1efaf80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b47d4306a0979b2a7801d2be03e566b1e1efaf80"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
