)]}'
{"id":"openstack%2Fdesignate~755127","triplet_id":"openstack%2Fdesignate~stable%2Fstein~I500c932a502db24332aad95c54713b7056bc1df4","project":"openstack/designate","branch":"stable/stein","attention_set":{},"removed_from_attention_set":{"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2022-09-15 00:17:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I500c932a502db24332aad95c54713b7056bc1df4","subject":"fix database sync always get the wrong  value","status":"ABANDONED","created":"2020-09-29 20:54:32.000000000","updated":"2022-09-15 00:17:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7ae1bd3338f5af65c595d07853ce6988b328d6f6","_number":755127,"virtual_id_number":755127,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-13 16:57:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 20:54:32.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-09-29 22:04:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-13 15:43:10.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"}],"messages":[{"id":"8c22b88b8e5ee4475740dfe3f3b9b7087efcabeb","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 20:54:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d0d56d080c7b2aa02d3781ed6983134ecfde4dd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:04:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a5a6952f11004a80a91163be034a1144 : FAILURE in 31m 12s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/f89f8cb371b64ed0a85d9fa52aaf5d69 : FAILURE in 20m 43s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/85cff0543b844ef2a4c7a6610e1875d0 : FAILURE in 26m 06s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/d58c20c741c740b0a8353446777e0d5f : FAILURE in 21m 41s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/0c46cc68358345da91012928374cde4b : FAILURE in 26m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ca6c0e518d6c4b4ea74037604bec18c4 : SUCCESS in 54m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3187a029189c4424bcac38cbcb307715 : FAILURE in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d830d0dbdb014528b1fc39a629940b1a : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82007a5fd4d445e489ff52bba13da7ce : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6712631acd124705868adee3c94e4487 : FAILURE in 4m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/732dbd59d5a2441da5d25c19a5214f08 : FAILURE in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d7d73406ba04d52833a1761a9ada481 : FAILURE in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c215676c3b0b4e19ac48f47c22ea2c54 : FAILURE in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77733b188af740f49512da46f8700c2c : SUCCESS in 4m 46s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/c4eeff0b613542a281c518a3673bf035 : FAILURE in 24m 01s","accounts_in_message":[],"_revision_number":1},{"id":"42111b9feeb944043378260d534eea78f3bb95e5","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-01-14 14:31:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0e954199377b96a5ec3bf3e84b594168cd75a418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 15:28:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/92d735c9e78e4ebbbe88d37bbbdbfa24 : FAILURE in 30m 00s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/de241ade92064075bd6457cf9d020855 : FAILURE in 26m 01s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/582fa2d60ba9452f90eadc71856ee99d : FAILURE in 29m 13s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/c874b11e16bf4e6e9eb58e35ae018ed4 : FAILURE in 24m 55s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8362dc4fbf6549e6955920f39fb66dc6 : FAILURE in 26m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c04b1f50d10f4edba22f99ba957cab3e : SUCCESS in 54m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06b81d531c4e470e9c6a42b7e5738f9c : FAILURE in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/759423dbe42f43389187467ccd730b1c : FAILURE in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/038547839ad94731b71500dbfc868651 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8910e07f15348af86211e5f0932e6ff : FAILURE in 2m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/21d9893fd1d145b994359d38383cdf0d : FAILURE in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1d6638dfff6457ea158ddd9d5b6a97b : FAILURE in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e149819cde14e74b07c8928a9301cb6 : FAILURE in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/123fc50ea7fd4b17a69e73f0027cd79a : SUCCESS in 6m 19s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/0f4820e0a282419aa0a81b927b01ce59 : FAILURE in 22m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8b1c60478d68cb4fab2cf7bf98bd15aff0e3ae69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-27 17:33:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cb453759e321bcbaf1b111813787da3aa4149da1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 18:36:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/67e4f369b53a4a6da4807cc0913776a5 : FAILURE in 28m 05s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/ac7133fbf9cb4e5ca76ed60223cfe776 : FAILURE in 27m 06s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/dc15a83131934d198429cb0388ab243b : FAILURE in 28m 00s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/a9480fc3d9674be38ac2d6679ca2ff8b : FAILURE in 27m 21s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/790b5287bf8841ce85cc6370f03271c3 : FAILURE in 27m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/64cd5a3a0ab3403bb9ade7f19b76b4d7 : SUCCESS in 56m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3985b0b4500c4f03aa21e370959e083e : FAILURE in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36556a8f056b47058df25dde8c85b1bd : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/414b70828f354f84802754e64c9e6978 : FAILURE in 4m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0a67691d044246b0a805330d66517b96 : FAILURE in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20c977163f0b40cebc0d9d992700aad5 : FAILURE in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cbf42ae288a4ad185aa6ea920bf2ac7 : FAILURE in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a287cbdc3c514d20852ad6609ac55af7 : SUCCESS in 4m 49s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/f7d4c541b4d94fc6876e108e3b302076 : FAILURE in 25m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3326e0997391b8dc86e2aa7100f48bc8ef701206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-05-13 15:43:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"df988d5facd23139cffdad691329d7bdc778dc99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 16:57:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1206cc9917f9417ba03adb3fa87bfed5 : FAILURE in 36m 05s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/812776975ba34160a2d698017f0c7d50 : FAILURE in 34m 02s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/684fa995c71e4dfb98e96b799055b2c8 : FAILURE in 31m 41s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/25c5ebff6bb640e3b0e583ec71ee4d06 : FAILURE in 34m 59s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/12aed49f4f764fa5afc1bcbd25d1ae1c : FAILURE in 31m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6c398a446bc44eabb347af275650d503 : SUCCESS in 1h 07m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5387c95ac6094664aba0b127c5064551 : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67494c13ebe240289cdf3dcd48799fb6 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8d1dd6b841d469898e7aa5f3924ab6e : FAILURE in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b7eec280a1f24fb385420bc881f9e1da : FAILURE in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac0f1d4e9b3a4faabc39486752a75f46 : FAILURE in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0c17b4f930c4ef5a034cbe60a5d2995 : FAILURE in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/438b768c4ffd4f738e733e1ac6484f25 : SUCCESS in 4m 13s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/ec1b5af0edc1439692f489991be90f90 : FAILURE in 31m 35s","accounts_in_message":[],"_revision_number":3},{"id":"7ae1bd3338f5af65c595d07853ce6988b328d6f6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-15 00:17:14.000000000","message":"Abandoned\n\nI don\u0027t think we have working nodepool images for this, so it can\u0027t be tested/merged anylonger. Restore it if you need it.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f496a6e5fbbd52d36a6bce9b24a04f51c04d0027","revisions":{"9d08244fbd29aa79dc11ae4100de787e95aec50e":{"kind":"REWORK","_number":1,"created":"2020-09-29 20:54:32.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/27/755127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/755127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/755127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/755127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/755127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/755127/1"}}},"commit":{"parents":[{"commit":"6307c84e83530ab27124ab9e6a8117ee44069614","subject":"Make greande jobs n-v for EM and oldest stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6307c84e83530ab27124ab9e6a8117ee44069614"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-09-11 10:57:49.000000000","tz":480},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-29 20:54:32.000000000","tz":0},"subject":"fix database sync always get the wrong  value","message":"fix database sync always get the wrong  value\n\nwe create a default pool  when exec database sync, and get the\ndefault_pool_id from designate.conf, but we did not register this\nopt before using  it, so we always get the default value which is\nwrong\n\nChange-Id: I500c932a502db24332aad95c54713b7056bc1df4\nCloses-Bug: 1881277\nRelated-Bug: https://bugs.launchpad.net/kolla-ansible/+bug/1855877\n(cherry picked from commit 609437e73deff55e9f5548538f093d01f0addd1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d08244fbd29aa79dc11ae4100de787e95aec50e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d08244fbd29aa79dc11ae4100de787e95aec50e"}]},"branch":"refs/heads/stable/stein"},"fadf048782bb878b74569f1e0d80113dc32d4d71":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-27 17:33:47.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/27/755127/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/755127/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/755127/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/755127/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/755127/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/755127/2"}}},"commit":{"parents":[{"commit":"acc82755f1e8df8bb8f0ab7b4524f691637a312e","subject":"Adding distributed locking to central","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/acc82755f1e8df8bb8f0ab7b4524f691637a312e"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-09-11 10:57:49.000000000","tz":480},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-27 17:33:35.000000000","tz":-420},"subject":"fix database sync always get the wrong  value","message":"fix database sync always get the wrong  value\n\nwe create a default pool  when exec database sync, and get the\ndefault_pool_id from designate.conf, but we did not register this\nopt before using  it, so we always get the default value which is\nwrong\n\nChange-Id: I500c932a502db24332aad95c54713b7056bc1df4\nCloses-Bug: 1881277\nRelated-Bug: https://bugs.launchpad.net/kolla-ansible/+bug/1855877\n(cherry picked from commit 609437e73deff55e9f5548538f093d01f0addd1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fadf048782bb878b74569f1e0d80113dc32d4d71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fadf048782bb878b74569f1e0d80113dc32d4d71"}]},"branch":"refs/heads/stable/stein"},"f496a6e5fbbd52d36a6bce9b24a04f51c04d0027":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-13 15:43:10.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/27/755127/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/755127/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/755127/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/755127/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/755127/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/755127/3"}}},"commit":{"parents":[{"commit":"0b5634643b4b69cd0a7d5499f258602604741d22","subject":"Update zones masters using pool target masters.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b5634643b4b69cd0a7d5499f258602604741d22"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-09-11 10:57:49.000000000","tz":480},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2022-05-13 15:42:51.000000000","tz":-360},"subject":"fix database sync always get the wrong  value","message":"fix database sync always get the wrong  value\n\nwe create a default pool  when exec database sync, and get the\ndefault_pool_id from designate.conf, but we did not register this\nopt before using  it, so we always get the default value which is\nwrong\n\nChange-Id: I500c932a502db24332aad95c54713b7056bc1df4\nCloses-Bug: 1881277\nRelated-Bug: https://bugs.launchpad.net/kolla-ansible/+bug/1855877\n(cherry picked from commit 609437e73deff55e9f5548538f093d01f0addd1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f496a6e5fbbd52d36a6bce9b24a04f51c04d0027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f496a6e5fbbd52d36a6bce9b24a04f51c04d0027"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
