)]}'
{"id":"openstack%2Fdesignate~755138","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I90af3a092296f943509833608f25522b6f8e9ab6","project":"openstack/designate","branch":"stable/ussuri","topic":"fix-pdns-stable/ussuri","hashtags":[],"change_id":"I90af3a092296f943509833608f25522b6f8e9ab6","subject":"Fix pdns4 devstack plugin","status":"MERGED","created":"2020-09-29 23:03:21.000000000","updated":"2021-02-17 00:40:02.000000000","submitted":"2021-02-17 00:38:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755138-fix-pdns-stable/ussuri","meta_rev_id":"08ceff93b50353bc02140e699206b4b125f5a51c","_number":755138,"virtual_id_number":755138,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-17 00:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-16 23:14:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2021-02-16 23:30:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 23:03:21.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-09-29 23:25:12.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-09-30 00:56:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-08 15:57:34.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"}],"messages":[{"id":"81a25ba155ff310c7ad6cb86ee8a3b8401b28989","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 23:03:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"946386c7a2bd01f3f74490c660a59655916a74d1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-29 23:25:12.000000000","message":"Patch Set 1:\n\nThis needs to be backported to Victoria as well.","accounts_in_message":[],"_revision_number":1},{"id":"50564950d55f8206c7ace85ddfb8e3c47fca5ee0","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-30 00:16:03.000000000","message":"Patch Set 1:\n\nDone","accounts_in_message":[],"_revision_number":1},{"id":"cc8d4dbce55f256cc9211f84e67fe69d4073f425","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-30 00:17:07.000000000","message":"Patch Set 1:\n\nWhat about further back? Should we backport to earlier than Ussuri?","accounts_in_message":[],"_revision_number":1},{"id":"13b9398689ef5b8d46531580470b11742de18341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 00:56:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/83eb6af16cce4ba28d46ab05ceb0fa16 : SUCCESS in 40m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b17ea318a3fb432088f9c02d725cdb75 : SUCCESS in 40m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/070198bff9694953895f2978f73bf923 : SUCCESS in 57m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/82bde47549884143b7c5ef3ce579928f : SUCCESS in 40m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3e502c60691f43b08e4b86e876d08bcc : SUCCESS in 31m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/678c13c2d2cd47009d6e48bc9257a8c9 : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9603baf1026e45edb92bd0d410df3bc0 : FAILURE in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f38772d084534ba4a3caa102d9195e25 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1853122f7e424c828fa4c46daabd05a8 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c37120fd7cb4fba98e88793bd5653ea : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e3a810c9a984b22a997a14752766488 : SUCCESS in 9m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f87f24a3ae64120886fe3dd1389e279 : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cf5f747224d640a298b94e918ccf0405c4fd5f89","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-30 15:48:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6de6323a99e46125e05909c46e5d9e28742f46df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 17:09:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/df4ec2d289cf487da5897fae2e396326 : SUCCESS in 35m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c6a4b2702ecf46cf9c14c47a52c5dc0a : SUCCESS in 35m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/78bf8cb3067a4ac58219f959a14cbedf : SUCCESS in 1h 12m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1b8bb6c3eb5c46a59ecea26ff4a0b060 : SUCCESS in 35m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cd14dc9562c9448692d5ed805293990b : SUCCESS in 35m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b79d77e72c444b5989a5743adf17552 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28e61c80da464a9188874cbfd1d43b69 : FAILURE in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b6aeb3e18ab408ab3b0003d6dda3790 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c5603e033bf4129a5c91a837e354b5d : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31a9689e0bb941dab5d688b63cea915e : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f61738f7138467ca51c1d01f9cf232d : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/daa5e841fce74089a84699ce18713e68 : SUCCESS in 46m 42s","accounts_in_message":[],"_revision_number":1},{"id":"514b7f8ada0c40ce9ad89b953c8636492dce8c2f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-08 15:57:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2413ed15223b796308b726d352648d3b3fbb9497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 17:16:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d4d5044b740d45f0b692469b385550d9 : SUCCESS in 37m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5a60fed7f8e2474785b00f4fca83d179 : SUCCESS in 35m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bb0e0ac6266b4f209120b3442774c6b8 : SUCCESS in 1h 04m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c90a53e31bff44238cbc9eecb48c2403 : SUCCESS in 34m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b68d216e63c3443184fa052551aae680 : SUCCESS in 46m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61d53b9a8c9d4f15b28ecd35489737b4 : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00c2de5a88754da48a197d9a454025eb : FAILURE in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5785365a35143fbafb8fc7d6d658af2 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3d0473fa1434a73b091b11fb9696662 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49ef77e6ec064958b4c344baa83bfe09 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1144d6979b1441f8915b394dd2ad2045 : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/843f53f4e5d2439ea196a23f1e4fcc35 : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4942736d2574642b82ee63b4d6754c6980e39647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-01-13 22:16:27.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2895c8b1e7da25fb75a63762bcd82713b84a52ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 23:17:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9ca7623113c248d388a9cd2287af9028 : SUCCESS in 37m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8bf8801943bb4698bd993cac023cb309 : SUCCESS in 33m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/72e5a30d43f248e7bf41164c4aa04bd0 : SUCCESS in 57m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b1ab5471cb464964b69b9e2fc87a83cf : SUCCESS in 33m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ada4f4564ada4eb09a3e3823f612adf5 : SUCCESS in 41m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62a9d6d447a6499ba47bbf1546ad4db6 : SUCCESS in 8m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ac6f55a8e7146088ff962e2b5442911 : FAILURE in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d844bee74414a819d8ac57b42bdb0fa : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8715e1bb5c24328abb07d6dfcea82ba : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7ec5dfe0fa24290912385e04f0f43ec : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a268f20ca8245658d928f3cc68894c6 : SUCCESS in 16m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dab9d3d8d895462096ba9230de6174ab : SUCCESS in 25m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2cca6a81d1646665eeb5b847fcf6e1231b12bdb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-16 22:22:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"bdc2f14bbcd85504921c9e362f42077312e5ed0b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-02-16 23:14:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4824eda23e9a33eb001f7657d575376bdf1330f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 23:29:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1964b07a686343f5a62ff5f519a642a5 : SUCCESS in 30m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/63d5b635e5a84c1e958c4ecee04b1bff : SUCCESS in 52m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c391934acb434f519b89f2a5916cc59e : SUCCESS in 1h 02m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3af1fa0a3e174e2b883f0d3404655f88 : SUCCESS in 34m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7f3adff68ced48b19f4bdfe594c3c906 : SUCCESS in 32m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dca0939d2b14c0bb703cd4147fc1d09 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/421e282b936646239960bcaa36a376ed : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17cef078770f438e8770713e2e854454 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a769667b962a4179848f2433e910b411 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/814b2c7fef6e44b098820f89a865674b : SUCCESS in 7m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/548c84908f8e404ab48ba320be3d3df4 : SUCCESS in 27m 18s","accounts_in_message":[],"_revision_number":3},{"id":"6baa43ea895ec4dd93db708d0dac93c3e5c377dc","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-16 23:30:34.000000000","message":"Patch Set 3: Workflow+1\n\nThanks Erik!","accounts_in_message":[],"_revision_number":3},{"id":"1d49d9ffd9454f0b063ca241289122a3c49c997d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 23:31:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8d5f30b4ea9b96982b4abd00a0cc3669bb2ea74d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 00:38:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/77cd98156ac5407eb33825046b55de6a : SUCCESS in 32m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9eafd5723fb1459dae99186b3cd11e32 : SUCCESS in 34m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/570ad808158d4109ae54ab4a8f7abc9a : SUCCESS in 52m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/af0a1ee39d144967a5baeeea9c131eba : SUCCESS in 35m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c8e2aff82e66403d99de5d12266ab5b1 : SUCCESS in 37m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4e262c7f9694f4faa6564df28983e0d : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e573e3ad8d5464ead56b5c610d78afd : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e619da81e6240d6b700b5c7270cad94 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a67a7841d3446daa477c449f0ea3ca : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":3},{"id":"7b95e5c7a5e4d6bf38448fe5b6a87bed71650eb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 00:38:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"08ceff93b50353bc02140e699206b4b125f5a51c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 00:40:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd8d143fc26473c8dbd9315d5c91319 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dc2426f7f234dac587b0725ad89f514f0afdcd06","revisions":{"8326963de7abeecbd59d2c3fe5ec1c0847d03d28":{"kind":"REWORK","_number":1,"created":"2020-09-29 23:03:21.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/38/755138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/755138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/755138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/755138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/755138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/755138/1"}}},"commit":{"parents":[{"commit":"d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-09-27 10:23:37.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-29 23:03:21.000000000","tz":0},"subject":"Fix pdns4 devstack plugin","message":"Fix pdns4 devstack plugin\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8326963de7abeecbd59d2c3fe5ec1c0847d03d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8326963de7abeecbd59d2c3fe5ec1c0847d03d28"}]},"branch":"refs/heads/stable/ussuri"},"2f65c51b2ac3ca90f2c42d47bc68b37f6c780d50":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-13 22:16:27.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/38/755138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/755138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/755138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/755138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/755138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/755138/2"}}},"commit":{"parents":[{"commit":"9f28e73823dfc50a7b4fb3f8020520131b452eb2","subject":"Merge \"fix database sync always get the wrong  value\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f28e73823dfc50a7b4fb3f8020520131b452eb2"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-09-27 10:23:37.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-01-13 22:16:03.000000000","tz":-420},"subject":"Fix pdns4 devstack plugin","message":"Fix pdns4 devstack plugin\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f65c51b2ac3ca90f2c42d47bc68b37f6c780d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f65c51b2ac3ca90f2c42d47bc68b37f6c780d50"}]},"branch":"refs/heads/stable/ussuri"},"dc2426f7f234dac587b0725ad89f514f0afdcd06":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-16 22:22:32.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/38/755138/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/755138/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/755138/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/755138/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/755138/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/755138/3"}}},"commit":{"parents":[{"commit":"5a526c7aafcd53aa4cfba090512eb687c26c4a67","subject":"Drop lower-constraints test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a526c7aafcd53aa4cfba090512eb687c26c4a67"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-09-27 10:23:37.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-16 22:22:09.000000000","tz":-420},"subject":"Fix pdns4 devstack plugin","message":"Fix pdns4 devstack plugin\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc2426f7f234dac587b0725ad89f514f0afdcd06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc2426f7f234dac587b0725ad89f514f0afdcd06"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
