)]}'
{"id":"openstack%2Fdesignate~755175","triplet_id":"openstack%2Fdesignate~master~I80cb1e658761aaf16943b12e3b4dc6f208cb2eee","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I80cb1e658761aaf16943b12e3b4dc6f208cb2eee","subject":"requirements: Drop os-testr","status":"MERGED","created":"2020-09-30 03:50:39.000000000","updated":"2020-10-18 21:39:26.000000000","submitted":"2020-10-18 21:38:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755175-1603057080783-81340386","meta_rev_id":"670f3ff2773720dfff98e9c4181395fb5301634d","_number":755175,"virtual_id_number":755175,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-18 21:38:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-10-16 09:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-16 09:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-18 20:38:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 16:01:47.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-10-18 20:38:45.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-10-18 21:38:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4578f3bc1a0344c066c61ba9be2460a7089f5ef0","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-30 03:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5924cc7c10c4ea40be0ef476b028e023d23d10e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 04:46:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1d3f3e7ee11545bd9ca2829539e03cca : SUCCESS in 34m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5156fe142da7404fafae41980bcf4867 : SUCCESS in 37m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7d26bd1d40be41b5a10607717ac3e589 : SUCCESS in 53m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/504ba9d3f52648feb87240741f614ef6 : SUCCESS in 43m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/12bc47c3629e4c55948b4624521e3826 : SUCCESS in 43m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68cf9e9265b14fb5b6085b58ea33c670 : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ba438584e3548e690a0b8272d61c538 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba4dd73df9a4fe19297da5ad830dcc6 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4deb240edba74a0d9e57e273283086f2 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d2292880ee8443583d0e21443128e15 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a97b977bf57e4a9589745dc01fc055e0 : SUCCESS in 7m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/59c008c72e9e43c194c1bba1ca257c8f : SUCCESS in 2m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/000a8ff4c98049f2bc7d1dba695889f9 : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5529af4cc59358d0cd82e5117dc10a575f19fb3b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-30 16:01:47.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"909202dd6dad00880d18b4973e7007e6214a7ad3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-30 18:10:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad624e18f3b9cdbc02e38c88e2a7b6b17485a929","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-06 14:15:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80d85eeb6aa1062489a1a5f4c241209e8c204da3","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-12 03:54:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71ae8684b22fa7746c4937cff99d5e5fa291be70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 04:42:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/db37e4037764440fbb21fa520c2bf626 : SUCCESS in 36m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9b4e54837767475eb104484fc3731729 : SUCCESS in 39m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/82c4abd03fec455e8594d127c3f260a9 : SUCCESS in 46m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ecab53cfeadb48f79f127601054614d9 : SUCCESS in 32m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/908acd8351434fd68bfcd8585684b43b : SUCCESS in 37m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76ff4007e9aa4584abf200006f0312e9 : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a53e00dc85904b0c87365f6c9ad85bcf : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/595d1f59435148ea95501508fabb3738 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c5f974ebe9b47b089ecf3b2887f0521 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3908147c7135445180274c737a8abab2 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cbacf2576b74bd9bd7d71911fc6e1e5 : SUCCESS in 7m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b8b2c6e890e465686795a28b0e08641 : SUCCESS in 2m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/286824d25410406b8b6fdccd468bdd55 : SUCCESS in 31m 29s","accounts_in_message":[],"_revision_number":1},{"id":"81046cd487898eb0a0e2d0e137c900d77f0c5017","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-10-16 01:16:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"363bb475ff92b2d5db8de70f989d01a682ee88c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 02:26:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d932e97b324a459693aa387b721b5847 : SUCCESS in 40m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/16e16d175cbf4397ae2c170c24d044bf : SUCCESS in 39m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7f1fbcf2d35a42e08afef408003c5ad9 : SUCCESS in 56m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b5f88aa6dbb24b65ace1dc10ef4af68e : SUCCESS in 40m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ad3a7ac1ae074b2f81754cb08b6b7bd9 : SUCCESS in 39m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db1a4324fd2d4a41bbf2946084d5627d : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b39e96ced9a48eeb38380dc28e7ced4 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a008c0e099184af7acff89a10285a962 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02f466e322364c78b2beaf59efec2796 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75b63f8ad1e343638c5e566127ce4afd : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ff84accf0164e77b32b37ec9dfd654e : SUCCESS in 9m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/acd68eea41d543ed9057460017fb687a : SUCCESS in 3m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f177b8fb1b794ba8b3eab0b6acf4f5c8 : TIMED_OUT in 1h 06m 45s","accounts_in_message":[],"_revision_number":1},{"id":"076b249634d1bc9f54fd3eed61719f93902003fe","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-16 03:54:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"506e99fea5518f41fe31e5b81e4076ca5b542e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 04:53:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0fa23c97acd44cebbb649be54c33d400 : SUCCESS in 38m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/867d4c02a3ad4e7f8abc3441f082345b : SUCCESS in 39m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/95de375b3e8048098ea303b88aac65f8 : SUCCESS in 58m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2bb742b36c0c4868898d037547beb9bd : SUCCESS in 38m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b343f944c5014bdcae5e89169b78604e : SUCCESS in 38m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7829c723768746f488eb3174b82aef41 : SUCCESS in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0536b72eb6042e59a14359be17c0d23 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a74bb0315042fda4a7bb2c1e2f2acf : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e0ba7ca90e44d44af85585deaf9f248 : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3a681d9e13f4ae1970cd61e5715c118 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99526fcca248485981da7bf0c3cdd9b1 : SUCCESS in 11m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a1654b6e92f94adb8609830c85fb765c : SUCCESS in 3m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8cd7a45ac02148d0b194d3eeb940bb94 : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b757a5fefeacb7755db80dc5e82da4e584fcccf4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-16 09:34:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ca582cd0860883a2ecb42f1732fc83a08f7f9934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 10:27:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0fa8eb194d1048a6b80967d2dadc0006 : SUCCESS in 47m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/73841d3729fd4a1386230436c1d2b911 : SUCCESS in 38m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9d8c88de85e493f98eb517a7858d2e3 : SUCCESS in 52m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5c6bef504d0a49678c982bc343fdb992 : SUCCESS in 38m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/609da59b801746698d4c1993be3aea3a : SUCCESS in 39m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bc16900110c4980abdbf106b19da074 : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eab1379c5ff743a7986f62d302faf91f : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be66167e8b3346be9e8bda3eaa472807 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a25fe66462084ecfb3d9c80b9ff4f98e : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1813ad3b9aae40daa6f693a45df43305 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b15d10cb29249cd970dafd9ca261d6e : SUCCESS in 8m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/19ed1183e1f1460998d99cf3826f4ff0 : SUCCESS in 3m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e684211c7cc74b66a572264579794d79 : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":2},{"id":"af8c5dd2e812f801b2417dd648270690574bbfbd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-18 20:38:45.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"160b22b34c746046c5f189ff62fc2e56af6d1758","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 20:38:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e03e5d8a7e815c37b8d9bc6a95c4633ec8f58ba7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 21:38:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c7ba6fc8f0552b5bcad0178b37306d69c6d95502","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 21:38:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/32282c63ff8445aa9ec3bf943edb4499 : SUCCESS in 42m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d9ee94043d1b4ee8bb101612ffe021e4 : SUCCESS in 40m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e7a72e8295c74a7eaf57f4cfdc4f029c : SUCCESS in 56m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/14ef3f68c1464e0fb6077927ca111116 : SUCCESS in 37m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/28dfc6a074ea4aea9670e2feb9081b58 : SUCCESS in 40m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a699119bf0cd4c5f8d16153af9737c95 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2abb6a32319a4f189679d518dc4e8e6c : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80c8fcb3ccdc4a38a376835807213e70 : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7440e4c0f2524c83a13547733824193a : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c7b2302faa44a27ac2654ba50a66890 : SUCCESS in 12m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a3eea6216d6d41b388d65359c4d46fe2 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"670f3ff2773720dfff98e9c4181395fb5301634d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 21:39:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3533bf93df984608bb5c4689a7190f91 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8459ed9460ab492adfd305875c717ebe31bf68f3","revisions":{"d07d4f2eaacd8d26dcc0deb1ce0a1ba9e0b384c1":{"kind":"REWORK","_number":1,"created":"2020-09-30 03:50:39.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/75/755175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/755175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/755175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/755175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/755175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/755175/1"}}},"commit":{"parents":[{"commit":"0850d2d3e53897dcf10afea588bf97a1eff3b815","subject":"Remove the unused coding style modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0850d2d3e53897dcf10afea588bf97a1eff3b815"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-30 03:49:38.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-30 03:49:38.000000000","tz":480},"subject":"requirements: Drop os-testr","message":"requirements: Drop os-testr\n\nWe\u0027ve switched to stestr.\n\nChange-Id: I80cb1e658761aaf16943b12e3b4dc6f208cb2eee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d07d4f2eaacd8d26dcc0deb1ce0a1ba9e0b384c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d07d4f2eaacd8d26dcc0deb1ce0a1ba9e0b384c1"}]},"branch":"refs/heads/master"},"8459ed9460ab492adfd305875c717ebe31bf68f3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-16 09:34:04.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/75/755175/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/755175/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/755175/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/755175/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/755175/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/755175/2"}}},"commit":{"parents":[{"commit":"9eae67abc6e5aaac061c4576a82f54f3653cef4a","subject":"Merge \"Remove the unused coding style modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9eae67abc6e5aaac061c4576a82f54f3653cef4a"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-30 03:49:38.000000000","tz":480},"committer":{"name":"Jens Harbott (frickler)","email":"j.harbott@x-ion.de","date":"2020-10-16 09:34:04.000000000","tz":0},"subject":"requirements: Drop os-testr","message":"requirements: Drop os-testr\n\nWe\u0027ve switched to stestr.\n\nChange-Id: I80cb1e658761aaf16943b12e3b4dc6f208cb2eee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8459ed9460ab492adfd305875c717ebe31bf68f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8459ed9460ab492adfd305875c717ebe31bf68f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
