)]}'
{"id":"openstack%2Fdesignate~755349","triplet_id":"openstack%2Fdesignate~master~I72bae79b4f8e0bab739875d6f783abdb4868950b","project":"openstack/designate","branch":"master","topic":"bug/1897936","hashtags":[],"change_id":"I72bae79b4f8e0bab739875d6f783abdb4868950b","subject":"Fix pool update with duplicate masters","status":"MERGED","created":"2020-09-30 16:50:31.000000000","updated":"2020-10-01 07:53:35.000000000","submitted":"2020-09-30 22:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755349-1601504163036-d925a52d","meta_rev_id":"d72dc2ee7da1fa0d189e29f97304077fd72a17b6","_number":755349,"virtual_id_number":755349,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 22:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-09-30 18:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-30 18:45:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-30 18:45:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 18:20:53.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-09-30 18:45:27.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-09-30 22:16:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86fb66f8e28666790dd29a3ae65096aa8c9ceba3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-30 16:50:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95d7636a1b8f5ae5070e9d0abb5cb4d28bfe615b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 18:02:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2283bf1a4a124178b86725ed64ae54c3 : SUCCESS in 40m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fa21bf66702d409da6aeabb9faf049fc : SUCCESS in 40m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bb2ecc0e63bc4b70b295a56dacf7a6c2 : SUCCESS in 1h 00m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d630173056b24ebd9ea7ea3f100162ce : SUCCESS in 38m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/448c28668ef6484c8a1dfe4f1c87c4f1 : SUCCESS in 38m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89621a60f38d4002b94178ef55bf1e5e : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d170b9d8c16440e39885bc9463a0baae : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc6498dd0aaa472fb8c533afc26ad739 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edbeaac0463f408f846babe936a07cac : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42db37b0f3c84cbfbf5d2efbe6353d49 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3e77962af284afab1977bac3e508e19 : SUCCESS in 7m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0b074f82610040ddb2c54ed63141f04a : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":1},{"id":"936f8e1ae1621e02eee1984a859492cbc23af465","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-30 18:20:53.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"1cf448899a8069359b155b15a34d5b2f40dcad49","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-30 18:45:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd63292c23ce1e6a474d337d4cb2444edec853dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 18:45:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb6c18d9bc28ffb527f9ad94ea593b79cbd37c30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 19:58:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/051e481da8f34dbe9bb182d1350fbe49 : SUCCESS in 40m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/13724d39115849e487475c1cce7a7cf5 : SUCCESS in 38m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/453d3fe4d3ca44ba87217a83c55e78c0 : SUCCESS in 55m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e4d20a90be204229bef6267d01be4007 : SUCCESS in 39m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f46e42224dfb48068c758b91fa63b2f6 : SUCCESS in 1h 04m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eb6663a523b4460b3964762473d47d4 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d499f23e30245feb7b3337903203bc8 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/660f44f339cd4bdd81da1cfd640d795a : FAILURE in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc55daec079743a7bdd778b2a3a81db6 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48fb53ee3dbd424a8d97caa749a73784 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ac8cbe7111bce33c292355672cdafbe21424f99b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-30 20:13:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff5734062794a81ec6fdfc5e106fdacd43cb63b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 21:10:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cce7bb019261450893bfae46cb76818c : SUCCESS in 30m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a2990b1d781e4e55bc7e2c0e30619686 : SUCCESS in 40m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5b547f8fb9e2485e898994920f71f969 : SUCCESS in 53m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/900529eef50e48d5b54e9e527e98ce25 : SUCCESS in 40m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9f301645e36e40e699d2a00af99033b8 : SUCCESS in 40m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b6956f3c74d4988b6fb1e4c987f438f : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c97131991794af4b67e1df014113256 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/090f145d93a3453892edfc17fb42b4c7 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7be201f9b6d430fbeb5e03fce4784ff : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8854e68e4f045d5bcab6ffb59040196 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70b68a1fc2c14e58a2621ef7b4cfd701 : SUCCESS in 8m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/937599ed1fdd4b28aa5a38992d471ac4 : SUCCESS in 36m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b185f356fa488554c117e3f792ebbe2c6b1bc4ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 21:10:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ec9f4d2a1b2621d160b50f9b6d724d04f2d661d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 22:16:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/35f8a3c00bd04a7494ead859201457fe : SUCCESS in 41m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/be9c7ebbe9b9410599adf16d612bcd81 : SUCCESS in 41m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a7eddd9db78b4cc6839dcae6915ae228 : SUCCESS in 58m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7ce9c72ae2934492b83c80c448a309b8 : SUCCESS in 42m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d38676dc88d54d9da94d2e21c20083e7 : SUCCESS in 45m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf7bee630bc7459da24b6f2ef8348593 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/666d50174ab742a2b616148f6eea2321 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fd26c7d583d4892901c9a6d03428433 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1211ed50206e438faada758793f04b16 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9061e8f565c9458dbe8081d62c50ed42 : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6092e47b525cee5d29d8bdd2ab60171439875a2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 22:16:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c6e959be88d18a67234e9d9759e0422ce00f9576","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 22:17:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a7998b049b4045a29c3bf76ea82073 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1bf4578444e56e27f86ac7ad6182d358bc37ed07","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-30 22:32:10.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 9e159d93d99bf1ed0c487bfe99450f59db509249","accounts_in_message":[],"_revision_number":1},{"id":"3a9e60932f403fdb67e41e10d4e405993b384e2f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-01 05:50:58.000000000","message":"Created a revert of this change as Ic794ae3091bf264213e2bdcdc24674e7801bfa0f","accounts_in_message":[],"_revision_number":1},{"id":"9e054bcd123ee619ba7dbe38cef52f4e82c14a2a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 07:53:35.000000000","message":"Patch Set 1:\n\nThanks for the fast merge :)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e86bc2a58f00371a2f22217cb8a79e5ac33677c","revisions":{"4e86bc2a58f00371a2f22217cb8a79e5ac33677c":{"kind":"REWORK","_number":1,"created":"2020-09-30 16:50:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/49/755349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/49/755349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/755349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/755349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/755349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/49/755349/1"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-30 16:48:25.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-30 16:50:28.000000000","tz":60},"subject":"Fix pool update with duplicate masters","message":"Fix pool update with duplicate masters\n\nIf any zones exist that belong to a pool that has any duplicate masters\namong all targets, currently updating the pool will fail.\n\nThis change fixes the issue by ensuring the pool target master list does\nnot contain duplicates.\n\nChange-Id: I72bae79b4f8e0bab739875d6f783abdb4868950b\nCloses-Bug: #1897936\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e86bc2a58f00371a2f22217cb8a79e5ac33677c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e86bc2a58f00371a2f22217cb8a79e5ac33677c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
