)]}'
{"id":"openstack%2Fdesignate~755361","triplet_id":"openstack%2Fdesignate~master~Idf48fad73572e632dc3c1a24b696fbf7bacf0ad9","project":"openstack/designate","branch":"master","topic":"bug/1852902","attention_set":{},"removed_from_attention_set":{"31039":{"account":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"last_update":"2022-09-24 05:53:57.000000000","reason":"Change was abandoned"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-09-24 05:53:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idf48fad73572e632dc3c1a24b696fbf7bacf0ad9","subject":"Pdns4 driver adding TSIG key through pool.yaml","status":"ABANDONED","created":"2020-09-30 17:50:58.000000000","updated":"2022-09-24 05:53:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd596776084d08d41b8a46d5b018538620a9e765","_number":755361,"virtual_id_number":755361,"owner":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-24 05:36:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 18:08:40.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-09-30 18:08:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-10-14 01:32:12.000000000","updated_by":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"reviewer":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"state":"REVIEWER"},{"updated":"2020-10-18 20:41:44.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-11-19 08:25:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73087623726efba3c851bed4fb7034673a7eb62f","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-30 17:50:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7272dd4bb91fbf701bbfaa59d43e7db94cf90c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 19:04:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c3cdf5542ea5469b8b6735bf5c3d80b2 : SUCCESS in 49m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d8507fd3559943409556974ba929b83a : SUCCESS in 43m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/63c184baf0c94877b1f09051f860db70 : SUCCESS in 1h 03m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/108fae3a2de4400fa29289d3ca326572 : SUCCESS in 36m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/469debf71c4b4161b036e30ffd023cd2 : SUCCESS in 35m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55772ad9e64340028013d1adb8199c18 : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88f38d126de9499c971c5a03fb7e3cd1 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34392ba8a2d6475185d7fa063ed85e23 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eae1cf1b5f5040749592adb961aeea6a : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c52ef01694340b0b857d9d915a486c5 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb5ee1d8a5341d189cf9eee486af756 : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddd5537862cd4f9c995c2a43ec637774 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0f3920b3cac4eb9bc802ca35b0d1ea9 : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ee298cb46da1c41969e33f8683197034e2078206","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-10-12 18:18:53.000000000","message":"Patch Set 1:\n\nany feedback on this","accounts_in_message":[],"_revision_number":1},{"id":"a529ea77c8f62bbce0e58aea0f60eb1c64350c94","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-10-14 01:32:12.000000000","message":"Patch Set 1: Code-Review+1\n\nLook good, are you planning on adding any tempest tests for the feature?","accounts_in_message":[],"_revision_number":1},{"id":"d49b73d22b0d7c70ac313f5122a46c40b69aafb1","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-10-14 17:28:05.000000000","message":"Patch Set 1:\n\n\u003e Look good, are you planning on adding any tempest tests for the\n \u003e feature?\n\nYes i can add test to https://review.opendev.org/#/c/755876/","accounts_in_message":[],"_revision_number":1},{"id":"0d8944ad1511a0b6f2ec7a372363590377a5a1f4","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-10-15 03:17:46.000000000","message":"Patch Set 1:\n\nThanks I\u0027ll take a look there","accounts_in_message":[],"_revision_number":1},{"id":"2cce6ba0726bd363081e1c1590b93de87853adb1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-18 20:41:44.000000000","message":"Patch Set 1:\n\nThis code path previously had 100% test coverage. It would be nice to try to keep the coverage at least near 100%.\nhttps://f480bb5fdd53fbb72663-fe8e452d8420a7f2256fc373ec02a39c.ssl.cf1.rackcdn.com/755361/1/check/openstack-tox-cover/55772ad/cover/designate_backend_impl_pdns4_py.html","accounts_in_message":[],"_revision_number":1},{"id":"f8c170ff74578945d9a040ae1e838629edabcd58","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-10-21 17:59:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78d8372959a86eefdbc917b2c8595c9b534c9554","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-10-21 18:00:42.000000000","message":"Patch Set 2:\n\n\u003e This code path previously had 100% test coverage. It would be nice\n \u003e to try to keep the coverage at least near 100%.\n \u003e https://f480bb5fdd53fbb72663-fe8e452d8420a7f2256fc373ec02a39c.ssl.cf1.rackcdn.com/755361/1/check/openstack-tox-cover/55772ad/cover/designate_backend_impl_pdns4_py.html\n\ni added more unit test for this","accounts_in_message":[],"_revision_number":2},{"id":"0db2ec024f172b57a6a71c04ef3c98a878998f60","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-10-21 18:09:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"88eb814a980761def3dbc83a5d07d05e32be5d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 19:19:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4ee6afc35ed847bb8e6f592bd5ba5521 : SUCCESS in 43m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f2ede5bf350640ffb8f5499e521d83c4 : SUCCESS in 47m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e2ed1f6f14454c2dab2f6ca9a1fcad59 : SUCCESS in 1h 02m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9839e4ac9a2649cdb557ca44201d0c28 : SUCCESS in 42m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fce60da0d15a43f58fbdddada8f5d786 : SUCCESS in 44m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d67a2d9fc72a4035bc8174d7f8afea58 : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9617991f975d4bd8a55cf562217da440 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c372ab3de94f47c69bdb2904dda7e227 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc59c273e7394ee888dac51fa2cb77ac : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/499ebb3bb9fe4465b46ddcbb7648665e : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2fc506a18649e28b633cc2a0dd9237 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edc8ac2dbffe497dac558278c911c0ea : SUCCESS in 5m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9386712336e64bbb9aeb2c3c37978c49 : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":3},{"id":"cfc643d649f71cd5613fda2f11a6953f7f5e9db6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-19 06:22:34.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"157d309db31c79a734b6d68de01f5c359b5c10ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 08:25:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4629250111064a0d8b07bf80058a69e5 : SUCCESS in 42m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5341744435804bf28ebd06185203b888 : SUCCESS in 39m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ca12bb133a3f459d867c9fbb8239e843 : SUCCESS in 55m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fefcc58943a7480c857c0e763f227b18 : SUCCESS in 40m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a9893fd1e3554eb6be19934aca7e1a79 : SUCCESS in 41m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a92062f7a14a4cb6a2ac77c929ecf2a6 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f47163f7c754cff86a60ebbab7c8e23 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2f3ea5a718746db93a4be67557a7d69 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f7b5b90a4cf44e1be6c6b0d1a68304f : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd204616472e46cc93d9bf4cd5e3f2de : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/459587d57e37459f82056343a25e0dfe : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5289219e199a46d8b46547de0ce49b7d : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ba8afab41f64a8c9f59ed3b1c561d7a : SUCCESS in 5m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e121c7e4528a4bf982bc57256ed7427a : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":4},{"id":"cafcfc99b408e973b1a570168306f61162d2f491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:36:54.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"205779b4e5e46e230c8e4c8664efaaba4f94cd03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 05:36:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5392f64b24cd498fa7e980a288d3f0d0 : SUCCESS in 40m 57s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/52846a817d8344d7adb815bf8c9bbe63 : SUCCESS in 43m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0af5af36a0e24fddaf2b1d141d9e5508 : SUCCESS in 40m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9f46d7a84afa43bcac298709d6a08238 : SUCCESS in 36m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/096124a7289247078c8406e38714e88f : SUCCESS in 57m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/759b99528c16437993c6d97f085969a1 : SUCCESS in 37m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d283a83fd9ac48adab69c2ba186fe113 : SUCCESS in 38m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82867ce8c9f04b91819411b7fef207ac : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22def3c7a7ff49029c07a339a213afdf : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fcb6607db4c4f4b82c5a90bcee0b86f : FAILURE in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab95500a52614b07b99d9515f42f633a : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4409b7936514217b1c8e635ac3415b0 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b29e6dab0d914233b55e3cee8a426626 : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89e12c6842694171aa7afd966333a3c3 : SUCCESS in 4m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eab17c47f0b04f5993c10223338758a1 : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":5},{"id":"bd596776084d08d41b8a46d5b018538620a9e765","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:53:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9e105129a3a71fc5298439868c24187ad13c6e7d","revisions":{"ffc54d118d644375fb1046dc9aac8b1742ea66b6":{"kind":"REWORK","_number":1,"created":"2020-09-30 17:50:58.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/61/755361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/755361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/755361/1"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-30 17:44:19.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-30 17:50:52.000000000","tz":0},"subject":"Pdns4 driver adding TSIG key through pool.yaml","message":"Pdns4 driver adding TSIG key through pool.yaml\n\nIf the option `add_tsigkeys` is provided when using Pdns4\nthe listed TSIGkeys will be added to PowerDNS backend\n\nChange-Id: Idf48fad73572e632dc3c1a24b696fbf7bacf0ad9\nCloses-Bug: 1852902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffc54d118d644375fb1046dc9aac8b1742ea66b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffc54d118d644375fb1046dc9aac8b1742ea66b6"}]},"branch":"refs/heads/master"},"59e2db434e865bece57757d5d177ecbd4daa4f50":{"kind":"REWORK","_number":2,"created":"2020-10-21 17:59:26.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/61/755361/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/755361/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/755361/2"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-30 17:44:19.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-10-21 17:58:59.000000000","tz":0},"subject":"Pdns4 driver adding TSIG key through pool.yaml","message":"Pdns4 driver adding TSIG key through pool.yaml\n\nIf the option `add_tsigkeys` is provided when using Pdns4\nthe listed TSIGkeys will be added to PowerDNS backend\n\nChange-Id: Idf48fad73572e632dc3c1a24b696fbf7bacf0ad9\nCloses-Bug: 1852902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59e2db434e865bece57757d5d177ecbd4daa4f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59e2db434e865bece57757d5d177ecbd4daa4f50"}]},"branch":"refs/heads/master"},"d678aa32e265fc1ccb8d6dae54ffc8d8170b7b5f":{"kind":"REWORK","_number":3,"created":"2020-10-21 18:09:18.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/61/755361/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/755361/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/755361/3"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-30 17:44:19.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-10-21 18:08:57.000000000","tz":0},"subject":"Pdns4 driver adding TSIG key through pool.yaml","message":"Pdns4 driver adding TSIG key through pool.yaml\n\nIf the option `add_tsigkeys` is provided when using Pdns4\nthe listed TSIGkeys will be added to PowerDNS backend\n\nChange-Id: Idf48fad73572e632dc3c1a24b696fbf7bacf0ad9\nCloses-Bug: 1852902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d678aa32e265fc1ccb8d6dae54ffc8d8170b7b5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d678aa32e265fc1ccb8d6dae54ffc8d8170b7b5f"}]},"branch":"refs/heads/master"},"b7be76fef1189ab8bf0edded641bf028fa281a60":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-11-19 06:22:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/61/755361/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/755361/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/755361/4"}}},"commit":{"parents":[{"commit":"e49b80703f002ad2f9c1dc5c35e7cc75a51ac9c0","subject":"Merge \"Add timeout to rndc commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e49b80703f002ad2f9c1dc5c35e7cc75a51ac9c0"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-30 17:44:19.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-11-19 06:22:34.000000000","tz":0},"subject":"Pdns4 driver adding TSIG key through pool.yaml","message":"Pdns4 driver adding TSIG key through pool.yaml\n\nIf the option `add_tsigkeys` is provided when using Pdns4\nthe listed TSIGkeys will be added to PowerDNS backend\n\nChange-Id: Idf48fad73572e632dc3c1a24b696fbf7bacf0ad9\nCloses-Bug: 1852902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b7be76fef1189ab8bf0edded641bf028fa281a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b7be76fef1189ab8bf0edded641bf028fa281a60"}]},"branch":"refs/heads/master"},"9e105129a3a71fc5298439868c24187ad13c6e7d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-12-24 04:36:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/61/755361/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/755361/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/755361/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/755361/5"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-30 17:44:19.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 04:36:54.000000000","tz":0},"subject":"Pdns4 driver adding TSIG key through pool.yaml","message":"Pdns4 driver adding TSIG key through pool.yaml\n\nIf the option `add_tsigkeys` is provided when using Pdns4\nthe listed TSIGkeys will be added to PowerDNS backend\n\nChange-Id: Idf48fad73572e632dc3c1a24b696fbf7bacf0ad9\nCloses-Bug: 1852902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e105129a3a71fc5298439868c24187ad13c6e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e105129a3a71fc5298439868c24187ad13c6e7d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
