)]}'
{"id":"openstack%2Fdesignate~755363","triplet_id":"openstack%2Fdesignate~master~I57a7649fde1cceda787622aa1975f3a7100aeb29","project":"openstack/designate","branch":"master","topic":"lc_vine_five","hashtags":[],"change_id":"I57a7649fde1cceda787622aa1975f3a7100aeb29","subject":"Increase lower-constraint for kombu","status":"MERGED","created":"2020-09-30 18:12:54.000000000","updated":"2020-10-15 17:32:55.000000000","submitted":"2020-10-05 16:48:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755363-1601916517732-5cec80e2","meta_rev_id":"1e3754ed5c9186f2a8bc7a36c5d305161be74cb3","_number":755363,"virtual_id_number":755363,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-05 16:48:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-10-05 15:29:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-10-04 07:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-10-02 16:57:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-10-05 15:29:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 15:49:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-02 16:57:55.000000000","updated_by":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"reviewer":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"state":"REVIEWER"},{"updated":"2020-10-04 07:39:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-10-05 16:48:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4360c83a078d98ea152137095fe9c0f05383e513","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-30 18:12:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba91eb33a22433679ca16bf799e70fc89b1a5d75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 19:31:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e9f6203db84b47bbb29df0a434d16f2e : SUCCESS in 1h 09m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/64f7055b65e74f448d04fe71ed0160b8 : SUCCESS in 40m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9436261677ea4015974066c27c499f39 : SUCCESS in 49m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ba04c9e7fd6c4a48b7b28c25ca443cc7 : SUCCESS in 38m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5ac7b9b3a55243f2a8ed6e902f6bcbdf : SUCCESS in 43m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec8468dc87ce4d7cb61e5b3808fc9b2c : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30fbbf0e0d9449c280c336850c8fca6a : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52dc1a9b1a034e77af5496b2c1c6222c : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36e802692fa843bd91797f31ccb75d38 : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/041fa7e0309543c6b8ec134f41e71ee4 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60febeaf7cbe41f79f617cf2eef96820 : SUCCESS in 7m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0195fb91ca584e6ca99f3f5734ad5e7b : SUCCESS in 2m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8c0cda543b1945a4ba4664289fbc4127 : SUCCESS in 35m 54s","accounts_in_message":[],"_revision_number":1},{"id":"18d76a9ed788a11b5d1cf85560baead471334caa","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-30 20:58:48.000000000","message":"Patch Set 1:\n\nI think going to 4.2.1 will be better based on this https://docs.celeryproject.org/projects/kombu/en/stable/changelog.html\n`The 4.2.0 release contained remains of the async module by accident. This is now fixed.`","accounts_in_message":[],"_revision_number":1},{"id":"f21e31da11d13fe9fac7f1bc9b1f60246f724c11","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-30 21:29:24.000000000","message":"Patch Set 1:\n\nThanks for the hint Hamza!","accounts_in_message":[],"_revision_number":1},{"id":"65e8301ef7e81f352f8871796f83580682083503","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-30 21:31:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d331582a27a572b490fff828675905b2aed8476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 22:29:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/07aaf4ffb7244748b816bcd2f555ff65 : SUCCESS in 37m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ac8334d600204ffd88d042cb7b72faa9 : SUCCESS in 35m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6c1da7d043cc4627a0a81870bcc96d1d : SUCCESS in 55m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c7c6b7c236d3458e8d138e6a870e7b93 : SUCCESS in 43m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/875c004f6f5a4983b9a2b30560d8b57a : SUCCESS in 42m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/104a5cc60d8149b098a3b0d26c72c151 : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc7841a051be4261b573718a8eaff0b1 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0bca55e7994062941a8fd762662751 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/999abf88dc6840d59ca860b7a86be390 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38fd018768d14ea791da0f3d85809147 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969ddd0163944be8af84838cfb3a9349 : SUCCESS in 7m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bfe164a13ea947769f9e1f935ed32f45 : SUCCESS in 2m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/557161cc51a54d2192e31171b5c7c94b : SUCCESS in 36m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5df1d789ac3a7183fd7adca6e71e67222f613cc8","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-30 22:31:57.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"5a127a608a91b87f2120d9cb033a4ede5971e323","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-01 15:49:27.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"29f11df3cd653ca67598ed541ee2224774371c6b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-02 05:51:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"233f7954ffacf0f57df9c34d9dc48d7d6b53362b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-02 16:08:08.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0c52516202252969cb6a71d1849dd6c8b9510d93","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-02 16:08:24.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks for the review!","accounts_in_message":[],"_revision_number":2},{"id":"39ae54c816ce4cf884b44edfe6b5e5c117b36f2d","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-10-02 16:57:55.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTTM","accounts_in_message":[],"_revision_number":3},{"id":"2e5644b60a3a8ed9868088c3c282a12e9b91b63f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:08:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5a661f7ab1f34528a579eac784b7a5a9 : SUCCESS in 40m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c9817fb324ab4f1abd48b2a363904e70 : SUCCESS in 40m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6a9d8249c9044084b7ddae54d96c0a0f : SUCCESS in 55m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8940d79106a34f4d8de622075dd0da75 : SUCCESS in 41m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b70a3d43420847d299f12b8ca465de8e : SUCCESS in 40m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27dfc1a6c82e4d0d832ad286fb38b6a8 : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0caca9b899bc471e8b7cc34c9ebbc972 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e1a612c705140ecbb5c9d553fb4c238 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c07607234bda495e96317ecf6c74773e : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8d28e999e0e462994dafd2d030631b4 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d014b5582281449c8758417ede804143 : SUCCESS in 7m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf9f60ad5db64b0c994f1e3297421ad1 : SUCCESS in 3m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/526a3d3d27084a6596d06dc9fcb09cb0 : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":3},{"id":"f4e4e84561d2345f02f376b6868ad69408f44b26","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-04 07:39:15.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3c7c4aace51a3e10f89a8720a49259d9b9d5c2b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 08:41:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/148b7af0c11a442d8452e8d842742909 : SUCCESS in 40m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fec884ac14004c8583456b73cdb06d39 : SUCCESS in 40m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f34656b89cb14d7fa285c32786de738f : SUCCESS in 56m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e49c8b52d8994c93915b2bce51802134 : SUCCESS in 38m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c6b7bafdc1c94dd4a9059aaeeec0fe4e : SUCCESS in 41m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9db7f01ec514a92a2cd0e9d5447cb93 : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3aeefef02f904aa4963831fc7ac2c2fa : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ace73ad482646998912757e290de2a8 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7a8b95754244b7e8ddb9e9260367bac : FAILURE in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4d6e804695e4c95828604c14505d7ae : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e89ee3c41974dad88206bf747966350 : SUCCESS in 6m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d6ec3a5dd1145dc9e9466cabfbcb1dc : SUCCESS in 2m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf147ccb812843db82f7d22718f30e89 : SUCCESS in 37m 33s","accounts_in_message":[],"_revision_number":3},{"id":"cbe4a10048e11dfcda30eef20c9aba7b2f7bed9d","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-04 12:34:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5071026b41654d374510c5681f304b7a5549574b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 13:29:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ab2192f7c67b419ea764cf8fa03f9202 : SUCCESS in 34m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5d9d0654a63d4ca9a24bf9e9fd4da315 : SUCCESS in 35m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8ab677472a754d278d848f6b5884c506 : SUCCESS in 54m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7894355426d24efd9810e52eeaddaf3b : SUCCESS in 35m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0c4c831247694f438c23c45a2b8f0e4c : SUCCESS in 36m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acc0bb3a5cdb47eab32e23afcb4e8316 : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ef8eafa351245aba14f636351d8f75a : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fba06aa945844d094ae3b5b6904ef01 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/650ee12316e74c7ba1d16ac64bc9c5bc : FAILURE in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9aa65fab883e48fcb5727280a47b58ee : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30e7e7b20a5b4a11b9f42bdf9cd5136c : SUCCESS in 7m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1dbd327ebff64b6f880e78f6df7ab646 : SUCCESS in 2m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bef299f501c146de94a52c5769bbd9af : SUCCESS in 33m 18s","accounts_in_message":[],"_revision_number":3},{"id":"ce73bc39d6c13b9a4540390db3762431b3cab786","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-05 12:18:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"08dfc8b5049e21be95b6c48fd9f7e53be25212b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 13:13:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a35202ae14944ac9b2256118f88c3583 : SUCCESS in 45m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4255cd86713a4d3cb91bd05a5e591847 : SUCCESS in 39m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/05952db3907c45ea966a3d2b16cd3434 : SUCCESS in 51m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0bf8042ef5bf44709328a2745e909345 : SUCCESS in 41m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/db1402b23cf84e96b83ba0695059f9e1 : SUCCESS in 47m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e477bf011cec4a238b4195b1dca1546c : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11aa5a70a81d490782cdeb30cd93f183 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d22b3c82584b8d88ba54e68183ca34 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2741f2d88b324de8a650a1542c55effc : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b796accc82214be6996e44d0e93c0d26 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27399b36bc95442da5e32ba635ae6177 : SUCCESS in 7m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/537ec0f667dc4266b26fe1988d3b311f : SUCCESS in 2m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf4c276396a14f8eb682f6680009ff36 : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":3},{"id":"d6dded881e4d2bf71ed6001541a8854b6084382a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-05 15:29:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"6938dde537736dadce672d7cbf4b4fcdeec2fddb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 15:29:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"50eb4032d488a8183d201113d6a4f8b04ac4f4f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:48:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7060163cd29a41ca9d92d5ad08a17bdc : SUCCESS in 39m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1c777ebde1bf4754baca24b35a54d259 : SUCCESS in 45m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f686549aea8f41b4a19549008382ad13 : SUCCESS in 56m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1f4f9c81d171452297f18fa5ff5a4dad : SUCCESS in 47m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d98ae4d23fb1457b8a238027f87019a6 : SUCCESS in 1h 11m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4aba28ef9d8a4ceb9062e66c5a74f650 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2f1c4acc421438f92dd5dd7787daafd : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2153cf21009f47bf806698ee74be3655 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cd662558180402cb50c496dbef51d2d : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e738dc0ed96e4c20900ec93102831095 : SUCCESS in 9m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/85f50114f00740cfa29bb96234c42cad : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":3},{"id":"c4d19e3c81728dc55f523b532c586f6cd72e2a1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:48:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e1c0c3b5d7667b08893490f5d8d78bd94814267a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:59:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae7db88960b64611a0c43fcb88690a01 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3},{"id":"19174bcbaa35d02e9cd1b93df4fc0353c5d73b8d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 03:50:23.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 490070d637b00d4138f9ed2dfbfccd10e2400444","accounts_in_message":[],"_revision_number":3},{"id":"5f3022366d3b3c1fedb89a4b17237c6b52b2e0f2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 17:32:55.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 4f42a815bd57afa66db1af8405717380af233369","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a0d20514c6ebbfe6b42b989b6192ae77312d3ba7","revisions":{"82594564fa511701fd3c9a9a5dc24786da48d986":{"kind":"REWORK","_number":1,"created":"2020-09-30 18:12:54.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/63/755363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/755363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/755363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/755363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/755363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/755363/1"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-30 18:10:26.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-30 18:10:26.000000000","tz":-360},"subject":"Increase lower-constaint for kombu","message":"Increase lower-constaint for kombu\n\nThere is no module named \u0027vine\u0027 on kombu\u003d\u003d4.1.0 version [1].\n\n[1] https://github.com/celery/kombu/issues/851\n\nChange-Id: I57a7649fde1cceda787622aa1975f3a7100aeb29\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82594564fa511701fd3c9a9a5dc24786da48d986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82594564fa511701fd3c9a9a5dc24786da48d986"}]},"branch":"refs/heads/master"},"0555bfbb48f48bff35d64ceb8111c7a707355af0":{"kind":"REWORK","_number":2,"created":"2020-09-30 21:31:23.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/63/755363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/755363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/755363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/755363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/755363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/755363/2"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-30 18:10:26.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-30 21:30:28.000000000","tz":-360},"subject":"Increase lower-constaint for kombu","message":"Increase lower-constaint for kombu\n\nThere is no module named \u0027vine\u0027 on kombu\u003d\u003d4.1.0 version [1]. Bumping\nto kombu-4.2.1 (and not kombu-4.2.0) [2].\n\n[1] https://github.com/celery/kombu/issues/851\n[2] https://docs.celeryproject.org/projects/kombu/en/stable/changelog.html#version-4-2-1\n\nChange-Id: I57a7649fde1cceda787622aa1975f3a7100aeb29\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0555bfbb48f48bff35d64ceb8111c7a707355af0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0555bfbb48f48bff35d64ceb8111c7a707355af0"}]},"branch":"refs/heads/master"},"a0d20514c6ebbfe6b42b989b6192ae77312d3ba7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-02 16:08:08.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/63/755363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/755363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/755363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/755363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/755363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/755363/3"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-30 18:10:26.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-10-02 16:07:44.000000000","tz":-360},"subject":"Increase lower-constraint for kombu","message":"Increase lower-constraint for kombu\n\nThere is no module named \u0027vine\u0027 on kombu\u003d\u003d4.1.0 version [1]. Bumping\nto kombu-4.2.1 (and not kombu-4.2.0) [2].\n\n[1] https://github.com/celery/kombu/issues/851\n[2] https://docs.celeryproject.org/projects/kombu/en/stable/changelog.html#version-4-2-1\n\nChange-Id: I57a7649fde1cceda787622aa1975f3a7100aeb29\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0d20514c6ebbfe6b42b989b6192ae77312d3ba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0d20514c6ebbfe6b42b989b6192ae77312d3ba7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
