)]}'
{"id":"openstack%2Fdesignate~756112","triplet_id":"openstack%2Fdesignate~master~Ibefa12ddedae14a93e9ab914f62ce8247e0ee4ce","project":"openstack/designate","branch":"master","topic":"cleanup","hashtags":[],"change_id":"Ibefa12ddedae14a93e9ab914f62ce8247e0ee4ce","subject":"Remove unnecessary lower constraints","status":"MERGED","created":"2020-10-05 15:32:17.000000000","updated":"2020-10-12 13:54:04.000000000","submitted":"2020-10-12 13:52:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756112-1602510748775-2e8b7072","meta_rev_id":"d685d53fdcb6ee0050445a2f04175b246c698115","_number":756112,"virtual_id_number":756112,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 13:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"recommended":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"all":[{"value":1,"date":"2020-10-12 01:38:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":2,"date":"2020-10-12 12:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-10-12 12:55:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":1,"date":"2020-10-12 12:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 01:38:26.000000000","updated_by":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"reviewer":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"state":"REVIEWER"},{"updated":"2020-10-12 12:55:10.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"},{"updated":"2020-10-12 13:52:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bcd5631675d5a1f3aeb1098fc8db9b8ecfa26e71","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-05 15:32:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5608fc3f8895c88e99367c977098102b15269c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:41:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3de6f8831f6e41dc82c035198d6e2821 : SUCCESS in 41m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5b25403bac0042c28425db9878b43d8f : SUCCESS in 42m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7a9ad0886f914a75ac3b52ca0268fe09 : SUCCESS in 58m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e92190b983434b38b54cf8adf4cfc024 : SUCCESS in 40m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/319fa4ed0daa448497b8a4a08bf02b92 : SUCCESS in 42m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60121d8218f0431aa5a5479973d54f5c : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e3cb7885238421cae7591dfd0ed48cb : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29d694c3b6db4ba19161771d2c369a6a : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbbaa769fd334d8cb49471b03165954f : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/860a42560970436ba1419bfab19d0fcb : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2ba1ee6e71419c87817843cdb54ca7 : SUCCESS in 6m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/015ac1ba237141559dcb63cf631247f7 : FAILURE in 6m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/92b0f8cd5ed44f6f99cac6815a5538d9 : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0c1b7ddff3dddabd1e1769828373e2282711d61f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-06 13:14:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aeb470f73baceb481bc5688848e27cd950eb0dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 14:22:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8297ce2563f44d658f2603227df1fca0 : SUCCESS in 47m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/29068b2ff1af42788eaefac851b92feb : SUCCESS in 44m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8034a993398847aea420600db0e93624 : SUCCESS in 59m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/713a43492c9f4ccfb26b16f91b284412 : SUCCESS in 37m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b3a06b3fcf214af6bbfdbaa70ba4bcb9 : SUCCESS in 41m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57b67329cc0b4556960b13703c18867f : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7557886cd0a54229abffc26d57f6d715 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00d0dcce8f1c4dbf85ff6bb2117adf84 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a0ff318b56141bb8086f003a4a2b039 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/898d9aedff1d4cfd8ccd63bf85d72eae : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/853017c7008a43b9974b2277516f8625 : SUCCESS in 17m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6f11ef56573845ef854614bb7d94768d : SUCCESS in 2m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/06b3f73a482e4a0c9791cdde6216fd77 : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c9090f89b29f497661b912dbbe267608d5809508","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-10-12 01:38:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"672d1ea51b3aa2b4178c88041687b595bb5b8313","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-12 12:48:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d42545ea4e601761c5cba5f3bc27ae3c5831766b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 12:49:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d289a459d4bf755a86fe313131334f7e98f18eb","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-10-12 12:55:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"83fabde81c2452b3e3d020431441254df4c8e1e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:52:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fc46077f2eeb4d8aa2eb550d6afdf633 : SUCCESS in 41m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e9c469d46cfd4e5aafcaf4d3aed3e2cc : SUCCESS in 38m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/52a23fc3a131404f94192a6f885697ab : SUCCESS in 55m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/806151aa60eb49e9a548dfa1680f80ac : SUCCESS in 37m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f1fa23bb807f466c87ac920c05437bb8 : SUCCESS in 1h 01m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59f01683688c41f39d3d8ffee600b359 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ee721697e247fd9a49f16a982417d8 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6de1377eb5ac4a83b2c85ba4599a108a : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/548dbd4e7a1b4c51b10df407424b9f3c : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95cd6bcdb1cb473fbd890f4e6dde7c92 : SUCCESS in 8m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/70e6ca44b6734cf198e911169a34e712 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":2},{"id":"e03a792a5c800167aded7550e5e6848a8fb03bcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:52:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"17e0586975e9a17cebdfccd06a50c72acb97e6b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:54:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e592fbfc50994d5fb7b27c06076b7c6a : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"348771951fa6e9740b629e81227cc0311751a471","revisions":{"06a29756f526bf459004d29ea3dc25c086676741":{"kind":"REWORK","_number":1,"created":"2020-10-05 15:32:17.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/12/756112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/756112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/756112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/756112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/756112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/756112/1"}}},"commit":{"parents":[{"commit":"19da5d79d5d24dfda31773f70971f640856b10b1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19da5d79d5d24dfda31773f70971f640856b10b1"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-10-05 15:31:42.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-10-05 15:31:42.000000000","tz":-360},"subject":"Remove unnecessary lower constraints","message":"Remove unnecessary lower constraints\n\nThis change removes unneeded lower constraints.\n\nChange-Id: Ibefa12ddedae14a93e9ab914f62ce8247e0ee4ce\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06a29756f526bf459004d29ea3dc25c086676741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06a29756f526bf459004d29ea3dc25c086676741"}]},"branch":"refs/heads/master"},"348771951fa6e9740b629e81227cc0311751a471":{"kind":"REWORK","_number":2,"created":"2020-10-06 13:14:58.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/12/756112/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/756112/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/756112/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/756112/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/756112/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/756112/2"}}},"commit":{"parents":[{"commit":"19da5d79d5d24dfda31773f70971f640856b10b1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19da5d79d5d24dfda31773f70971f640856b10b1"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-10-05 15:31:42.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-10-06 13:14:50.000000000","tz":-360},"subject":"Remove unnecessary lower constraints","message":"Remove unnecessary lower constraints\n\nThis change removes unneeded lower constraints.\n\nChange-Id: Ibefa12ddedae14a93e9ab914f62ce8247e0ee4ce\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/348771951fa6e9740b629e81227cc0311751a471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/348771951fa6e9740b629e81227cc0311751a471"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
