)]}'
{"id":"openstack%2Fdesignate~758266","triplet_id":"openstack%2Fdesignate~stable%2Ftrain~I00ae6227edf21007c39a7d55221898ae5c10dda2","project":"openstack/designate","branch":"stable/train","topic":"bug/1880230-stable/train","hashtags":[],"change_id":"I00ae6227edf21007c39a7d55221898ae5c10dda2","subject":"pdns4 backend: check if zone exists before attempting delete","status":"MERGED","created":"2020-10-14 23:39:06.000000000","updated":"2020-11-03 11:04:36.000000000","submitted":"2020-11-03 11:02:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758266-1604401361073-846bd11f","meta_rev_id":"c86d938cb89e1e26cdcd8704ad7ed6063154ff7c","_number":758266,"virtual_id_number":758266,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-16 12:46:49.000000000","_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-03 11:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-11-03 05:37:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2020-11-02 21:16:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-11-03 05:37:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-14 23:39:06.000000000","_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},{"value":2,"date":"2020-10-27 22:21:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 23:39:06.000000000","updated_by":{"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},"reviewer":{"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},"state":"REVIEWER"},{"updated":"2020-10-16 12:46:49.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-11-02 21:16:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-03 05:37:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-11-03 11:02:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ec2092645ef3a4797bab4db17ad69ded85d5c0d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-14 23:39:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5c391f15856127f0a90db22ea0aabe081202638d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 01:00:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d604db3ad9fd4396bf32d8833807cc0d : SUCCESS in 36m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e68e58bbb75846c5abd47e4ccc3961c7 : SUCCESS in 41m 19s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/d7635aaf63f648028d5f1fadaf468c0d : SUCCESS in 39m 01s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/0134e2184f084cd1bdcf255ae9b36011 : SUCCESS in 1h 01m 30s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/191a11e269e24181a4dde0bb5f3d5e56 : FAILURE in 34m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ee2cd80f43d341dca679e471e95bb6bf : SUCCESS in 1h 13m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/601a9d80c787415ab2e8ee0d189d5dfa : SUCCESS in 43m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bb5b25aafc7641b5bba15e96b07f138b : SUCCESS in 48m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64766a18ef7e475797e62202decc09ca : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6cdeefc0fc44cd68a97beaf6046557c : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed25d44becd84b34a9e230e5670e74a1 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2baf69b4a8d44a46b63ede9ac1eb279c : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b31119be821c45409cedc6a38bf7615a : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc79ef26219d4676b62c4a316ab608f3 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ae05c735b045f7b124508f84ec930b : SUCCESS in 7m 34s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/d7dec35f48b74ada8c60fd64e356dee7 : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8fca70e56d41704b397aeb8998cdb0bf3218c8c1","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-16 12:46:49.000000000","message":"Patch Set 1:\n\nNot sure what to make of that postgres error :/","accounts_in_message":[],"_revision_number":1},{"id":"33fe30ad88d09e165e36418681f1d821fb12ede6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-16 19:05:35.000000000","message":"Patch Set 1:\n\n\u003e Not sure what to make of that postgres error :/\n\npostgres is broken :\u0027( unrelated to this change. We really should fix it, but I don\u0027t know enough about postgres to fix it.","accounts_in_message":[],"_revision_number":1},{"id":"13d00a85d5366d522d11990558b41cf93ac2abd5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-22 01:21:27.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"01aef11e7ac0384ddcc96dea2e4bc08757698e91","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-27 22:21:53.000000000","message":"Patch Set 1: Review-Priority+2\n\nBumping review priority on this bug fix. This is an incredibly frustrating but that usually requires the user to modify the database to fix.","accounts_in_message":[],"_revision_number":1},{"id":"713f344dff698f1a50cb8ae36b33891b7b1315e2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-02 21:16:27.000000000","message":"Patch Set 1: Code-Review+1\n\nThis looks ok. I might have just added an except block for the 422 response to save a round trip, but this is fine.","accounts_in_message":[],"_revision_number":1},{"id":"3d55c6c1c5f275c9414d4d9e09bf54c2070d846a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-11-03 05:37:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77a4a0e66a4fb2d52d19f0206156e24e7fb6358b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 05:37:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1be7c86c851a83949acc7f23e9d4a1d48553bb34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 06:49:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9339d9fe72bb48238916fc689feb5c8d : SUCCESS in 31m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b8bbdb24219c46239851507bcd513281 : SUCCESS in 43m 06s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/0ddcea0055844cbdb15eddee9680ee2c : SUCCESS in 43m 27s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/ae4bfc8aa3154ade8e2ec98db9716959 : SUCCESS in 40m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1a8ad6a33deb4ca9a87ef0ef81172161 : SUCCESS in 1h 03m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5be75386d930406686ae708f9bef9906 : POST_FAILURE in 35m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e7e6bd8c9e504d98a5e49b8bcdf9b33d : SUCCESS in 43m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2df6c6782ba408cbc7e443b7cdb9dd3 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1bf36cc4c7644fba54210f20ad37d16 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1a7ca8417df46bdacdd4685517f256f : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/562d9d1698994a4eb6735fefa66e3b4e : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b24ae512567a4f0ebf476b7ea08a3f62 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04998c6df8574d6591c0d8082b16f571 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4cd67ad4118ed28f4716f916c7b5ea083dfc3233","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-11-03 08:41:47.000000000","message":"Patch Set 1:\n\nrecheck unrelated mirror failure in post","accounts_in_message":[],"_revision_number":1},{"id":"f676ccd8bfb1a9b8587fe321db260bd74982f28a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:49:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0663ac4fd74c4b278cc5ab59cd803830 : SUCCESS in 39m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1714dc7d79b445e785447777b16613c0 : SUCCESS in 40m 59s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/420d6a20d85848448f491c2c80adb78e : SUCCESS in 46m 44s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/9b5d7996d1a1458881eb8d640300ea39 : SUCCESS in 42m 41s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/025031533b5e4106a12159f91324fb30 : FAILURE in 29m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/761b6160f2104f45a777e24cfb91fc5d : SUCCESS in 59m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/342042aee0634a96a1588d815c172bd4 : SUCCESS in 40m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7ab580214a5c4103911f76e18810de9e : SUCCESS in 40m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaa591efbbd5483d87d846f779c6fe5b : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db120c9ec11448b5aeb9fb0d6ccbaa7f : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/800fd24f4d8e4e99ad1dc2e505dc43ba : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1af19a76ed84c59beef42fafb496e21 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2510512b7fbc4e718fb0be1021c2a486 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ffb74325475249cfbe39f188baed4d38 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c6f707e0a54b4c84ff22267f61c221 : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/53e76321ec3643c8ac6ee6dad8f90a24 : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":1},{"id":"317294cc98273afe54fb34d947cc34513cdbfb2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:49:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50d49a5f57e9f6a6094bbb23d41bf455cbc315c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 11:02:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6e5f494e58e347598bd93fee0ddfef30 : SUCCESS in 33m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/52421166c9514b289e8da0c9b33637e7 : SUCCESS in 31m 32s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/3665255c82434abf87de328df387debb : SUCCESS in 46m 58s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/8d79fc93ddde4a9791815d5fee68c103 : SUCCESS in 41m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a74e76ed45ae4784b80366e1c839a63e : SUCCESS in 1h 05m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fe23a30ab6374fa7a42edf3d9a5747ed : SUCCESS in 46m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2b5cb5a3e9694abd91967a398a12d28f : SUCCESS in 51m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11dbd3abba574fa6b9456e00a1f5f963 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8effcaeaf2947679c9af8943196b813 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6994740646aa4ae8b674b5cb98a5a225 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24ec5ad6f3d54ad9bd7fc50dae850b34 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2158a66f52d42389550ca6e1c6b4e87 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc8279f334d64dc5a48435b7e476cc57 : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":1},{"id":"87f30589a0b46ea79631011867328f73c0e8dbbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 11:02:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"762c30839ecd7d185c907bc4c476dea59f939306","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 11:04:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6312568e83849b6aac6407e9a686471 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0dd12f3746e52dba2e6cc78984525845c473e836","revisions":{"0dd12f3746e52dba2e6cc78984525845c473e836":{"kind":"REWORK","_number":1,"created":"2020-10-14 23:39:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/758266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/758266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/758266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/758266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/758266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/758266/1"}}},"commit":{"parents":[{"commit":"4fc04ee416d13f7e4e27c6b522b42a3c69c271b5","subject":"Merge \"Adding distributed locking to central\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fc04ee416d13f7e4e27c6b522b42a3c69c271b5"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-01 15:04:12.000000000","tz":-300},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-14 23:39:06.000000000","tz":0},"subject":"pdns4 backend: check if zone exists before attempting delete","message":"pdns4 backend: check if zone exists before attempting delete\n\nThis should prevent us from obsessing over zones that are in the\ndesignate DB but not in the pdns backend due to races/incomplete\ntransactions/etc.\n\nBug: 1880230\nChange-Id: I00ae6227edf21007c39a7d55221898ae5c10dda2\n(cherry picked from commit 5acba9443e2fc1f6e78f9a32cd85eb168d2a7cb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0dd12f3746e52dba2e6cc78984525845c473e836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0dd12f3746e52dba2e6cc78984525845c473e836"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
