)]}'
{"id":"openstack%2Fdesignate~758326","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I57a7649fde1cceda787622aa1975f3a7100aeb29","project":"openstack/designate","branch":"stable/ussuri","topic":"lc_vine_five-stable/ussuri","hashtags":[],"change_id":"I57a7649fde1cceda787622aa1975f3a7100aeb29","subject":"Increase lower-constraint for kombu","status":"ABANDONED","created":"2020-10-15 03:50:23.000000000","updated":"2020-10-15 17:32:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"350023f0cdb8511eb7024003e58b2babe226d51f","_number":758326,"virtual_id_number":758326,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-15 04:58:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-10-15 03:50:23.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 03:50:23.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-10-15 04:58:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"384b83d07ef66a54bf40f0f9f5b31f7734289492","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 03:50:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bd8ca8433ad47982fe42cb4054a366dee6004490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 04:58:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1fb910bcc32347688966e5ec0fbefde1 : SUCCESS in 35m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/025efcb7371d48eb9c38790ff285d740 : SUCCESS in 33m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f8da8547e1b14695833b0ffb0fbdbff4 : SUCCESS in 1h 05m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a822b34fe2684675af0e04cfa519ae08 : SUCCESS in 35m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/783a66b88c4c4dc19bdb728a4f59afc6 : SUCCESS in 57m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab43ee5bb9304035ba9756be7ac98bd2 : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80589aa6773b4b9a9b4d91c617ecea8d : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c1f2d989d0d4159a7569112d54fc26d : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ca4a888cbae41fe82000b3789e76418 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f55ff7b951a24ed08b56cf8bcaab4114 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1cb3171999412c8bf680fde8e04526 : SUCCESS in 7m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/30a7e49fcf434e04b3ecfa872b4c254c : SUCCESS in 2m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09fdf32aa6e94f4fb9d5f29674c38113 : SUCCESS in 30m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3de9d88672bd99c654770aef7d358731f4e07201","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 05:20:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"a0d051c7a93c48ef48b5b3d46652cde82fcd82d1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 17:31:16.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"2f87dddaa9b085bc7fa6c78bc166bbe3a20d4a66","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 17:32:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"490070d637b00d4138f9ed2dfbfccd10e2400444","revisions":{"490070d637b00d4138f9ed2dfbfccd10e2400444":{"kind":"REWORK","_number":1,"created":"2020-10-15 03:50:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/26/758326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/758326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/758326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/758326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/758326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/758326/1"}}},"commit":{"parents":[{"commit":"d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-30 18:10:26.000000000","tz":-360},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 03:50:23.000000000","tz":0},"subject":"Increase lower-constraint for kombu","message":"Increase lower-constraint for kombu\n\nThere is no module named \u0027vine\u0027 on kombu\u003d\u003d4.1.0 version [1]. Bumping\nto kombu-4.2.1 (and not kombu-4.2.0) [2].\n\n[1] https://github.com/celery/kombu/issues/851\n[2] https://docs.celeryproject.org/projects/kombu/en/stable/changelog.html#version-4-2-1\n\nChange-Id: I57a7649fde1cceda787622aa1975f3a7100aeb29\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit a0d20514c6ebbfe6b42b989b6192ae77312d3ba7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/490070d637b00d4138f9ed2dfbfccd10e2400444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/490070d637b00d4138f9ed2dfbfccd10e2400444"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
