)]}'
{"id":"openstack%2Fdesignate~758338","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~Ie05efab2f52fe3e1641c9863f6f85c813158e748","project":"openstack/designate","branch":"stable/ussuri","hashtags":[],"change_id":"Ie05efab2f52fe3e1641c9863f6f85c813158e748","subject":"Fix lower-constraints test not constraining properly","status":"MERGED","created":"2020-10-15 05:19:37.000000000","updated":"2020-10-22 02:46:25.000000000","submitted":"2020-10-22 02:44:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758338-1603334684086-2c452473","meta_rev_id":"8061cdd2a160560f6c3fcb1590535bb16a3b3708","_number":758338,"virtual_id_number":758338,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 02:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-10-16 12:32:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-10-16 11:33:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":1,"date":"2020-10-16 22:12:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-22 01:20:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-10-16 11:33:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 11:33:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-10-16 12:32:51.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-10-16 22:12:46.000000000","updated_by":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"reviewer":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"state":"REVIEWER"},{"updated":"2020-10-22 02:44:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca277928f1b23144d054c2e228b1d5ee524fa060","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 05:19:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"be2e08298c69bf7888375ebc1df18cc69900f927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 06:45:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/675ef5ad921648cc9f4d5ec92e51b985 : SUCCESS in 37m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b45a55280f61404b9291028ff60c4f0a : SUCCESS in 41m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6aee4d1af7ab4bcea32a7055f8f32afe : SUCCESS in 1h 18m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ca5d6813f4db44399fcd646ad4fa28a4 : SUCCESS in 41m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eb52a30d30a440f7b4cf67f275ff6bfd : SUCCESS in 42m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f556285bbc0548d9a022568d290c149c : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fd161d7caf3417782fdad1e99373dbe : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d636d99a92b645dc946a36ddbd7fe0c2 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90d64f834b8e4bb5841fc75ffedbc53f : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b08d702e469d486d8959d8b708a9e8ce : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f033329fc1c24caf864df8e2ad85f9a4 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2df648bc299b4d86944fb05c244584d3 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3db30158178d98bcc46985da74dfbc40aff667d0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 06:57:01.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"bf91f72cb0cb67ee1136c40faccc93efdaa4bce6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 07:09:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cadf5d0ab57deaf2f5e3f3d1e89013ae28c188ab","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 07:19:37.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"c73e0e332a817029aec91aba0aa7c6a5e96bfbd5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 07:42:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"05b87c9d0758d52bcdcd938b25d53a8ff95aff96","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 07:54:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1f9966977cceb4c3f8d3a3928d965eb280bcd4f2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 08:25:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"85902599fa352cb1fb4d928762f0bec3f6da882f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 11:28:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f338a7ff30df4920929b2dad8226864f : SUCCESS in 45m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/66b12222454240c39c47ac5aeecc3030 : SUCCESS in 38m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/afc58ab3b643412eb5321d2f10b5c48b : SUCCESS in 1h 14m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1e80f138b58044bba2a070945d96309a : SUCCESS in 37m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4627a30cadf248338ac33a5afea3810d : SUCCESS in 37m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fdb3943f77244d0b1cbf30d60b6cab6 : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78b4c7fec55e47a794c4aa9b5f61c866 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9278d6fc270843bb99b83ec86f681940 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f07dfabc14e34d03acdecb3064e27b39 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07604848f0e34694927df1ba86e3b43e : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d5122470b9549b79d472b85e8523392 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4752c7c616474ae994cfc5293722d238 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":5},{"id":"0631ea00a265424599553ab8de4b542e4d6a2151","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 17:49:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5},{"id":"d9eca905b5c278022762ec66f9db927ce3f341b1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 18:59:55.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"2ea03f76737de3259f848865033c62ac350ceaa9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-15 19:01:18.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"1a0984c6fc62e32a1a1e249d49e3a509d496ba95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 20:23:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/831a787a2de94153bb244cd047f9648b : SUCCESS in 41m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1aa9fc183bb7420c9d5aacf445db5497 : SUCCESS in 34m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/52b4285377ba418b9714524a3715fad1 : SUCCESS in 1h 13m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/766adf928d1844a19caac54ba93daa47 : SUCCESS in 40m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e91984aee8e1489cbfb5bb110431b4eb : SUCCESS in 40m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7d89f7e93bb4731ae0bb1aeb66e26ba : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3960e19ba7148b9aac7e5c4b89bf0e0 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6140f1783f34408d8127ff01e6c6b09a : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/853fdd61fdbf43c98a4fea1e291716c9 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49df7d0bfd63401786536a99f2b0ebee : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45120521532d46a8984308b124fff4c8 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cbd46478e70452e987a39dbd52284b7 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":6},{"id":"7ea4a2dd653a2fcb685bad597bbdb4eaf8f19d15","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-16 11:33:50.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+2\n\nWaiting for patch to be merged in master","accounts_in_message":[],"_revision_number":6},{"id":"2cde4bf4365e7a43cfca47191e541309a2d628ac","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-16 12:32:51.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"402492cdec01f14559b947cc5a8a1fdd05a8a7c3","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-10-16 22:12:46.000000000","message":"Patch Set 6: Code-Review+1\n\nJust question:\nin https://pip.pypa.io/en/stable/reference/pip_install/\nanything you missed in your requirements file—and it’s easy to miss things as your project evolves. To be safe, install your project using pip and --no-deps.\n\nbut `-U` Upgrade all specified packages to the newest available version. The handling of dependencies depends on the upgrade-strategy used.\ndoes that mean it will get newest version","accounts_in_message":[],"_revision_number":6},{"id":"5717ef7fb2d64a1e08eb4b59a220d994db25ffb9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-17 01:18:14.000000000","message":"Patch Set 6:\n\n\u003e Just question:\n \u003e in https://pip.pypa.io/en/stable/reference/pip_install/\n \u003e anything you missed in your requirements file—and it’s easy to miss\n \u003e things as your project evolves. To be safe, install your project\n \u003e using pip and --no-deps.\n \u003e \n \u003e but `-U` Upgrade all specified packages to the newest available\n \u003e version. The handling of dependencies depends on the\n \u003e upgrade-strategy used.\n \u003e does that mean it will get newest version\n\nThe U (upgrade) might be redundant to be honest, the intent was to make sure that it always installs the latest version of Designate, while the --no_deps prevents it from installing any of the Designate dependencies, since that is handled before.","accounts_in_message":[],"_revision_number":6},{"id":"737529f27d720eecb8126672c60172bf98b9c4a4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-22 01:20:53.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1694fe746e24c2c05c2dcd17e25006094b1cdcf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 01:21:11.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c69a6f69f327e63d2c2ce07cec6cd083703ffc85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 02:44:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/de6217b4c8524fc1b1354e9794aa2fa6 : SUCCESS in 33m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0c5754c205894ef6a93cd1850b3672c3 : SUCCESS in 35m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/296c20cede6e4657a7e0801b84987a26 : SUCCESS in 1h 00m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9efb5346476c49598303028adcae2ea7 : SUCCESS in 39m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7ec7405242b94eeb8f553832d813fb4b : SUCCESS in 33m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5d4907e0ab14590b265fb387d36a092 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/705a6b804b9344d0b8000a284eeed7a2 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/603a70dee3f6463a9d315b04f3dc9b3c : SUCCESS in 7m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5076566bc450440da133ee7d1cebba37 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5675cebedc4268951b1aeee0ccafec : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1caf0ad566f94690ad1162a517309ad2 : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":6},{"id":"f6748cd5068d4a08f46bd441c3450e8b302ef0e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 02:44:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"8061cdd2a160560f6c3fcb1590535bb16a3b3708","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 02:46:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d5b6fb88ec4f0680526dee81f1e817 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f57155e498d4f08a889f25638ff055f : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"688987ab44e9dc167ba9d5928d72748313196fc9","revisions":{"6ffe5cfebb9708049978e9e905e485913cb8c530":{"kind":"REWORK","_number":1,"created":"2020-10-15 05:19:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/758338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/758338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/758338/1"}}},"commit":{"parents":[{"commit":"d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 05:14:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 05:19:37.000000000","tz":0},"subject":"Fix lower-constraints test not constraining properly","message":"Fix lower-constraints test not constraining properly\n\nWhen running the lower-constraints test we are currently\nusing pip install -e . to install the source. This caused\npython libraries to unintentionally getting upgraded beyond\nthe intended constraints.\n\nChange-Id: Ie05efab2f52fe3e1641c9863f6f85c813158e748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ffe5cfebb9708049978e9e905e485913cb8c530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ffe5cfebb9708049978e9e905e485913cb8c530"}]},"branch":"refs/heads/stable/ussuri"},"4d8801af542b7883b8c24e3f49682fd9144a4950":{"kind":"REWORK","_number":2,"created":"2020-10-15 07:09:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/758338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/758338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/758338/2"}}},"commit":{"parents":[{"commit":"d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 05:14:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"me@eandersson.net","date":"2020-10-15 07:09:17.000000000","tz":-420},"subject":"Fix lower-constraints test not constraining properly","message":"Fix lower-constraints test not constraining properly\n\nWhen running the lower-constraints test we are currently\nusing pip install -e . to install the source. This caused\npython libraries to unintentionally getting upgraded beyond\nthe intended constraints.\n\nChange-Id: Ie05efab2f52fe3e1641c9863f6f85c813158e748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d8801af542b7883b8c24e3f49682fd9144a4950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d8801af542b7883b8c24e3f49682fd9144a4950"}]},"branch":"refs/heads/stable/ussuri"},"322af6d6a7ed0f77ef20e0ef5c89f9f4de613c0d":{"kind":"REWORK","_number":3,"created":"2020-10-15 07:42:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/758338/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/758338/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/758338/3"}}},"commit":{"parents":[{"commit":"d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 05:14:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 07:42:36.000000000","tz":-420},"subject":"Fix lower-constraints test not constraining properly","message":"Fix lower-constraints test not constraining properly\n\nWhen running the lower-constraints test we are currently\nusing pip install -e . to install the source. This caused\npython libraries to unintentionally getting upgraded beyond\nthe intended constraints.\n\nChange-Id: Ie05efab2f52fe3e1641c9863f6f85c813158e748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/322af6d6a7ed0f77ef20e0ef5c89f9f4de613c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/322af6d6a7ed0f77ef20e0ef5c89f9f4de613c0d"}]},"branch":"refs/heads/stable/ussuri"},"8465e01dd2136c50e964bb465c489aeb337067f6":{"kind":"REWORK","_number":4,"created":"2020-10-15 07:54:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/758338/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/758338/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/758338/4"}}},"commit":{"parents":[{"commit":"d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 05:14:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 07:52:47.000000000","tz":-420},"subject":"Fix lower-constraints test not constraining properly","message":"Fix lower-constraints test not constraining properly\n\nWhen running the lower-constraints test we are currently\nusing pip install -e . to install the source. This caused\npython libraries to unintentionally getting upgraded beyond\nthe intended constraints.\n\nChange-Id: Ie05efab2f52fe3e1641c9863f6f85c813158e748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8465e01dd2136c50e964bb465c489aeb337067f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8465e01dd2136c50e964bb465c489aeb337067f6"}]},"branch":"refs/heads/stable/ussuri"},"690d1562e89f5b78f15bc3541bf59146ddc50a4e":{"kind":"REWORK","_number":5,"created":"2020-10-15 08:25:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/758338/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/758338/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/758338/5"}}},"commit":{"parents":[{"commit":"d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 05:14:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 08:24:51.000000000","tz":-420},"subject":"Fix lower-constraints test not constraining properly","message":"Fix lower-constraints test not constraining properly\n\nWhen running the lower-constraints test we are currently\nusing pip install -e . to install the source. This caused\npython libraries to unintentionally getting upgraded beyond\nthe intended constraints.\n\nChange-Id: Ie05efab2f52fe3e1641c9863f6f85c813158e748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/690d1562e89f5b78f15bc3541bf59146ddc50a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/690d1562e89f5b78f15bc3541bf59146ddc50a4e"}]},"branch":"refs/heads/stable/ussuri"},"688987ab44e9dc167ba9d5928d72748313196fc9":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-10-15 19:01:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/758338/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/758338/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/758338/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/758338/6"}}},"commit":{"parents":[{"commit":"d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 05:14:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-10-15 19:01:06.000000000","tz":-420},"subject":"Fix lower-constraints test not constraining properly","message":"Fix lower-constraints test not constraining properly\n\nWhen running the lower-constraints test we are currently\nusing pip install -e . to install the source. This caused\npython libraries to unintentionally getting upgraded beyond\nthe intended constraints.\n\nChange-Id: Ie05efab2f52fe3e1641c9863f6f85c813158e748\n(cherry picked from commit 568b559b67b0d9a663172d681fe887bf06feea0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/688987ab44e9dc167ba9d5928d72748313196fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/688987ab44e9dc167ba9d5928d72748313196fc9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
