)]}'
{"id":"openstack%2Fdesignate~759174","triplet_id":"openstack%2Fdesignate~master~Iaf95ed023d5d665cb7bc885b931fd29ccffe1c9f","project":"openstack/designate","branch":"master","topic":"tox","hashtags":[],"change_id":"Iaf95ed023d5d665cb7bc885b931fd29ccffe1c9f","subject":"Update requirements URLs in tox config","status":"MERGED","created":"2020-10-22 03:04:35.000000000","updated":"2020-11-12 08:56:09.000000000","submitted":"2020-11-12 08:54:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759174-1605171279964-28503777","meta_rev_id":"4840458b2bb823ea92dcc42adc41a5a21daa8d28","_number":759174,"virtual_id_number":759174,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-10 12:57:25.000000000","_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-12 08:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2020-11-09 10:12:49.000000000","_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":2,"date":"2020-11-10 14:24:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-11-12 07:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-11-12 07:47:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 20:44:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-29 14:16:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-11-10 12:57:25.000000000","updated_by":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"reviewer":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"state":"REVIEWER"},{"updated":"2020-11-10 14:24:16.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-11-12 07:47:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-11-12 08:54:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cbd993a840c75ed2a972500d08262faf17490d5","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-22 03:04:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3adb3910fc283f7bdc27e38b88e2520b68dc232f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 04:10:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cebddc3312fd4b278b9a5ae26c1a76a3 : SUCCESS in 5m 24s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b1b35643495f4ce2a785ec0c41f5c67e : SUCCESS in 39m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9df4d57d55eb488fb59d1e42ce6a9a24 : SUCCESS in 40m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/47c6574024cb416288678683259386f1 : SUCCESS in 46m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/44df94a88a86497ba37a01881bc9ed57 : SUCCESS in 38m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fb99d5abccdb405eb3ece91f8e1e3b0c : SUCCESS in 40m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7869c3e995984242a4f7259339400999 : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/316cd6b5dd1e46e5a7fe36285d315367 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e6f26124bc243cf849125526b49b0d5 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/344c85b5c7be4ebba3247f18ef9adaed : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95312642f4d24cb19c2ecb114209d271 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c8effb055d4e78a7ccdf83f82b7c13 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3472dba87c4648d0a51b6a5865f0daa5 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1b97bf1f3213d6732a250eabb8463635e2947649","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-22 07:17:36.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d6d952a406b40ec017999e592993a40c513fd355","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-28 03:01:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a77aa7e9061d3709147a6b9576c8131ee6c7fab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 04:13:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a94769de17274fa489a9516a2ac73650 : SUCCESS in 6m 18s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9dc5e442a1b24783ae15f14bff28f85c : FAILURE in 42m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c3b3d2b6884146108013b1daf55234d0 : FAILURE in 36m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b38d2714229147a592683d2414ae6965 : FAILURE in 47m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f0eff776be554dfb8b72ae79fec6f158 : FAILURE in 25m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/df02d400400043d7b1260d0c1dea7f7e : FAILURE in 28m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77ad25f797b34693aaf45a51c8a2cb1b : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45d6b9e819244fb29760d46ff85b3c9c : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544417c6427249e082d4fba83913b1ba : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f3871d1099a49f5a1c6f7d1c3ec6f39 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e08dd830484e457eb247098cff947ea4 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f946ab03f2348a58afc6aee457d87d8 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea572ac484844bbb800ee95020d3d3ba : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"a85a1f9cfe94cf2bd334a4fc9025a90b14400414","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-28 20:44:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ee82504245cf1f29829f3168d7da4e264037bd08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 21:48:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/23537905b360455994423bdf2a0749ee : SUCCESS in 3m 34s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7205b6c2f9d341039d5c6cfe9b61b6bf : FAILURE in 35m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0eb618e6a8874ce6bc8d8a0a40bc33ff : FAILURE in 27m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/763c5e76d5b745aa90bd22c68a7b7f6d : SUCCESS in 53m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/46bafb969fcf495fb838915d59190b14 : FAILURE in 55m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d91cca577ac9415296e7067f55a11477 : FAILURE in 35m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19c37fc6fc9c4e81b63f3e262c265ab1 : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6f802d046c94848bbe6ede5e3d964cf : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a49171121844a399b9ab4299bbb698a : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0087863e5f1e4cb1a3a477fc14e19480 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb422608cacc4ee1be13f878a7ab22d1 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffeac0b8a622416eb3ff686db74b7468 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19b5ba1cf3a64864872b2f1858ba471b : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"b28330631c7a8cc6e946bba320819f455b559895","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-29 14:16:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00c7472a88f85e05f52e9416c73e18149d582c2d","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-30 03:57:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a81521f35f822b423b50c2f88c06f1a5937ec69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 05:06:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dc37f59f253643f08a91b25431bbd21a : SUCCESS in 4m 10s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/825fd1b6e1594b189b0b7c88e4183946 : SUCCESS in 34m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/43983a2a90bd4e049f5222bdc2e28302 : SUCCESS in 43m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a172e6441cb6461fa9db7048b5171176 : SUCCESS in 57m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b42018a613ef4f38a3aded42e638f897 : RETRY_LIMIT in 20m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cbd5e37813fb4d66886b00c3065b434d : SUCCESS in 34m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/441523ab7dd543a6bb3d0145fdcfa681 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8420b696d82949bf97e62f05ebd9df4b : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc2ab74d91d44d4095ba06d84fe29d74 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c18c1995e0ae4988bfa00a7aaad3174d : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d138fc9914464ce48f986a7a9e024454 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1518ea58b9747588f4ec1778fa56f98 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54b84425a88740afb77398c58a806db9 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"8d42ad28f0ba1b1863f7297a9be42211e05ba773","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-09 10:12:49.000000000","message":"Patch Set 3:\n\n(4 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"72ba15ba378af293fa283bf1577699be20a5bd71","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-11-10 12:57:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"09cc92691a706e20dbc4d8070675ea8034e88e46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 13:56:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/21363bd693d647e2ae619cf91b1fd24d : SUCCESS in 4m 23s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cc82732899234b6a970da3ca61b69a64 : SUCCESS in 41m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/81e667b92c844a4e8348fb0bd800f61f : SUCCESS in 34m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f9204c0f2e1742fc88bb8017272ce764 : SUCCESS in 49m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2b9686e00720498c9f3f682c0735f4be : SUCCESS in 44m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8d409c2cf89e40a3a4c951fe36f14c76 : SUCCESS in 48m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55d8f872cbb249d797617f22b92ec9bd : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20e40434602f4fa7acfdfcd9485d3a29 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39e2e147aaea43c69acd4f6729bc6d4d : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b3a5c729c134c8ca411d98c20cf3acc : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e524aa97e5a54fbbbe62d344c2c04bd6 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a82711f125384554b9275215f1190f52 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa27482ed45b4f50963d66bda6ee01b1 : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/102cb43e36f44e2b9458b86f91984852 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"cbb3f9eb83513af1e3f9ec4e959f47ea0d9ca3de","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-10 14:24:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0937c5644a92b87af45f9cf73b359c5d2568271f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-11-12 07:47:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"40258d7290a964eb0ea5930e60bf3142c4108e72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 07:47:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ac7ecfc3b769a9df724473404a8c3473a90e9c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 08:54:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8619abaddbb846e9a2c304f06b2e9172 : SUCCESS in 3m 24s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/325851732f344bd0bb081220e215869b : SUCCESS in 41m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2a8520b6814f40488a6e96b087cac5ff : SUCCESS in 51m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e077bf2f113e4b39a42e3f4f8dda5d6a : SUCCESS in 58m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/29833da3d27748de83552defecf51821 : SUCCESS in 43m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ae3bf9334c5e4f1fb869681cda2b2818 : SUCCESS in 39m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cf8ff52617d4cee94c11df301069114 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2f7237eaff04eb18611dad516b78f25 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ea7173ecfea42f499bdf7ef8ae355cd : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b707fc5d6b4c420a899e19ded819d643 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8930950c164b4794981ad00548521ce8 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40027448308d4ace8713722412c7cb1a : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"4e4e3b1b33e9567ad9396bc3e02c1a36b80ba3d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 08:54:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e460a2dfc4211c4b3177f5073719a04c69649f7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 08:56:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ebbda121f6124b3cafe22106db7e22c9 : SUCCESS in 52s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/befd4656982b43289a8b430681cc38bb : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22dc9d19390b43c5be69bea906e15e20 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8834e8b2f91fe494f072ae6675657d687ee67713","revisions":{"c0f868d5629551ef82f8eec455f84088089c62c7":{"kind":"REWORK","_number":1,"created":"2020-10-22 03:04:35.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/74/759174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/759174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/759174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/759174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/759174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/759174/1"}}},"commit":{"parents":[{"commit":"8459ed9460ab492adfd305875c717ebe31bf68f3","subject":"requirements: Drop os-testr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8459ed9460ab492adfd305875c717ebe31bf68f3"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-22 03:03:25.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-22 03:03:25.000000000","tz":480},"subject":"update doc url","message":"update doc url\n\nChange-Id: Iaf95ed023d5d665cb7bc885b931fd29ccffe1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c0f868d5629551ef82f8eec455f84088089c62c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c0f868d5629551ef82f8eec455f84088089c62c7"}]},"branch":"refs/heads/master"},"8a9bddd09d966f26a8ddb401f65b34844592d2c5":{"kind":"REWORK","_number":2,"created":"2020-10-28 03:01:00.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/74/759174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/759174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/759174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/759174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/759174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/759174/2"}}},"commit":{"parents":[{"commit":"8459ed9460ab492adfd305875c717ebe31bf68f3","subject":"requirements: Drop os-testr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8459ed9460ab492adfd305875c717ebe31bf68f3"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-22 03:03:25.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-28 02:57:43.000000000","tz":480},"subject":"Update requirements URLs in tox config","message":"Update requirements URLs in tox config\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so will switch to the correct\nupper-constraints list automatically when the requirements repository branches.\n\nChange-Id: Iaf95ed023d5d665cb7bc885b931fd29ccffe1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a9bddd09d966f26a8ddb401f65b34844592d2c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a9bddd09d966f26a8ddb401f65b34844592d2c5"}]},"branch":"refs/heads/master"},"8834e8b2f91fe494f072ae6675657d687ee67713":{"kind":"REWORK","_number":3,"created":"2020-10-30 03:57:35.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/74/759174/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/759174/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/759174/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/759174/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/759174/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/759174/3"}}},"commit":{"parents":[{"commit":"8459ed9460ab492adfd305875c717ebe31bf68f3","subject":"requirements: Drop os-testr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8459ed9460ab492adfd305875c717ebe31bf68f3"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-22 03:03:25.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-30 03:55:01.000000000","tz":480},"subject":"Update requirements URLs in tox config","message":"Update requirements URLs in tox config\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so will switch to the correct\nupper-constraints list automatically when the requirements repository branches.\n\nChange-Id: Iaf95ed023d5d665cb7bc885b931fd29ccffe1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8834e8b2f91fe494f072ae6675657d687ee67713"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8834e8b2f91fe494f072ae6675657d687ee67713"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
