)]}'
{"id":"openstack%2Fdesignate~759609","triplet_id":"openstack%2Fdesignate~master~I0ebe726eec0fb7c7578c8d08d09bed5c9d306709","project":"openstack/designate","branch":"master","topic":"test","hashtags":[],"change_id":"I0ebe726eec0fb7c7578c8d08d09bed5c9d306709","subject":"DNM: Debug gate issues","status":"ABANDONED","created":"2020-10-26 05:32:03.000000000","updated":"2020-10-27 15:47:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6e53d5898cc0fec9055ee58be5294edabda8eb0f","_number":759609,"virtual_id_number":759609,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-27 10:37:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-27 10:37:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2fb14879b37b8f65516cdd62b9610c0d8d7770d8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-26 05:32:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db60bf08d9cfffc18698e153ce5e66235015edf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 06:22:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b94aeb59c0404e688bfee216ca87a400 : FAILURE in 33m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cb0c704e76ee4df2877bf0fad626584f : FAILURE in 29m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8a72c63673754be195fc24e3bffdafa3 : SUCCESS in 49m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/997b03ff20414c6ca060fe0bc8d65720 : FAILURE in 30m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fe1b4557d01040a3b129884e33861d69 : FAILURE in 28m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/458774fb353648238ec5928ce1759688 : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7468336a320c412a9f8fc88b564713de : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcf02e63b43840ec89223a09f82b5b99 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3050d0990fce4a3dac73cb3a6ba3c766 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65d31404dbd447019e6987a5d7c1a2c7 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc13471235946d9b296badcabfdb4ab : FAILURE in 5m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/679640d0ffec4f4283a29ee9528f6533 : FAILURE in 28m 31s\n\nWarning:\n  Comments left for invalid file doc/source/admin/blacklists.rst","accounts_in_message":[],"_revision_number":1},{"id":"d3ae4efde8024d48b64b55fc1e064e7471dddb70","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-26 11:21:06.000000000","message":"Patch Set 1:\n\nrecheck another eventlet bump","accounts_in_message":[],"_revision_number":1},{"id":"9a1e899ad840299a417577331f3ac1b8c9d95697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 12:25:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5605c59b816449ba80c66b259d15e5a0 : FAILURE in 34m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2914136c4f184d40b2b10fd32f857066 : FAILURE in 34m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c313bbec55f14bcdbc212f31310250de : SUCCESS in 1h 03m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ee038e622a224ede9bd6ec9c5890a116 : FAILURE in 31m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eb3dffda56724699a1f3c6ea9c0cd12d : FAILURE in 31m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec1d5b6048d74dd1b6a5bd9ae267e0ef : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3891f919ab7b4454ba980a2441c41109 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45cd7c7df1274ff3897ab60b27129e98 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24c9bbeca34942029dfc1644503d98fe : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e00f26657b9441aa85ff327d8137ed70 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56160b0d6a6b49dcbc77dbc5cb36d8b9 : FAILURE in 5m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f0b6864188724fc2ab001684e207b9a1 : FAILURE in 34m 35s\n\nWarning:\n  Comments left for invalid file doc/source/admin/blacklists.rst","accounts_in_message":[],"_revision_number":1},{"id":"64dc36d3a63af1ebe23484f4b9d35a5c2510ccc8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-26 13:25:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4c17a2a60dbab5727cf7aa91c0d1f74838c27bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 15:14:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/defc6d5e826a43f88006d7c19ce0dbf1 : FAILURE in 58m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7babff1b07094a7b9fc2b893a2af8250 : FAILURE in 4m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/27a11f00437c4f45a9b5b0e4ea252208 : FAILURE in 22m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/98543157f2fe4a4883194b8c7f5dba2b : FAILURE in 1h 38m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8fabbb052b404e1f8255f28429c4c3d3 : FAILURE in 31m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e80ed87ee5b24cd7b3f2533825eac195 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d97c0088db9249b28067e93734221da6 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b242002013b24f93adad0d0684a76f56 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73adef0083724963a786a1a06a3c7f1c : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49541fba7b3848d380b752c5b89521b1 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3ede66c7174d7487ddea2d7bd2a1fe : FAILURE in 5m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e64c5f0f489d4a6d8e7cf377ae1eba7f : FAILURE in 34m 29s\n\nWarning:\n  Comments left for invalid file doc/source/admin/blacklists.rst","accounts_in_message":[],"_revision_number":2},{"id":"13cb00e1798088f154c2dee2e708d6cd84b66a63","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-27 09:25:08.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3d240bccd467c897598b215436eaa2483ac4d2b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 10:37:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/246a559b85274bd5bc2959d40cf07fb0 : FAILURE in 28m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/276a089ac3d648f1a679b46b17582840 : FAILURE in 28m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a176f3818d794cb688fd6209f986f579 : FAILURE in 43m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/93a74fe56fce42428d00bea1210441f2 : FAILURE in 38m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/523b6ebb237b4cd7a01f3c00b31084b3 : FAILURE in 31m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ded5082e477b4f5697d1aef6a07a25ec : SUCCESS in 15m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe0f76742ffc46a3bda5cdb09dd4273a : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c155b88ca104f0da09b64c34866030a : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c91b52d757984ed7ae35da0a0317dcb1 : SUCCESS in 16m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31652c9cfaf445ba87e89f8adf303b5f : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac2213039d341f38ca4e25c8fa717fc : TIMED_OUT in 31m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a077b450c37f4a4c9eb5814d0d0b9cbe : TIMED_OUT in 1h 02m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9ac0bf68a9f0165dbe3c550164672c59aca6d8ce","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-27 15:47:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"654d5ed033b22091f3072e03cabaad71fb0c8306","revisions":{"b43d7a3af81a22f211c6be7c088301a7133ee7e6":{"kind":"REWORK","_number":1,"created":"2020-10-26 05:32:03.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/09/759609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/759609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/759609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/759609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/759609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/759609/1"}}},"commit":{"parents":[{"commit":"769a55c2e523d749628d6ff4a813344801312d13","subject":"Merge \"Fix lower-constraints test not constraining properly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/769a55c2e523d749628d6ff4a813344801312d13"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-26 05:22:45.000000000","tz":60},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-26 05:22:45.000000000","tz":60},"subject":"DNM: Test eventlet cap","message":"DNM: Test eventlet cap\n\nDepends-On: https://review.opendev.org/759552\nChange-Id: I0ebe726eec0fb7c7578c8d08d09bed5c9d306709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b43d7a3af81a22f211c6be7c088301a7133ee7e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b43d7a3af81a22f211c6be7c088301a7133ee7e6"}]},"branch":"refs/heads/master"},"b90bb55973aebb31e407c3fa7fc5b97a4c0fea49":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-26 13:25:48.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/09/759609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/759609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/759609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/759609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/759609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/759609/2"}}},"commit":{"parents":[{"commit":"769a55c2e523d749628d6ff4a813344801312d13","subject":"Merge \"Fix lower-constraints test not constraining properly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/769a55c2e523d749628d6ff4a813344801312d13"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-26 05:22:45.000000000","tz":60},"committer":{"name":"Jens Harbott (frickler)","email":"j.harbott@x-ion.de","date":"2020-10-26 13:25:45.000000000","tz":0},"subject":"DNM: Test eventlet cap","message":"DNM: Test eventlet cap\n\nDepends-On: https://review.opendev.org/759691\nChange-Id: I0ebe726eec0fb7c7578c8d08d09bed5c9d306709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b90bb55973aebb31e407c3fa7fc5b97a4c0fea49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b90bb55973aebb31e407c3fa7fc5b97a4c0fea49"}]},"branch":"refs/heads/master"},"654d5ed033b22091f3072e03cabaad71fb0c8306":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-27 09:25:08.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/09/759609/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/759609/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/759609/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/759609/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/759609/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/759609/3"}}},"commit":{"parents":[{"commit":"769a55c2e523d749628d6ff4a813344801312d13","subject":"Merge \"Fix lower-constraints test not constraining properly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/769a55c2e523d749628d6ff4a813344801312d13"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-26 05:22:45.000000000","tz":60},"committer":{"name":"Jens Harbott (frickler)","email":"j.harbott@x-ion.de","date":"2020-10-27 09:25:04.000000000","tz":0},"subject":"DNM: Debug gate issues","message":"DNM: Debug gate issues\n\nChange-Id: I0ebe726eec0fb7c7578c8d08d09bed5c9d306709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/654d5ed033b22091f3072e03cabaad71fb0c8306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/654d5ed033b22091f3072e03cabaad71fb0c8306"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
