)]}'
{"id":"openstack%2Fdesignate~759733","triplet_id":"openstack%2Fdesignate~master~I2ec7a5e641e6c1d2fa52ad410bb86de446694716","project":"openstack/designate","branch":"master","topic":"no-tail-log","hashtags":[],"change_id":"I2ec7a5e641e6c1d2fa52ad410bb86de446694716","subject":"Drop deprecated tail_log function from devstack plugin","status":"MERGED","created":"2020-10-26 16:25:47.000000000","updated":"2020-10-29 06:42:47.000000000","submitted":"2020-10-29 06:41:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759733-1603953680913-12b68651","meta_rev_id":"ed1aa072aff2404515e7626d59f80717990c4ac4","_number":759733,"virtual_id_number":759733,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"date":"2020-10-28 13:09:56.000000000","_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-29 06:41:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-10-28 21:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2020-10-27 19:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-28 21:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-27 19:54:14.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-28 21:09:30.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-10-28 21:27:48.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-10-29 06:41:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af637a78f0c69a1e38affb897b75d4943923a189","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-26 16:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"effe51268912f7e9858d5568630adaf22d5ca865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 17:25:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e76a1e005af14bd9bf3ec6f6f6bd598d : FAILURE in 41m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2097638aa06443e5aa29dccdfb62275b : FAILURE in 27m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3bb9d02bd41441d29e14499e4de13673 : FAILURE in 49m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/189b16bb03ce418ab06131321df1fafd : FAILURE in 30m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ed98f28f4fa049a0b33adc3f055b54b1 : FAILURE in 28m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d385f0b7bc8145508cf7cb2896897bbb : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce650e1da61545a59d5d12f72f41c060 : SUCCESS in 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21065e523c344695a3f4a7f0aca0d0bc : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e8f66d4a4794d219c59a4f6723a2153 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5b0ea20553a408fbdb17fc6b9047cf9 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06fafdaaa48040239282f666e1564fc5 : FAILURE in 5m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/598399b6341642ed989b8284dc814a9c : FAILURE in 27m 34s\n\nWarning:\n  Comments left for invalid file doc/source/admin/blacklists.rst","accounts_in_message":[],"_revision_number":1},{"id":"9be9274f60ffe1297b5bc04fbcd6c4397d66dcae","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-26 17:39:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64598aeca03871e2a84b56ecf766c1b394c028c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 18:44:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/654c21022a0241a69c84479f58c2f8c7 : FAILURE in 28m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b39bfa4aaed145c79fe2cb59705347be : FAILURE in 33m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f5bb00f616b64639b2185c32e6e3838c : FAILURE in 56m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6c22d54c0ef54e8685d759d8658bc521 : FAILURE in 32m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/229bf8b0f956466c8d2b85bb2a4f6995 : FAILURE in 29m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c881dc509f24cc5bb8e0356c68fa60f : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40e7fb5b6895480e9a3b6f5332ef47ff : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33bad0bc998d4ccd8d0281e4a28f5b9a : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c7774703d864d0784c6ed7c8136ea2a : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94bc9c24a8c447178eee0b1c2d00a25b : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f6effb463ec438b9045a88c6b9a4ad1 : FAILURE in 5m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eb9ff43504da454f8f8485bd5343ff80 : FAILURE in 33m 29s\n\nWarning:\n  Comments left for invalid file doc/source/admin/blacklists.rst","accounts_in_message":[],"_revision_number":1},{"id":"4e2f611ee03b2ef04e3a82746603a729f6b1b5fd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-27 15:46:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d40fb3c705c1d2c246eeffae62f8e3fc04de6778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 16:40:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/060fdc0cb125483abde98299c14e7b69 : SUCCESS in 35m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/13513ca993e144a9b30ef3685edb2277 : SUCCESS in 41m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b33df06b7b4145dd8429a45288d3e52c : SUCCESS in 46m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9c94fb85b4ed44ae9470ad9aabf4ebde : SUCCESS in 36m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0949bb6a245e4cb199df4c046ab57b0f : SUCCESS in 42m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6effbe2804e47659dcd447e84557e26 : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0c561c08f924eb199927ce6c763cfd1 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6760e876768c45b7934f657d7d52ccc7 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/640df8ea0367408e87630be7043b5e20 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c24e990b77434c90be3006b0c6134c66 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9217780dc24646c8be19a51108aa46e1 : SUCCESS in 6m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae380d37813d4c668916b6100d171d66 : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":2},{"id":"1c51b02cb632704fbd9013c861e836f80363099a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-27 19:19:58.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"45e25cfdc70553306f9e5f9cd30fbd48a33166f6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-27 19:54:14.000000000","message":"Patch Set 3: Code-Review+2\n\nIt would be nice to have a reference link to devstack, but I am aware this is deprecated/going away.\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"2a9413f7a3300b3d6ac789b1392374f3684b7af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 20:18:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/94a6558f69ac4b77bf83beb9cb5bbae2 : SUCCESS in 36m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/173821e497d24d4ba4e97614cf376359 : SUCCESS in 36m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ac35db66cf60427b88fb417cd8f47a1d : SUCCESS in 53m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fa2ec3662ae0431c85dff327cf8034b5 : SUCCESS in 32m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c314b5f31ec3462c95c444e6a0ba2e67 : SUCCESS in 43m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02931da87c8b4a07a4b78bee46f3ac31 : SUCCESS in 13m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8246d2b19e7940e195122006f8ef09e7 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43a1d160345c4e47a5d4ce2272a4a50e : SUCCESS in 13m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fab9c4c776264374adb8c14783519af0 : SUCCESS in 14m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c84597be4d8446d2a2beb3bd001139dd : SUCCESS in 14m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590bdae151784fc98845dfd4631c6a06 : TIMED_OUT in 30m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/05da6b6edb2a4ea38e4b6656ae4ab1c5 : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":3},{"id":"43c086c68cb09242b3fcebf15551809a391ab7f7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-28 13:09:40.000000000","message":"Patch Set 3:\n\nDevstack change is https://review.opendev.org/760126","accounts_in_message":[],"_revision_number":3},{"id":"537469596e452b7c5083301494088ec0febcc4a1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-28 13:09:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b1643d8b82be0aaf6c8cd97166883cc47172d18e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 15:02:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8ac52162f1ef44e490ba50bd475e9032 : SUCCESS in 45m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3efd0c7d185849fab2913bab25abe673 : TIMED_OUT in 1h 13m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c272b178af3f4c36836c0c5782c27d87 : SUCCESS in 1h 30m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/144791b8fe3440439340e7a2a57349ba : SUCCESS in 37m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c991e2eabcfc40dbb8cbdf59e75a194d : SUCCESS in 35m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2df368b0ba8144a5a4f9d55cbb870f6c : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24008f89c9df4bcf9d214307652db609 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5519a547f5414db39f80b936d8d5a765 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73a63de4e5504836a74402dbc2632378 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dffe5b4616e476e93f6240afb060d0a : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae862940f244e138bd4afc020f0861c : SUCCESS in 7m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/78745f97ff98479a9b130a52a0c39340 : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":3},{"id":"227b5ee46af27e01bf56ccc6df260e1379c12430","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-28 17:56:21.000000000","message":"Patch Set 3:\n\nrecheck\nsetup was a hour before tempest even started.","accounts_in_message":[],"_revision_number":3},{"id":"0a47eafceb1c05eb9de12bac5794af274fe58139","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 19:10:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/85aca006b6ab4ab3b5a2ccaedf2f9d3e : SUCCESS in 38m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a6829c9d97bb4dd39eff7db83fabfa07 : SUCCESS in 38m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9ef7263bf8ac47658345a35b5a3bc77b : SUCCESS in 1h 05m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f9c64f6447ca4f4c9279e9f4d601674b : SUCCESS in 45m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3bcb0ccb3af04ddeacebc5499713a84e : SUCCESS in 33m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c99dde1fe13a4b62b5d98356a834c37b : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a78b855c7aa4dc59d8f0640612e0997 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/056c56732f824dc7b0ceae5be7406b37 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb2bb283d3b04ad99bc95de0b0e08b72 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f89f2bd00d3a48c88bdd66f74e508b66 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ebb111e6d31463280ac1d4f888e5bc2 : SUCCESS in 8m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/589bdeb9ef3d4b52bf057c3ede209b2b : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":3},{"id":"802f71cbd16bd342a0fb160cf57a0075fffc732f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-28 21:09:30.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"1e5e93bed011113234be76623e9ce859a5f9c9d1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-28 21:27:48.000000000","message":"Patch Set 3: Workflow+1\n\nI feel like something might be broken with logs. I see almost no api logs, but does not look like it is related to this change.","accounts_in_message":[],"_revision_number":3},{"id":"e34e90ddb7a901806babfa27ff33684787c15964","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 21:58:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2c8ad02c6542ab3a29d8434a3051128b675230d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 23:06:50.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a3b2b40d0b074517a70a616066d3b8d4 : FAILURE in 4m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c7b56ce26d3a478580549ece1bc85f66 : SUCCESS in 37m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/707cd55d20af438f9855aa3b613c182a : SUCCESS in 1h 01m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/07ba930c4dcf4ec29b98d84ffe11db1a : SUCCESS in 48m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3903abed2e8c48e7b86042c114a0d368 : SUCCESS in 44m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63709a5255804def843d7fe7affae245 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85246d39c6a44acdb62c1a4e41b78079 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6af04d27abf1461a819ca49f7865b294 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dacc1ba4262641d1aa81103af9f64ee8 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a18cfa430de24c9682035b345f3e6a00 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":3},{"id":"d47e43a0d583135beee2eaa6552c05c99b3a6b38","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-29 04:55:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fcfa8bb935595de5af191918bcc6dd3287a0c97f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 05:49:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d612e22a07a349aea392bac6acd8ff8b : SUCCESS in 46m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8247223cb27946f08b2dc549e18375f8 : SUCCESS in 32m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d51c1f5c77df45dcae189afe3e330bbc : SUCCESS in 49m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e5c14547bde34583b6859ce255b3931f : SUCCESS in 39m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ccff9f81a0244120acd8651c6989e59a : SUCCESS in 37m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de8c802465e34e4fb9788a1fdc0fe665 : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25e4c684cef941cc837a860bd1b1af7f : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11c9c3c33fac49f9913ab2056792e326 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59f583bfdfe745b1a2da1daa6d4030f4 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a9767c52aab49148cbd4b03b1ccade5 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c0704d1df064f9c8bc0d031a2137471 : SUCCESS in 8m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4eb6ec485e064e798ee22bfa8483cd2f : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":3},{"id":"642970d690d0ff2fc6c8a87fdb82a0d1aee05211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 05:49:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e2a5140b367ded91b29b30c141e0d72899d136bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 06:41:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1ec6e71b0cfc45ec8435d94e84284a6c : SUCCESS in 38m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/da05bdf7b6ef40f89d1213001c5a0c4a : SUCCESS in 39m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f2bc82dfb22f4f179603c0b2f35ace65 : SUCCESS in 47m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e8077a32dd9a4f2ea4fc9aca677f8e54 : SUCCESS in 42m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/29fbcf40aa9a4f768db28b4c2f7c6780 : SUCCESS in 40m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5289736da22426e82adfec8fdc158f4 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f25a926b3a4e639e9266127ef612ad : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6ee7d3120a24662a43cd667ad786bfb : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/755feda0aefe4f0f862e6b894719336f : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ffed1b645b74140b8480967dd13b51c : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":3},{"id":"492eb0557fad1642f5048ec9e995f2f197ad5bf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 06:41:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8ffbffd00a713902680be694c9dbf1fc21448a66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 06:42:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5a6e7c17c34e58810002d7b3b1cde3 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2255830a6a3e77b754a3d7264163e5e11860ef21","revisions":{"8a1d547b4ccfceadebc78bb1057c8e431134c03a":{"kind":"REWORK","_number":1,"created":"2020-10-26 16:25:47.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/33/759733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/759733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/759733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/759733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/759733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/759733/1"}}},"commit":{"parents":[{"commit":"769a55c2e523d749628d6ff4a813344801312d13","subject":"Merge \"Fix lower-constraints test not constraining properly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/769a55c2e523d749628d6ff4a813344801312d13"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-26 16:21:49.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-26 16:23:58.000000000","tz":0},"subject":"Drop deprecated tail_log function from devstack plugin","message":"Drop deprecated tail_log function from devstack plugin\n\nThat function is to be removed from devstack, remove it in order to\navoid our plugin breaking.\n\nChange-Id: I2ec7a5e641e6c1d2fa52ad410bb86de446694716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a1d547b4ccfceadebc78bb1057c8e431134c03a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a1d547b4ccfceadebc78bb1057c8e431134c03a"}]},"branch":"refs/heads/master"},"bbed0402d537b59acb519a377bcde43c782baa4f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-27 15:46:55.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/33/759733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/759733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/759733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/759733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/759733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/759733/2"}}},"commit":{"parents":[{"commit":"95a92ee5aada5886582a97658ade5021a20f6168","subject":"Be explicit about the coordination URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/95a92ee5aada5886582a97658ade5021a20f6168"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-26 16:21:49.000000000","tz":0},"committer":{"name":"Jens Harbott (frickler)","email":"j.harbott@x-ion.de","date":"2020-10-27 15:46:55.000000000","tz":0},"subject":"Drop deprecated tail_log function from devstack plugin","message":"Drop deprecated tail_log function from devstack plugin\n\nThat function is to be removed from devstack, remove it in order to\navoid our plugin breaking.\n\nChange-Id: I2ec7a5e641e6c1d2fa52ad410bb86de446694716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbed0402d537b59acb519a377bcde43c782baa4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbed0402d537b59acb519a377bcde43c782baa4f"}]},"branch":"refs/heads/master"},"2255830a6a3e77b754a3d7264163e5e11860ef21":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-10-27 19:19:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/33/759733/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/759733/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/759733/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/759733/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/759733/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/759733/3"}}},"commit":{"parents":[{"commit":"512943a037f24c1add0a79093a0cef1fe9da8789","subject":"Be explicit about the coordination URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/512943a037f24c1add0a79093a0cef1fe9da8789"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-26 16:21:49.000000000","tz":0},"committer":{"name":"Jens Harbott (frickler)","email":"j.harbott@x-ion.de","date":"2020-10-27 19:19:58.000000000","tz":0},"subject":"Drop deprecated tail_log function from devstack plugin","message":"Drop deprecated tail_log function from devstack plugin\n\nThat function is to be removed from devstack, remove it in order to\navoid our plugin breaking.\n\nChange-Id: I2ec7a5e641e6c1d2fa52ad410bb86de446694716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2255830a6a3e77b754a3d7264163e5e11860ef21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2255830a6a3e77b754a3d7264163e5e11860ef21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
