)]}'
{"id":"openstack%2Fdesignate~763870","triplet_id":"openstack%2Fdesignate~master~Ica9db41939e17fcd67b97dce5191e75bfb396330","project":"openstack/designate","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ica9db41939e17fcd67b97dce5191e75bfb396330","subject":"Add useful common policies to base.py","status":"MERGED","created":"2020-11-23 22:49:51.000000000","updated":"2020-11-25 17:39:33.000000000","submitted":"2020-11-25 17:37:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763870-secure-rbac","meta_rev_id":"6799070295d51a24a30f50e283183a9fdf24bb67","_number":763870,"virtual_id_number":763870,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-25 17:37:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-11-24 13:08:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-25 02:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":2,"date":"2020-11-24 20:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":1,"date":"2020-11-24 20:36:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 00:12:48.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-11-24 00:21:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 20:36:19.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-11-25 02:39:53.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"c9038b9e9a552caa594e68dc88d00d531012f1ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"436cbd73d217199bb0ed4104925b5ad49d3a0c50","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:52:58.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"3851008bd5185924d6192ff6bca37e4447e85b7d","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 00:12:48.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"7d27bba101b62045474c14859d5004b0ad76b172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 00:21:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3f88644b8ab2461aaf4acae46246348a : SUCCESS in 43m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b046f1d736934b3683c8be5414fc0878 : SUCCESS in 37m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3fa2e94e1a404490bc7a7db5a966b0cd : SUCCESS in 1h 11m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bb3a3b20e12748939978d437c648bd5f : SUCCESS in 46m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/81270e2dc02e4d6c8423d8860c2bdf35 : SUCCESS in 42m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52972e5f372344b687397a334e865384 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77e74d6f16bf4faabebfe7d6e86b73b1 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db5fa43e0c8242349746d67686a37303 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be13283018924f9a83272bb91cf6c490 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/842243fe55984ac9a76cc72725648a60 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d47b3d13203d4994985c946f6d3eaf7b : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e417b16eae946d396964036d2c96d7f : SUCCESS in 7m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/78f86f7397eb4046bd29a59239e5cd15 : SUCCESS in 44m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7c8f59a265d71de9b3e5f28d34ce5be8a93c6a7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 00:50:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"44e115d7e04237f8dd7b8fd2d0fcf1c04d447942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 01:04:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c1ec76d325d844785c52645d5550c23cf9d7965a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 02:10:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d0af65af00ff419cb6bd0cb3d1a246bf : SUCCESS in 42m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f1186172834e485e80dd52508eef3f14 : SUCCESS in 50m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6c0045d6d93b42c48c4ced4aca4d02cf : SUCCESS in 59m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e201022ee00c458e9a7a40b321ea5b38 : SUCCESS in 45m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/75661bc22ad44f87851184e1800ab273 : SUCCESS in 41m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c96375a7fd14a8b84f4686bba3c5740 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/517b0b85a9e84de2b30a2c86288f2cea : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab0c73ec2a8948f3935d2e366b56f8a6 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b51026ca76364fd2847b96e6daee1758 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38315104cc974e328d001c63bfb907fd : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef027699ef044550bae1652ef6a994d9 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1607f64ff0e24c18b7359ef2490cc7a2 : SUCCESS in 9m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6336d658809244608faec7c2956a2f93 : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":3},{"id":"18bcbbec66b4916ede8e4d79afda8c9f450d79a4","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 13:08:54.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"f8dd7ac0ced2d822f091ce28f417b963e6549719","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-24 20:36:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4725ae3a7c1f0b514ed2e43052b02776a5455f2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 21:54:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d24130e98bed4beb7b032a185ecc5433d5f47f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 23:10:55.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"bdb1718860b672a37fa757eceba6d5e34085b1b3","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-11-25 02:39:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"054337d9fbd0d5f1a0fffad5c8528f351482ab62","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-25 13:44:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"110d5212381c6041ddb6bbe29ddce2ee11404fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:54:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/75241294a171457ea52750b737abd771 : SUCCESS in 39m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4bfa47e22879430aaf96788358db35e9 : SUCCESS in 33m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5a5a5b8f176940c191b31bbc25d379c8 : SUCCESS in 1h 05m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c21930fe7f9949429b2cbaaf26e00d60 : SUCCESS in 46m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/806bb7a4450b418994365f0c26bc1fa1 : SUCCESS in 47m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f16fbc91f5664138863d4eee266428e4 : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d4a6e77f828456c91b66da62bb5deef : FAILURE in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/804628b36264451185adaa1c9cc6b761 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/222c7ca9cf074961a826240da41fc748 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f811211fa65b4edc9a0acc57ed9e6cc7 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa4233d0f2984d0fb11ef02dcfb28670 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace18bd6a69449afa814ccc7d43557c6 : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8762da18422d46c9a6925046eeb06e2c : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":3},{"id":"98e97c09c05769f348d4619e8eddb4b9be761839","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-25 15:29:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"34cab6a9ce323dcfca5b41be9b37e62154c0d9b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 16:33:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/93aa537a804a45fba47a1dcfc2f9c5be : SUCCESS in 34m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/45018bcd371646e086eeb09125d77e5d : SUCCESS in 32m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/275e3fbc77a0493b98c8f32c60ee1df8 : SUCCESS in 57m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/62bcba1d311d4cfabc7535267c450c2e : SUCCESS in 41m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f49dbec58f4f4756a452a52118eec1c4 : SUCCESS in 43m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81f79095f5b848428df21ad8aafcdb4b : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d945e56230f9445cade92689e7f74395 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bafe49337834100b6df75b05e8731bb : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf28fdef804e4551b49f71033e054a0d : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aaf9d08027541bfbe84a7ef69e05b9e : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed85561433044bd39fa90e494ed235eb : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f82e19a0b85427683dc5d5177f476bd : SUCCESS in 7m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d72417a25b544f8b95bec9d8dbd23262 : SUCCESS in 37m 01s","accounts_in_message":[],"_revision_number":3},{"id":"758801d61f62e6084f29aac7eba1fe0aadd76a0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 16:33:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a9d2e54fcf834680a5f050e265d9de653d23e643","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 17:37:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7fb5fcba2f0a47cda93c3b0fd2e76c41 : SUCCESS in 42m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ba6d3c14a5d341f0ac259d90dd345859 : SUCCESS in 44m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/308fe667a0054ffa863952ccf12185e6 : SUCCESS in 57m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e88aacbe2a8e45dfb195f1db8bf56a2d : SUCCESS in 34m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/954403d1872e470f96ebf40c2a7f80b8 : SUCCESS in 45m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fde949d9e3347818e1048cc0f620d37 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9499bf25d16a4b8f908885e88c5476ca : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/799bbb1a5c89450d86c031829bb1cb58 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffe23d891e4e455d80e6d6e4ea2e84b1 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2acc9c303c344164855909f9007ff662 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":3},{"id":"8fdd170efa987158fd237eab6e8affc2b30cd8d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 17:37:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6799070295d51a24a30f50e283183a9fdf24bb67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 17:39:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9d0462df3c2458386349bafce2b705b : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f4d35c02dfd3850596b6188b52051567fbb895a7","revisions":{"06c9ded69d5fd203db4189d8d1fc9b9291af1f45":{"kind":"REWORK","_number":1,"created":"2020-11-23 22:49:51.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/70/763870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/763870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/763870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/763870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/763870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/763870/1"}}},"commit":{"parents":[{"commit":"c62e3fa1976086aca799098ca017805a432d2109","subject":"Bump minimum version of oslo.log to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c62e3fa1976086aca799098ca017805a432d2109"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 19:20:57.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 19:20:57.000000000","tz":0},"subject":"Add useful common policies to base.py","message":"Add useful common policies to base.py\n\nThese common check strings are useful for implementing consistent policy\nchecks across OpenStack by adhering to a common persona. This also\nincludes support for a default read-only role.\n\nSubsequent changes will update the policies to use these check strings\nwhere applicable.\n\nChange-Id: Ica9db41939e17fcd67b97dce5191e75bfb396330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06c9ded69d5fd203db4189d8d1fc9b9291af1f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06c9ded69d5fd203db4189d8d1fc9b9291af1f45"}]},"branch":"refs/heads/master"},"b83afeb8d3ab55c4a020ff1dd733b9808c01e558":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 00:50:53.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/70/763870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/763870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/763870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/763870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/763870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/763870/2"}}},"commit":{"parents":[{"commit":"eebcc0d240b8d3371102f248b2b1aa6f47d2113e","subject":"Bump minimum version of oslo.policy to 1.32.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eebcc0d240b8d3371102f248b2b1aa6f47d2113e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 19:20:57.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 00:49:26.000000000","tz":-420},"subject":"Add useful common policies to base.py","message":"Add useful common policies to base.py\n\nThese common check strings are useful for implementing consistent policy\nchecks across OpenStack by adhering to a common persona. This also\nincludes support for a default read-only role.\n\nSubsequent changes will update the policies to use these check strings\nwhere applicable.\n\nChange-Id: Ica9db41939e17fcd67b97dce5191e75bfb396330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b83afeb8d3ab55c4a020ff1dd733b9808c01e558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b83afeb8d3ab55c4a020ff1dd733b9808c01e558"}]},"branch":"refs/heads/master"},"f4d35c02dfd3850596b6188b52051567fbb895a7":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-11-24 01:04:14.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/70/763870/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/763870/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/763870/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/763870/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/763870/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/763870/3"}}},"commit":{"parents":[{"commit":"11d03924e73cd152e377e24304905a072cc4366e","subject":"Bump minimum version of oslo.policy to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11d03924e73cd152e377e24304905a072cc4366e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 19:20:57.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 01:02:56.000000000","tz":-420},"subject":"Add useful common policies to base.py","message":"Add useful common policies to base.py\n\nThese common check strings are useful for implementing consistent\npolicy checks across OpenStack by adhering to a common persona. This\nalso includes support for a default read-only role.\n\nSubsequent changes will update the policies to use these check strings\nwhere applicable.\n\nChange-Id: Ica9db41939e17fcd67b97dce5191e75bfb396330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4d35c02dfd3850596b6188b52051567fbb895a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4d35c02dfd3850596b6188b52051567fbb895a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
