)]}'
{"id":"openstack%2Fdesignate~763871","triplet_id":"openstack%2Fdesignate~master~I8ba1c1fb57eb8a7abf2c101d446259cfc7a8d34a","project":"openstack/designate","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I8ba1c1fb57eb8a7abf2c101d446259cfc7a8d34a","subject":"Implement secure RBAC for blacklists","status":"MERGED","created":"2020-11-23 22:49:51.000000000","updated":"2020-11-27 00:50:11.000000000","submitted":"2020-11-27 00:46:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"763871-secure-rbac","meta_rev_id":"d9844ce4f2fdeb947cbb3e13d1b547071c9391ac","_number":763871,"virtual_id_number":763871,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 00:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-11-24 22:00:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-26 00:17:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-26 00:17:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 23:53:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 00:15:32.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"CC"},{"updated":"2020-11-24 00:44:35.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-11-26 00:17:30.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"6c8babe1d6d4bf02aeae79b4b2f985cf4633d38e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13761005225bb07a19b6b99b39697a8e3700e429","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:53:05.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"2d06f8ed292cb67997faddc226c7c17a18bb3ed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 23:53:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/20f945300a244d75af7ec7bacf8ff186 : SUCCESS in 47m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/197bde3fa549460d82d9e49a04713f42 : SUCCESS in 47m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ddea9030654942cbb0fdb77a8d6b9638 : SUCCESS in 52m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/28ac77dd3a1d427893e91b24149d5467 : SUCCESS in 33m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/63a4bcca26e0431e8923d6a8c91cc979 : SUCCESS in 41m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb26c0905b0348efbb9462623002a537 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3457dc84e3a94395becf6b507dda8db2 : FAILURE in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373ee39993254d05b9ae53b16a3221d0 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cf81496e91842188c34ea0f2d3275e2 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f343431002964eddbeb8234a927e18b4 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/557eb717eff04f81a5950835819b3156 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/defdfe8fff6b409c88708d748b835398 : SUCCESS in 7m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d72cc2e115794dacb1bf2cd195f5f57f : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":1},{"id":"87424668cd3c22dc2cd0a1e52e7d100841305935","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 00:15:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79ae642a850ad29fb2fbb13af2a8492f27451940","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 00:22:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c2c5ae2f26cc8f6b0790a7b8fcc9e7534b7bf12","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 00:44:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17beebbc50b22a93f6ac7454ac8c21cf5cc9c209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 00:50:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"85ace5acb34b12e6b3b7dc6e6688b74a66c7af00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 01:04:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4f1b61415714847522753fbca07cdedd92c46612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 02:00:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c44c9d68b5d04544b14f0cddbce28ed5 : SUCCESS in 36m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5f2fb4d872d946d0983636edb4e8bad4 : SUCCESS in 50m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f63088a1e1da4c42a01b5f83640c704a : SUCCESS in 47m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0d0056c3a02b44cdae2a4810f62129b9 : SUCCESS in 37m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f3756c73b29f4cd39669c04766ede0db : SUCCESS in 52m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cc4c4d5e6244a6bb4de638fb10aada9 : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eef3a076334d477c9228dd17e034c7af : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92a2f97d971b42bc8302a07ac2a98aac : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2868c5065afa47c4918b0f1813ffad18 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91a9e3f8540245c1a8377f27755ba403 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2ad89c15edd47c9b5b4dde58ac01017 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4fbd210d2584a8484d857a9148d244e : SUCCESS in 9m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/125a51ca91f1413f8610fa75a1ba988b : SUCCESS in 30m 46s","accounts_in_message":[],"_revision_number":3},{"id":"ae4185fed58ffdf750debcfe31fdc80c51269398","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:11:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3ca8f0023d0c6ee258a891b094e5b2306b4e6556","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:12:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b256fbbbcd7c18f66f91bd99f722dddabaa75009","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:12:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"660ebee5d5796a53f6549456322aa7eada83a92a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 05:17:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/44c7b34d58c54f65b13b0dc68402d30f : SUCCESS in 38m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/56797a156e47482b964d87b47dd8fe09 : SUCCESS in 43m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ab6796877ee141a196f2f1d86ed479f5 : FAILURE in 1h 04m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/df433e24943a42029a4d02393ab39f19 : SUCCESS in 43m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c8469773c44c4de58d5848f08ab9e0dc : SUCCESS in 46m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b64b6f8bf6e42d4a6bd101e298ea40d : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4cc09d9f7104b59acf56314a49b5f1e : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec723bdc72a14f8381380df0b1d60328 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65d62740dce640958db0454e756f9cd1 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8410d84a70154e2aa53a7032dcae595f : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5eea058e3e148768657addbfcacc923 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae284308a81c461d8e858c5b221b17a6 : SUCCESS in 6m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe062bc8e599488a871cf860e4ceb3c0 : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":4},{"id":"372d7976486fdba7fd9398bd3ad214d9184899e8","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 13:10:32.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"df6dca84768f6bb5cc87ee1c30dafb69662e7e48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 14:14:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0840673860354879a0b0ab8c52285e9c : SUCCESS in 43m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/71c52b27667147a49c03aac29a306773 : SUCCESS in 36m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3bd90e7eb6794fd0b65b0ee11b5fcba6 : SUCCESS in 56m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d5869b81cfcf41b5bc4faf4d9d4f1eb4 : SUCCESS in 42m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/944d8865c86546bf8970f4b3ecac2958 : SUCCESS in 43m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bf12970471248b1847faa418acae08e : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6c5f4d4d03242f09051dd5ca247a8d6 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ef5185243604fbba38816281b0148c5 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0274746b4da44769628bddcb1ee2fe8 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99e568b0daf744c79745f1de7caf2561 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2290b485525d4d75bd7aa356676b560d : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d45cb197e00b47a589611cc9e73273db : SUCCESS in 8m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44aa7ea900db4f8f9c53d92fa2190271 : SUCCESS in 26m 48s","accounts_in_message":[],"_revision_number":4},{"id":"c61ef11678c2f13366b8d21c7de4020fb9a98a40","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 22:00:05.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"71a698bed4c41b59ef96d1701ed660335d6b3466","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-26 00:17:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3ae583c69aef1e6e5ba345abdbb0eb445c35f90f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 00:17:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fb4ab12ca179247b5b4057b39bb717113ae6c10f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 01:37:08.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3093a759b7d1499bbf038061cbdc1b54 : TIMED_OUT in 1h 13m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6d78b2d5e7cc4c10a7a8b903c3696c24 : TIMED_OUT in 1h 12m 45s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/400fefaa16464ade958b845ad4ce2fdb : SUCCESS in 58m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5c12044adebb470fa259dd3fd3c9519b : SUCCESS in 37m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c3f46e5fd2db43938d704e34cb582ec7 : SUCCESS in 47m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9601bbaf0f3c49589da74300cdbdaaa6 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/807b449b2aaf4e898eb5313ce7fb5076 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebdafcd96ad448e886f8d4715e02f2f3 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/935299a92d4a4446936d24f626c18f63 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4944a8b09c8342219ac896040ee7bb48 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":4},{"id":"42960881818b5e4f00395b3fc3c5d3d836e8571f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-26 03:00:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6d6c29ac7f13125f0d8b22b2de22dd39cec56bd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 04:15:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2a2566eeb65942fdaafa49d7c370b11a : TIMED_OUT in 1h 12m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0e2f75b5960c4166831e0542fb36ca6c : SUCCESS in 1h 03m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cb6b5e0fc30f45e4b5a1747827ec30ca : SUCCESS in 43m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/41093f1d17bc4224be6f8145f9d10a7a : SUCCESS in 42m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/505c855a5fa4467688527988665062fd : SUCCESS in 42m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eadeb3245c134d20b2e19c8cbfd25168 : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc4ae89f0a5142d3a27bcf946941f43e : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b1efe12969452284cdc8eea9c0deab : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f79e21a0e2494f8bbd2a8666e3b29cdb : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82174e93a7a841159e1c4d42165d55e4 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5eea3b95a2fb4b3d98bcf2c9f2cddca3 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4265b4cda44676847ea030bf59ef89 : SUCCESS in 6m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/604a26ab876b40b9acb07643fdcdc5a2 : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":4},{"id":"4b93dfa67a531b33382bcd3eb793942c79a4bf89","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-26 22:34:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ae783a7482d4e4532a64daf653895751b6a11c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 23:34:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/02e0ca0fe6c74f3490eb82922f2bc91c : SUCCESS in 39m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8845acfc2adb460aa0ebe562ee22b2b2 : SUCCESS in 42m 45s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0e1b5d905ca74135bfac4ab4f19aef74 : SUCCESS in 56m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ce89ffbb3844441aaec25f97ef6e78ba : SUCCESS in 42m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d3d2e0f030d946178a529796d01f7ce1 : SUCCESS in 44m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96dd84534c994d3ab83d2dbca2bfc0b9 : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/628aa74c021a47918dc0be4491d07283 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c81c326036ce43c6a3e35b282328cf0b : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e28617565d04ff88f33392bd0c0a14d : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b0819714630475cbca119b440094ddd : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84da601ed64d473c91892d84c3de0081 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/841bcbc85195404aacfcd746aa6012cc : SUCCESS in 7m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf22552da4e64adb8a76f5292a982dba : SUCCESS in 36m 07s","accounts_in_message":[],"_revision_number":4},{"id":"9779d6aabf51761ae246734939c2aed55c42f717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 23:34:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dcef2a208abb37e27bc8d439e8a0cba36c46f518","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:46:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8d3d25f7fece4d1790cd21650e6ce2b7 : SUCCESS in 29m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3ba0b584efbc4e7eb541cb56fad57cee : SUCCESS in 1h 00m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4f669add36cb474d9a33aa9dd9b9c67f : SUCCESS in 1h 09m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/34992fbd4763448198722bc49cd78eae : SUCCESS in 55m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a653018f90ba4e21b37ae79016b99c6f : SUCCESS in 1h 00m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b15ade54c15498c9eebf5a25c95d95a : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45422de0450c41ad97d103236110e9fe : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a217aecf5a654ab3a53357bcbb6a7497 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a617e3248d245e08ea89320170793e3 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce908a04e8864ddc9bbbf341ccfad7bf : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":4},{"id":"4062ab3f2c10b5ab564aa39a9ea33ef119969281","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:46:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d9844ce4f2fdeb947cbb3e13d1b547071c9391ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:50:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c296c8a2358047ad89553355652a871d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"638d6b700ddbc733c24b64ccf056f96c63c1c67b","revisions":{"90c0f9a028c57d63a3258ed2db91c45211b11dbf":{"kind":"REWORK","_number":1,"created":"2020-11-23 22:49:51.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/71/763871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/763871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/763871/1"}}},"commit":{"parents":[{"commit":"06c9ded69d5fd203db4189d8d1fc9b9291af1f45","subject":"Add useful common policies to base.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06c9ded69d5fd203db4189d8d1fc9b9291af1f45"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 19:25:42.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 19:27:27.000000000","tz":0},"subject":"Implement secure RBAC for blacklists","message":"Implement secure RBAC for blacklists\n\nThis commit updates the policies for blacklists to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I8ba1c1fb57eb8a7abf2c101d446259cfc7a8d34a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/90c0f9a028c57d63a3258ed2db91c45211b11dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/90c0f9a028c57d63a3258ed2db91c45211b11dbf"}]},"branch":"refs/heads/master"},"eb0bd7bd9cd490a30c127c68fd34307f459171a3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 00:50:53.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/71/763871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/763871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/763871/2"}}},"commit":{"parents":[{"commit":"b83afeb8d3ab55c4a020ff1dd733b9808c01e558","subject":"Add useful common policies to base.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b83afeb8d3ab55c4a020ff1dd733b9808c01e558"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 19:25:42.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 00:49:27.000000000","tz":-420},"subject":"Implement secure RBAC for blacklists","message":"Implement secure RBAC for blacklists\n\nThis commit updates the policies for blacklists to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I8ba1c1fb57eb8a7abf2c101d446259cfc7a8d34a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb0bd7bd9cd490a30c127c68fd34307f459171a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb0bd7bd9cd490a30c127c68fd34307f459171a3"}]},"branch":"refs/heads/master"},"44bb2a855d6c6ea46bcf85fc80bb786fc3d94244":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-11-24 01:04:14.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/71/763871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/763871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/763871/3"}}},"commit":{"parents":[{"commit":"f4d35c02dfd3850596b6188b52051567fbb895a7","subject":"Add useful common policies to base.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4d35c02dfd3850596b6188b52051567fbb895a7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 19:25:42.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 01:03:03.000000000","tz":-420},"subject":"Implement secure RBAC for blacklists","message":"Implement secure RBAC for blacklists\n\nThis commit updates the policies for blacklists to understand scope\nchecking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC\nexperience and improve security.\n\nChange-Id: I8ba1c1fb57eb8a7abf2c101d446259cfc7a8d34a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/44bb2a855d6c6ea46bcf85fc80bb786fc3d94244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/44bb2a855d6c6ea46bcf85fc80bb786fc3d94244"}]},"branch":"refs/heads/master"},"638d6b700ddbc733c24b64ccf056f96c63c1c67b":{"kind":"REWORK","_number":4,"created":"2020-11-24 04:11:24.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/71/763871/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/763871/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/763871/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/763871/4"}}},"commit":{"parents":[{"commit":"f4d35c02dfd3850596b6188b52051567fbb895a7","subject":"Add useful common policies to base.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4d35c02dfd3850596b6188b52051567fbb895a7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 19:25:42.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:08:04.000000000","tz":0},"subject":"Implement secure RBAC for blacklists","message":"Implement secure RBAC for blacklists\n\nThis commit updates the policies for blacklists to understand scope\nchecking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC\nexperience and improve security.\n\nChange-Id: I8ba1c1fb57eb8a7abf2c101d446259cfc7a8d34a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/638d6b700ddbc733c24b64ccf056f96c63c1c67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/638d6b700ddbc733c24b64ccf056f96c63c1c67b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
