)]}'
{"id":"openstack%2Fdesignate~763874","triplet_id":"openstack%2Fdesignate~master~I093223790b699e35fb151e17111e34c24153d4c4","project":"openstack/designate","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I093223790b699e35fb151e17111e34c24153d4c4","subject":"Implement secure RBAC for records","status":"MERGED","created":"2020-11-23 22:49:51.000000000","updated":"2020-11-27 00:46:39.000000000","submitted":"2020-11-27 00:46:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763874-secure-rbac","meta_rev_id":"722083f82226de02ff73afd17c93d330ed118939","_number":763874,"virtual_id_number":763874,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 00:46:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-11-24 13:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-26 22:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-26 22:35:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 00:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 00:50:53.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-11-26 22:35:54.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"a20deebee7a52892f80e185e15c9727a362b33fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dfa9a7aef6866dc1efb4d1c52f3cbaa5e88832d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:53:29.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"fb572656ca0b704b6e217f9b54fde78cd0a81fd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 00:04:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3fe0f8b4e52649b7be2f06771b236350 : SUCCESS in 46m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4c9accc95b9341abbf6666eaea58031f : SUCCESS in 43m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bc97118fc8aa4f98bf447cb3f5761e0a : SUCCESS in 57m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/60f189cafff843e6bc4b0365b8da1d46 : SUCCESS in 40m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1c335a85dbf94ad7bd9306d92bb58b4d : SUCCESS in 43m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95a61a0f7dbd4840991c82f281f50199 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dda3ec6fe143454d923ffcf1b1bea09d : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fee571a055a47f6a7bba07aaac6b49b : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fde9831f0eb846a4a07f17f6d1de107c : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/693700f2c40b44c5bfb90297ec6939de : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af871f1afd224888afb9d40450a20ce2 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08b600458454efaaede14266e3740d7 : SUCCESS in 7m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e0dabc9de664c3cb23498fb4d382894 : FAILURE in 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"de0a4dbfdaf5b79ab7f1c107e9dea8f8268b7b65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 00:50:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"996c6e9d283f5ebfc433cf38db1ee562fb856f15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 01:04:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"410c01730464e6906527f175f8142d5a475762a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 02:47:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/994824b3224841fd8d17a82614fcbce5 : SUCCESS in 35m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/714eb20c680f43d581f10c5a1c67d133 : SUCCESS in 40m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fba404d2cf1440f2a1e1c708eb687a77 : SUCCESS in 55m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7460e90098f54d06ae9f859304d8614a : SUCCESS in 35m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f74d0f88129543399c9ab5f1eca9d333 : SUCCESS in 1h 35m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5583319287d04ca5a9e23f7144ad2e14 : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b74229d01c6048c9afb25f0c329087d1 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbb3e051ed04badabd29f78b8ca53f6 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23d8c95fc3b44efd900c6fcde7f3f0ec : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cfcc344ef794825a1d9e2ef8c40db6d : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a33a1a1ae6634f3c92b864ab284f8632 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0243195002994b3a9ee3dc8c7aa30269 : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/901214d0558d46ad800335286e12b18c : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":3},{"id":"54e7af0c2210bd5a0d05b9d28dbf01bae3241220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:11:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8d1a4fd02301b3b9feee046de0729da663e38dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 05:30:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/de0478d68732426cb120071410fb9da1 : SUCCESS in 31m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3445d30cbb2c40949c929953bafd75e0 : SUCCESS in 37m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b80955d21e124d8da83ba62dbc7458b6 : SUCCESS in 1h 16m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2715ba8584734e5b889bb900ff9dec3d : SUCCESS in 35m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/60646e5761214594a890c40abcbe43ea : SUCCESS in 31m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f5fc63742a64816b04d8fd62d428596 : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2e7147824fc4499825a8d7b842b22e5 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe4493f6f55d4a66906e553cd5d179cf : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94bc489bdbc448cf8e6fe86257b5b964 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a22ad7032fa04ca78c5042e7eb110b0f : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7d08d5d18364b1787f07c158e206614 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e90b7fd414446529f7627d346f2a18b : SUCCESS in 6m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fa9cf04397e0498890d2f6400ec8f0c2 : SUCCESS in 27m 17s","accounts_in_message":[],"_revision_number":4},{"id":"41e144fe8724c19ed206e6620f1691e6eb22b992","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 13:21:04.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"251a11a4fe3fc01f44b8ced0ab3eb916e45fcf9a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-26 22:35:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9c0521acea9dfc54b01489d25974b5c30a39184c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 23:34:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e5c83af401670846c3f0a685be70ee06b61effd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:46:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bc12c8c4015d48048d6538e5333e4e8b : SUCCESS in 35m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/af7fafad5435492d8243db0b9cc43297 : SUCCESS in 34m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/82ab2603167c47a993bff2f1ad644e78 : SUCCESS in 56m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e7906367adc2451ab5f8524a181a9bda : SUCCESS in 32m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/166d89ceecf2450eb67a2aa562c1ee60 : SUCCESS in 33m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f8ebde8440e48b483d3ca4e107be12f : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e15d8d0bf64e05b1f5aa4d35d8cbc0 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f03fe4505b904855a70c4b55d3afa2a3 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31b677b7d09b46a992c984f967cdef8b : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/834478b2ec54482f986dcc2989422a2e : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":4},{"id":"722083f82226de02ff73afd17c93d330ed118939","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:46:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e9ecd1b5fedd0fed92c61c7fbcba1de5b9f28eb3","revisions":{"9bfa8096f8a5c52047698ee90ae3c9ee1048178a":{"kind":"REWORK","_number":1,"created":"2020-11-23 22:49:51.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/74/763874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/763874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/763874/1"}}},"commit":{"parents":[{"commit":"2155f6fb50df724d857761df3dac787f4001df4d","subject":"Implement secure RBAC for quotas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2155f6fb50df724d857761df3dac787f4001df4d"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 21:15:26.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 21:15:26.000000000","tz":0},"subject":"Implement secure RBAC for records","message":"Implement secure RBAC for records\n\nThis commit updates the policies for records to understand scope checking and\naccount for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I093223790b699e35fb151e17111e34c24153d4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9bfa8096f8a5c52047698ee90ae3c9ee1048178a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9bfa8096f8a5c52047698ee90ae3c9ee1048178a"}]},"branch":"refs/heads/master"},"9b36639e536ac3b9ecaec580ba97297a10d13e6a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 00:50:53.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/74/763874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/763874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/763874/2"}}},"commit":{"parents":[{"commit":"2aedbbc255ef95791c707f04c20f8c1048cbddcf","subject":"Implement secure RBAC for quotas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2aedbbc255ef95791c707f04c20f8c1048cbddcf"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 21:15:26.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 00:49:28.000000000","tz":-420},"subject":"Implement secure RBAC for records","message":"Implement secure RBAC for records\n\nThis commit updates the policies for records to understand scope checking and\naccount for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I093223790b699e35fb151e17111e34c24153d4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b36639e536ac3b9ecaec580ba97297a10d13e6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b36639e536ac3b9ecaec580ba97297a10d13e6a"}]},"branch":"refs/heads/master"},"4b453e8be59bee6c3a9a70b72a70032345462c0e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-11-24 01:04:14.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/74/763874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/763874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/763874/3"}}},"commit":{"parents":[{"commit":"b958bd3d16e39f22bbe9466c9cce26ee21c343db","subject":"Implement secure RBAC for quotas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b958bd3d16e39f22bbe9466c9cce26ee21c343db"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 21:15:26.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 01:03:16.000000000","tz":-420},"subject":"Implement secure RBAC for records","message":"Implement secure RBAC for records\n\nThis commit updates the policies for records to understand scope\nchecking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC\nexperience and improve security.\n\nChange-Id: I093223790b699e35fb151e17111e34c24153d4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b453e8be59bee6c3a9a70b72a70032345462c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b453e8be59bee6c3a9a70b72a70032345462c0e"}]},"branch":"refs/heads/master"},"e9ecd1b5fedd0fed92c61c7fbcba1de5b9f28eb3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-11-24 04:11:24.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/74/763874/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/763874/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/763874/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/763874/4"}}},"commit":{"parents":[{"commit":"5bab4daa36d39f09c343546797e2148d9168f848","subject":"Implement secure RBAC for quotas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5bab4daa36d39f09c343546797e2148d9168f848"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 21:15:26.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:08:05.000000000","tz":0},"subject":"Implement secure RBAC for records","message":"Implement secure RBAC for records\n\nThis commit updates the policies for records to understand scope\nchecking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC\nexperience and improve security.\n\nChange-Id: I093223790b699e35fb151e17111e34c24153d4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e9ecd1b5fedd0fed92c61c7fbcba1de5b9f28eb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e9ecd1b5fedd0fed92c61c7fbcba1de5b9f28eb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
