)]}'
{"id":"openstack%2Fdesignate~763879","triplet_id":"openstack%2Fdesignate~master~I8ad4e61337f00a2c7b0019a6748c2fab42a65255","project":"openstack/designate","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I8ad4e61337f00a2c7b0019a6748c2fab42a65255","subject":"Implement secure RBAC for tsigkeys","status":"MERGED","created":"2020-11-23 22:49:51.000000000","updated":"2020-11-27 00:47:21.000000000","submitted":"2020-11-27 00:47:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"763879-secure-rbac","meta_rev_id":"9668351fd7f47433a77c07ba86dca5565cc0c8e3","_number":763879,"virtual_id_number":763879,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 00:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-11-24 13:31:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-26 22:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-26 22:38:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}],"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 00:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 00:50:53.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-11-24 21:05:35.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"CC"},{"updated":"2020-11-26 22:38:29.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"8a59a934ccb4205624bda7fb2325798be0b676f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e01af8f24721098cb7486a01eeedfddd7b6e160","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:53:51.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"cb5e8692ffd03faba360e6626ee1dced5745b6e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 00:16:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e9af8d88e01e48d58a40a0da07748b40 : SUCCESS in 36m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/19fb68b7869a4cccb67ee358e2cbeb28 : SUCCESS in 41m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5b24cfa4fcc04763833802b6addf062c : SUCCESS in 1h 01m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6c22be2fbe5b44888a7c1918c58386cb : SUCCESS in 41m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f0f8d8c72f5f4c4683a16b19a82b4574 : SUCCESS in 45m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02a409bb0844432f8962fbc7c9af32bb : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/daf47164021d446b810582d19d3e42ed : FAILURE in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/595b3e8369924a9dadf024be71813943 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bfe1ce2a70548779687dd23e9298bc9 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3f0a011a77c4a989e5fe353a43575c9 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a42fe5bc6104b4eba4718f9a9d345a8 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9c04e341704cbda277f47bb55a25d9 : SUCCESS in 6m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4667118e8b3941e7bbbbf3066c9e8550 : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b33a6cedaf2107bab7a7b09cd097164efffb6cb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 00:50:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2a3ca6e42c21d87e45c6ee9a107951b8be9a49ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 01:04:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2fabc1a505d894aa7999ae178be1e28712394057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 02:07:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/63ad92e968604269bc083317f07bdb86 : SUCCESS in 32m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/87fe78076b1c4962aed73453c6c8ece7 : SUCCESS in 40m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c8199bb380574a88ab0dca19277c142e : SUCCESS in 59m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1bdec36f89af44a0bd1dd4292e1d627a : SUCCESS in 34m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9b8eeb9a2a2549c69ef43c3a7b0ccce2 : SUCCESS in 43m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8a0bdce637543f09a5cff1ade0239df : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa2bb10cbd1c4645b882623bfd31aeb4 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac25f4c52f5404997b30894ef5785fa : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0b222dbc0884109833bff37dc88cc00 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/173786f03dfb47fa965662edecfdea56 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9a66ceddb704ac0858f329612991015 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a435900804a4c6ea68fe92f394d7ca1 : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6604c686082c4aec837e582e44861c4f : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":3},{"id":"8bf439045058f773a40cf36eac6285119027b722","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:11:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"76ac910851d301398c46d6c8aef10d70820b52f3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:16:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b0c4f25b8dbb14ba706e2936e0b6cdfc2365bd45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 05:03:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/44f1bbbb785e453fb907fd3b8d8baec7 : SUCCESS in 37m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/026eee326ab04de3af498a955f9e1a95 : SUCCESS in 35m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5dff7b3ec9b948bfa751b88853c23c11 : SUCCESS in 47m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d21f61b756414605b4252468ecac63ba : SUCCESS in 37m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b25f6d8b896d435581544a6c7c94065f : SUCCESS in 34m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8d2523d42964330beb69bee56dd6263 : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/617fa4cc85bf4def8c78dd9bd3413e9f : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/442aed9d607c4320bbde500fb20cd5ce : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a82f25e67754c2ca02a42e4474e3fda : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/793ec317cb5f4dd88e2b3654bff84152 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8904a042c75d416ab04a2e2f507908cf : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47de2f650d00454b84299d3c97bd4fc2 : SUCCESS in 7m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d212ed0af6246beb56b25b338f21933 : SUCCESS in 27m 34s","accounts_in_message":[],"_revision_number":4},{"id":"9a97ef86328f09bd5413fe91d2074408b9b92d66","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 13:31:32.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"958f46dd03dffd530002f20221afb82589184dc2","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-11-24 21:05:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f2abdae8ce396fabb0303e51a729799bb31f9db5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-26 22:38:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"490aa837cd75a906123f7dc9a50ebb984f884723","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 23:34:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2e1bb4bea5938c61d2bf3651f9e9b7f1e5eaaf4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:47:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/07059e1d79324b83b134a9327f7a80d4 : SUCCESS in 33m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0abdb708cd3743abb9dd0ee98dfdcb86 : SUCCESS in 34m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d88210a358344c5ca0f7165aeb1eb22e : SUCCESS in 50m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0e5eebb0ab604720b68005682f6f232a : SUCCESS in 33m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dbd63f8af7004c90aea59ca34439c2e6 : SUCCESS in 41m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f56689fef2ae4cce8b33ef12130cb149 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c2ffd31c71648e4a88592eab0c944a0 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22726bc5daa24068b5fb9891b78f1cd7 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62d8b075da12493a8400dfac10a32754 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f85fc29cb0934e3087253748670eba9c : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":4},{"id":"9668351fd7f47433a77c07ba86dca5565cc0c8e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:47:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"da1c94e47b5829fe14abd91bc835dbb83a984e9d","revisions":{"adb5ff45768af8761458b0b57dd5832e18a5c196":{"kind":"REWORK","_number":1,"created":"2020-11-23 22:49:51.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/79/763879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/79/763879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/79/763879/1"}}},"commit":{"parents":[{"commit":"3253551904c8626861db155770254ebd0cc2c55d","subject":"Implement secure RBAC for top-level domains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3253551904c8626861db155770254ebd0cc2c55d"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:04:16.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:04:16.000000000","tz":0},"subject":"Implement secure RBAC for tsigkeys","message":"Implement secure RBAC for tsigkeys\n\nThis commit updates the policies for tsigkeys to understand scope checking and\naccount for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I8ad4e61337f00a2c7b0019a6748c2fab42a65255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/adb5ff45768af8761458b0b57dd5832e18a5c196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/adb5ff45768af8761458b0b57dd5832e18a5c196"}]},"branch":"refs/heads/master"},"9277ab7f37cfbcb6c4a6bc258c25a265ac0d4a28":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 00:50:53.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/79/763879/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/79/763879/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/79/763879/2"}}},"commit":{"parents":[{"commit":"7c088bae546df39141be524be053b0add66c6d53","subject":"Implement secure RBAC for top-level domains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c088bae546df39141be524be053b0add66c6d53"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:04:16.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 00:49:30.000000000","tz":-420},"subject":"Implement secure RBAC for tsigkeys","message":"Implement secure RBAC for tsigkeys\n\nThis commit updates the policies for tsigkeys to understand scope checking and\naccount for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I8ad4e61337f00a2c7b0019a6748c2fab42a65255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9277ab7f37cfbcb6c4a6bc258c25a265ac0d4a28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9277ab7f37cfbcb6c4a6bc258c25a265ac0d4a28"}]},"branch":"refs/heads/master"},"e041d33606a5da3fb92df4c102ef60a393d7b5e9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-11-24 01:04:14.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/79/763879/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/79/763879/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/79/763879/3"}}},"commit":{"parents":[{"commit":"2d7a382a49403ee71f80a1347d56c02763c52734","subject":"Implement secure RBAC for top-level domains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2d7a382a49403ee71f80a1347d56c02763c52734"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:04:16.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 01:03:34.000000000","tz":-420},"subject":"Implement secure RBAC for tsigkeys","message":"Implement secure RBAC for tsigkeys\n\nThis commit updates the policies for tsigkeys to understand scope\nchecking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC\nexperience and improve security.\n\nChange-Id: I8ad4e61337f00a2c7b0019a6748c2fab42a65255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e041d33606a5da3fb92df4c102ef60a393d7b5e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e041d33606a5da3fb92df4c102ef60a393d7b5e9"}]},"branch":"refs/heads/master"},"da1c94e47b5829fe14abd91bc835dbb83a984e9d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-11-24 04:11:24.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/79/763879/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/79/763879/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/763879/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/79/763879/4"}}},"commit":{"parents":[{"commit":"e99f3588f1624f806265b264db1f82bf56fdedb9","subject":"Implement secure RBAC for top-level domains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e99f3588f1624f806265b264db1f82bf56fdedb9"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:04:16.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:08:05.000000000","tz":0},"subject":"Implement secure RBAC for tsigkeys","message":"Implement secure RBAC for tsigkeys\n\nThis commit updates the policies for tsigkeys to understand scope\nchecking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC\nexperience and improve security.\n\nChange-Id: I8ad4e61337f00a2c7b0019a6748c2fab42a65255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da1c94e47b5829fe14abd91bc835dbb83a984e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da1c94e47b5829fe14abd91bc835dbb83a984e9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
