)]}'
{"id":"openstack%2Fdesignate~763882","triplet_id":"openstack%2Fdesignate~master~I319b2398de9bd9d841bfb3bbdbe8f50434762602","project":"openstack/designate","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I319b2398de9bd9d841bfb3bbdbe8f50434762602","subject":"Implement secure RBAC for zone imports","status":"MERGED","created":"2020-11-23 22:49:51.000000000","updated":"2020-11-27 00:48:18.000000000","submitted":"2020-11-27 00:48:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763882-secure-rbac","meta_rev_id":"32ba198336fc3f6523b9f025421b79722b650fbc","_number":763882,"virtual_id_number":763882,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 00:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-11-25 03:32:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-26 22:39:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-26 22:39:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 00:22:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 00:50:53.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-11-26 22:39:48.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"adb99e23923b0f0003c3fc2390ab2b16e1a3bd94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8dbd28f2a2e1b58b7c139843cd672542c7884e4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:54:07.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"5d1d2ea7e1eda10641e2202c99539ff093b7b317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 00:22:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1877f0d329ea4298873c2d7d866294bb : SUCCESS in 34m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1d49daf12ff6426380c2cc763f071680 : SUCCESS in 43m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9e7095b660d245b28caa8bee0f59c551 : SUCCESS in 58m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4b542156aa50499984c677302f256347 : SUCCESS in 40m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/72ef60bab7fe44529fe79da92ec912ac : SUCCESS in 44m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a804297b9a3a470488feacd442cb062c : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20694b64eb0a43eb9f19e7aed06998a1 : FAILURE in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca0c773079ae4abb93538e189495e9c9 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b4d660a32ca43a99c11f1bda9410438 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ad22b18c32b4ad79c88716f3e7903a9 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c5bee8250b3415b8e7bae0f75cf81de : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6662e5e4b9a1492d8d9cf5d67183a9fb : SUCCESS in 7m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d4ea3ddcb2e4ae795bcffc843f825c4 : SUCCESS in 40m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2305c57fa03f416d49c4d5e1adf05a068541cce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 00:50:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c0e1a70c5486a4f8b12894addb99e95075eb4c20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 01:04:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4dfb26087160952239f9b82423a8d11c405d5d71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 02:01:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bcee96fa141347feb449d904aa274241 : SUCCESS in 42m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cedc5876e5244fabad5584ce40b0c7e2 : SUCCESS in 44m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7bfc6f8531d546cdb2dcfb421262c1f4 : SUCCESS in 52m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3bb97b1a0f3c4caaafec7b6a4e49af09 : SUCCESS in 32m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8968a3f4afb5451082b8a5a02146c992 : SUCCESS in 39m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a779141953b64f2aae05aa3d18cd0d7f : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42b35ff709004b748f3d04cd5854cdc8 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c57fba47a954016b94d008265c26c7b : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b9967786e744cd7b9640c4aafc56cab : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb86e607009a42f6b7d8cf9ba9111bbb : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e83faa93cd4b42efb136e5f1a00125e1 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26415fc965c948828ee50a7c94d0efbf : SUCCESS in 6m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb545a3f51fa4154bab114719ead2aa8 : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":3},{"id":"bff61e7e20c8b43e119113899585a0bb579dfeba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:11:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"f0a8f2433109520dbc194c158d039a260f310039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 05:29:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6f6992fac2a84151be1f2d0e5dcbd7d7 : SUCCESS in 43m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ef0f2417b2fd496da7223c531460d971 : SUCCESS in 44m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1982d9f4288540c4be2da96f422e24e3 : POST_FAILURE in 59m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/063ebb70a5484521bd3b114f89c11c32 : SUCCESS in 31m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/462459c82da9479b8d1af2f5a2287bc6 : SUCCESS in 43m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7406aea4799941b88cd247b442c2c1dc : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b5c635379b24085b725a4b400279ddf : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/767b3b10a47549a8bf105acee7525c1e : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5b18cf9e25b4c85877cdcc260b5f4ee : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b68fa7d0737a49e8bda5bfaed586b025 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07cecb5bc3b84319a3faffec97443154 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3e02b2106f4195841d73e3e9f5815d : SUCCESS in 6m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0744e68f3b1e419a84c9d7461ed0724b : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":4},{"id":"76a04736d4b86300d9723ad5521c7dd86ecc74f2","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 17:41:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"61c0d40890e3fd2b2b9e9eac3e8714d06ea292ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 19:13:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ac5785e22e1947ee9ceab36a050fbb22 : SUCCESS in 46m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7c4bf88856af40169e19a45f5cbb00ef : SUCCESS in 40m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4bfeaa9051094865be065c93b0fa2fbd : FAILURE in 1h 23m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b50ab51c29c74037b21bfd981924367f : SUCCESS in 44m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6e4e7abb6a24432ba27dce1ac8f7836c : SUCCESS in 36m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3f8d5cb856349f09d3d19d098f22be1 : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d587fb41154b42dc88cdf214e8f23d67 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e5b8eb43604ea8802b80de26bcda13 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cf1611f1d744d4d8b38f5ab334d03f2 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f851b9b0082e4f26ac320b79d850ff04 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa9c1773641044c287e886a32bf95002 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8badbe3b8f714f8f8a7b447f3fe3ae37 : SUCCESS in 9m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fb322c27787468abab1492e7c58ab16 : SUCCESS in 33m 15s","accounts_in_message":[],"_revision_number":4},{"id":"f6d6383a4b1b117c892c1bd5fed3357b3dc55b36","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 22:00:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"706e32824f74878c90bfd493cd5f58a96851293a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 23:15:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3eeea7c3009d42ab89dd2f8bf17d456c : SUCCESS in 41m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b4b88f5ea5594a21a04a1ecd973acb05 : SUCCESS in 43m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d4c325b329db46b387e5d487da36ee01 : SUCCESS in 1h 01m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d45203875db44f4494f33d89245a629d : SUCCESS in 39m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/66519668186e4919abdbbcb0fa5bd630 : SUCCESS in 44m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15d52be836204b7096e4a4e109bfd536 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cafe3a08be34d4b9abf5fa4917d7be3 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f81536fb204071b9fbd804c7cea8da : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4f12cc362dd4514af24bd90f7e62dae : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d4c6092617748d9923ba5f62a92d1c1 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/437269179ec547c68714cc6b54c45e0d : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69d4bdaa2d6c4a6ebd69d36d525f639c : SUCCESS in 8m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1e09d80334a49c192aa253b7a0b7ae7 : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":4},{"id":"6ce9f73f3a1f33969d9daf2a92ee6ed23621f1af","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-25 03:32:30.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"805de1ce034d664481acc76f103f9e4feb902a80","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-26 22:39:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8364d7b0293019125996bf7304c27dd77be0ca5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 23:35:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a3e0c301af2efbfce7f6e8d6138ba81637880783","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:48:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fb8fe283803c458b843ba5a4caa19031 : SUCCESS in 39m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/549baa7d03894fe48a2e9b0f1ab36342 : SUCCESS in 42m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e71e52e522ab429cb9cb6ec3a058aea4 : SUCCESS in 57m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8cfb0e98011045cf908c7301acac5782 : SUCCESS in 44m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/55338cd4bd194657ae6d33967bd77590 : SUCCESS in 36m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f91f16fda07a4f9590c1dc9999be59b0 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1adae6e1bde4841af517abf34bd0bba : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1282b0c30654d3ebbea3639f0516f99 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4574e050d3d846e3a37075c6fd19fdbd : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bf08d92c79d4291948e68a96b3f0fa1 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":4},{"id":"32ba198336fc3f6523b9f025421b79722b650fbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:48:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"40eb2626f0a1ab6d07c3913c6e03a92e70f112e3","revisions":{"c8936b4d094b9d7572cf16c9e629ec3d678e0d98":{"kind":"REWORK","_number":1,"created":"2020-11-23 22:49:51.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/82/763882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/763882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/763882/1"}}},"commit":{"parents":[{"commit":"b4886bca7e27a6927826140691b7e52ecc3be544","subject":"Implement secure RBAC for zone exports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4886bca7e27a6927826140691b7e52ecc3be544"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:31:26.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:32:05.000000000","tz":0},"subject":"Implement secure RBAC for zone imports","message":"Implement secure RBAC for zone imports\n\nThis commit updates the policies for zone imports to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I319b2398de9bd9d841bfb3bbdbe8f50434762602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c8936b4d094b9d7572cf16c9e629ec3d678e0d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c8936b4d094b9d7572cf16c9e629ec3d678e0d98"}]},"branch":"refs/heads/master"},"c3a78626934e5389c5c6fba3a087fdc3a1441ef2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 00:50:53.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/82/763882/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/763882/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/763882/2"}}},"commit":{"parents":[{"commit":"6fb6ef1e904893cf8d6a80c802e96f64bd45efc8","subject":"Implement secure RBAC for zone exports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fb6ef1e904893cf8d6a80c802e96f64bd45efc8"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:31:26.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 00:49:31.000000000","tz":-420},"subject":"Implement secure RBAC for zone imports","message":"Implement secure RBAC for zone imports\n\nThis commit updates the policies for zone imports to understand scope checking\nand account for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I319b2398de9bd9d841bfb3bbdbe8f50434762602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3a78626934e5389c5c6fba3a087fdc3a1441ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3a78626934e5389c5c6fba3a087fdc3a1441ef2"}]},"branch":"refs/heads/master"},"d2c04adbab296e4b2d32aa708e94f630ef9d548f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-11-24 01:04:14.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/82/763882/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/763882/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/763882/3"}}},"commit":{"parents":[{"commit":"44ab2a5e6faebe66d90728f4ab6d840468ec3b38","subject":"Implement secure RBAC for zone exports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/44ab2a5e6faebe66d90728f4ab6d840468ec3b38"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:31:26.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 01:03:44.000000000","tz":-420},"subject":"Implement secure RBAC for zone imports","message":"Implement secure RBAC for zone imports\n\nThis commit updates the policies for zone imports to understand scope\nchecking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC\nexperience and improve security.\n\nChange-Id: I319b2398de9bd9d841bfb3bbdbe8f50434762602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2c04adbab296e4b2d32aa708e94f630ef9d548f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2c04adbab296e4b2d32aa708e94f630ef9d548f"}]},"branch":"refs/heads/master"},"40eb2626f0a1ab6d07c3913c6e03a92e70f112e3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-11-24 04:11:24.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/82/763882/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/763882/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/763882/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/763882/4"}}},"commit":{"parents":[{"commit":"d9ee5be3b9f13489cfcbdcb0ec200254a254df08","subject":"Implement secure RBAC for zone exports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9ee5be3b9f13489cfcbdcb0ec200254a254df08"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:31:26.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:08:05.000000000","tz":0},"subject":"Implement secure RBAC for zone imports","message":"Implement secure RBAC for zone imports\n\nThis commit updates the policies for zone imports to understand scope\nchecking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC\nexperience and improve security.\n\nChange-Id: I319b2398de9bd9d841bfb3bbdbe8f50434762602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/40eb2626f0a1ab6d07c3913c6e03a92e70f112e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/40eb2626f0a1ab6d07c3913c6e03a92e70f112e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
