)]}'
{"id":"openstack%2Fdesignate~763884","triplet_id":"openstack%2Fdesignate~master~I56ae44c6ae302d521d8ec52c871f77ef1bbb072d","project":"openstack/designate","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I56ae44c6ae302d521d8ec52c871f77ef1bbb072d","subject":"Implement secure RBAC for zone transfer requests","status":"MERGED","created":"2020-11-23 22:49:51.000000000","updated":"2020-11-27 00:51:14.000000000","submitted":"2020-11-27 00:48:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763884-secure-rbac","meta_rev_id":"055f043d8f5e13ccdb532e288b4621c3e0efae2c","_number":763884,"virtual_id_number":763884,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 00:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-11-24 17:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-26 22:40:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-26 22:40:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 00:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 00:50:53.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-11-26 22:40:24.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"7dc8c5407d19c749a3caa51d091c74a06ac7ddb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e0def92d6c5581be4ada5a70b8e168bb25bb0ce","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 22:54:25.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"11c05823b9ebf411e08e06cc60ad615966cae6a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 00:24:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/618e11340d364275b01a739fe01a4650 : SUCCESS in 43m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0a08318ff8ad406cbd6c0697beb25522 : SUCCESS in 45m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9a2bad188aeb4794b79cef3598bd1f3d : SUCCESS in 48m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/386972f005e44eed8b6b11925d021712 : SUCCESS in 59m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a9d62144981f42dca2fa9d3d686dc01c : SUCCESS in 38m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6726c55f0e9a4f1fb7ec13cb324a151c : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/780b59ba1ceb43a29232a9903edbcb5e : FAILURE in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93d45f1e37d34d4288605e5415a53427 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c08757ed2fc54943baae26e7d1b37857 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eadf8870dd3c47a78d14bb399465dd97 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/173a3f8a2e2345c8901229c90e2cfe08 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/070d453d07cc4ad380d96121dc0cfe86 : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6377157765a44b62a2958a0f865da7a1 : FAILURE in 35m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e5b4c474ddf451a41f81e847981545edd57d13bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 00:50:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e273425fbbdcd1db12602f72eca60d20efecbb70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 01:04:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"020f35dd6de9443d7f241dbbb93f20cc474db43c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 02:11:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3dd685aa16a84311a1e3d09d314a5e3b : SUCCESS in 43m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/74510fc08150463daf2fcad8e1e71666 : SUCCESS in 39m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0dba58baa835487e936330ec08cc2336 : SUCCESS in 1h 03m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/493efccd2a04440abe42ab9ac6699b8c : SUCCESS in 34m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0fdde3c4477941049024986df152c5fb : SUCCESS in 44m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/982743b667dd4edb94c0dc5cb8c5402a : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/950c1eabee8a4048a4f2060f9e1d1865 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3c7784fbfe54b2b9ff00d87b20d9648 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fee177246414046ab0be428ea93d17f : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a9506042668402d9b8a840aa5eae314 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6aebf8cccce4ec9993b251093562d65 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d11f5a263348beaa6c57efe9ad23dd : SUCCESS in 7m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/492ffd0386a340808a91f191de79bbe2 : SUCCESS in 28m 35s","accounts_in_message":[],"_revision_number":3},{"id":"d7d35d93b232a9371d6cbe7ef59c6632d9cce4cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:11:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"1f1a6194ae03bfd6936b6870b7c12fa84f5c8f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 05:22:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/87601222a967430e864ce29c55118c33 : SUCCESS in 32m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8d528e3b319b415babf2bd07fa177641 : SUCCESS in 31m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/91e79264d72b4c79bd636097469685a3 : SUCCESS in 1h 02m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bae6ca972eae4fcc9c1ba8948648945e : SUCCESS in 39m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4157528a34984f18ae521f13b1b1c23d : SUCCESS in 42m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ceb5d1cb1d6473badef7d61d5194bc9 : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7074a86221a249739dc9a30bf48240b4 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12409fd5caaa48eb933b53e05a5e7903 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a51ba2380724e40bcd5199a04d188b4 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64dae5a467d74f78850a579460fff114 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04cc9dabd69049d29f196147dd48430b : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79f584a32d24c7fa0bb978f0e84b51a : SUCCESS in 7m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/69ba543dba44465a9937f36e0f094940 : SUCCESS in 26m 24s","accounts_in_message":[],"_revision_number":4},{"id":"3832443d54e014f55d9826c2401061a150f1a15a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-24 17:43:10.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"71f69b6db02014acf6f03eea4c395afc29853c41","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-26 22:40:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"22bc848cbf46a42db8a8a7c569d4aecbd111d126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 23:35:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"90a5b9a0a5a326cfdc4b29a1e8f38d03a8358f80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:48:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b975894fcd4b42f5afa4fff39e32cf05 : SUCCESS in 43m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c2a4751d7fd04328ab35fff4ad438e42 : SUCCESS in 41m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8280c7f5d052492784cb7de813de4146 : SUCCESS in 59m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f4097465bba5440383bc8c5ab2e5c818 : SUCCESS in 43m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2a3648d25b1d439c926844e52bc52fef : SUCCESS in 41m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b70add76e7c3444fb4cce372ac9be457 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a4c76244dde48d5abf371caf74ca005 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b9f1209a5284617aa2d1069173b5964 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7767b3de59ac490bba5009ad7435ae9d : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad19366272eb4725bc8eebaba35cfc27 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":4},{"id":"2ceb534931dc10bf371026d772c62b7fb05dc625","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:48:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"055f043d8f5e13ccdb532e288b4621c3e0efae2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 00:51:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0797ee42b7844b280a84e49b3eaf4bc : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1ea6d44a51df01f5a15eca2d380c849b8e297941","revisions":{"016f17a7505e295c21e4d00bf31d76613b4c5b0c":{"kind":"REWORK","_number":1,"created":"2020-11-23 22:49:51.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/84/763884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/84/763884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/84/763884/1"}}},"commit":{"parents":[{"commit":"1288ad1b392b394b87e7767d4e8b2b0cb2205d58","subject":"Implement secure RBAC for zone transfer accepts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1288ad1b392b394b87e7767d4e8b2b0cb2205d58"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:46:02.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:46:02.000000000","tz":0},"subject":"Implement secure RBAC for zone transfer requests","message":"Implement secure RBAC for zone transfer requests\n\nThis commit updates the policies for zone transfer requests to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: I56ae44c6ae302d521d8ec52c871f77ef1bbb072d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/016f17a7505e295c21e4d00bf31d76613b4c5b0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/016f17a7505e295c21e4d00bf31d76613b4c5b0c"}]},"branch":"refs/heads/master"},"585932503f3d5be1f6e0040d4a4aea4c76bc966d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 00:50:53.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/84/763884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/84/763884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/84/763884/2"}}},"commit":{"parents":[{"commit":"6af316f523531b05fcbd0b7bf65a6b095d389b93","subject":"Implement secure RBAC for zone transfer accepts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6af316f523531b05fcbd0b7bf65a6b095d389b93"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:46:02.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 00:49:32.000000000","tz":-420},"subject":"Implement secure RBAC for zone transfer requests","message":"Implement secure RBAC for zone transfer requests\n\nThis commit updates the policies for zone transfer requests to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: I56ae44c6ae302d521d8ec52c871f77ef1bbb072d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/585932503f3d5be1f6e0040d4a4aea4c76bc966d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/585932503f3d5be1f6e0040d4a4aea4c76bc966d"}]},"branch":"refs/heads/master"},"d2eb67ac8d66778b2ab54db34c2fe2acc3c8cee6":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-11-24 01:04:14.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/84/763884/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/84/763884/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/84/763884/3"}}},"commit":{"parents":[{"commit":"ede8c910bdce232bb1ccd933d8d98dccc583ea94","subject":"Implement secure RBAC for zone transfer accepts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ede8c910bdce232bb1ccd933d8d98dccc583ea94"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:46:02.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-24 01:03:50.000000000","tz":-420},"subject":"Implement secure RBAC for zone transfer requests","message":"Implement secure RBAC for zone transfer requests\n\nThis commit updates the policies for zone transfer requests to\nunderstand scope checking and account for a read-only role. This is\npart of a broader series of changes across OpenStack to provide a\nconsistent RBAC experience and improve security.\n\nChange-Id: I56ae44c6ae302d521d8ec52c871f77ef1bbb072d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2eb67ac8d66778b2ab54db34c2fe2acc3c8cee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2eb67ac8d66778b2ab54db34c2fe2acc3c8cee6"}]},"branch":"refs/heads/master"},"1ea6d44a51df01f5a15eca2d380c849b8e297941":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-11-24 04:11:24.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/84/763884/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/84/763884/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/763884/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/84/763884/4"}}},"commit":{"parents":[{"commit":"62cef160c02b3523b48f04809f7b0dcbd69039c8","subject":"Implement secure RBAC for zone transfer accepts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62cef160c02b3523b48f04809f7b0dcbd69039c8"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 22:46:02.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:08:05.000000000","tz":0},"subject":"Implement secure RBAC for zone transfer requests","message":"Implement secure RBAC for zone transfer requests\n\nThis commit updates the policies for zone transfer requests to\nunderstand scope checking and account for a read-only role. This is\npart of a broader series of changes across OpenStack to provide a\nconsistent RBAC experience and improve security.\n\nChange-Id: I56ae44c6ae302d521d8ec52c871f77ef1bbb072d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ea6d44a51df01f5a15eca2d380c849b8e297941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ea6d44a51df01f5a15eca2d380c849b8e297941"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
