)]}'
{"id":"openstack%2Fdesignate~764468","triplet_id":"openstack%2Fdesignate~master~I3f6b5d9871304bb2e4d9b91759b3ba64076fe09a","project":"openstack/designate","branch":"master","topic":"tld_policy_description","hashtags":[],"change_id":"I3f6b5d9871304bb2e4d9b91759b3ba64076fe09a","subject":"Update tld policy description","status":"ABANDONED","created":"2020-11-27 15:01:10.000000000","updated":"2022-09-24 05:53:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"88f8595fc295bcf1b6e18fcd2f745cb56ba13b81","_number":764468,"virtual_id_number":764468,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-27 16:11:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-27 16:11:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-28 07:19:45.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"03e49baa7ef3d8caf0226b97967c1ac012a09541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-27 15:01:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd19b7fbad3d0e9cdeec415834a744574fcbf0f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 16:11:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e72b33446f874bea9ca6487368e7c504 : SUCCESS in 41m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0993cf6d854a4287866eceb88a05e38f : SUCCESS in 45m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/feba351ce6c04701ae3e8854f3127b0d : SUCCESS in 1h 02m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8b39495bc225480ba5447008c005d7b4 : FAILURE in 24m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/29b7a2eb7b6a45a48fbcc574ecd4f6d8 : FAILURE in 25m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f39276602ee944a68c6e2c8368381357 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/676d5153c8bf4a1f8a538c6d560fe25c : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caf2181894ea492da310c56c67617ab7 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac81d0d5f29e4c4f9a7aa2665d768e25 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69933a98953f4cd49459d51c29a8a17f : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fb8a1b9b7b9488ca89803ed8fb5c060 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26b8a78f14d4ac399e356ba3504867c : SUCCESS in 7m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5674cb0788414efeb21abcd34f9ce5f7 : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3e6931976cade9d17c2532413bb2fb65afda15a5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-28 07:19:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e1e6e677c68182f45730c3ab2b3ca87e9d206303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 08:17:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5335f19d29bc4539a7c18b14eedd4980 : SUCCESS in 38m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/51f0f2c0cd4741e4bf7cf2ae6f17391d : SUCCESS in 36m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e4bd3de10c254e6baa69e526ee0d4b3b : SUCCESS in 55m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/71e9acbdbedc4fee8c4ac8f1c7219e31 : SUCCESS in 36m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c36bca77e97f4371ae2f1f2c917d6690 : SUCCESS in 37m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5d46f8d4148457591423572fc5f30c4 : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b686030feb04b8385888dc50d2c466c : FAILURE in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95cb740a739b4eaa9e83611e569f2b81 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc4b3c23e4d9428c96498a825e1e53e1 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/853bf07593a045ebb605f970b8b28ea2 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48d4ea43aca84d15975fa4139e4a7e96 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f601d11043447987ce715fba3c72f3 : SUCCESS in 6m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4eadba819e554efaae02b89abf38ee33 : SUCCESS in 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5869f0e537b57acff3d9627c02101d3b06b85b88","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-29 14:42:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9d9880b9c1865c4593256b262b9ed54bd4ea56ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-29 15:29:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9507b98cdef34456ad7bd814fe0adae6 : SUCCESS in 30m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/263ba27b9bef4b03aa852282e0a8c49e : SUCCESS in 31m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1fc8bfda530a43b19cbc6aa984137b93 : SUCCESS in 45m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/11749e74d1ce4e11a274192dc828455d : SUCCESS in 31m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/870439219c5b4273ba5a1ee38884b5a8 : SUCCESS in 30m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9f8774878c44125a7ddea83135ad4c6 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bbcf14b4ea44e259eb1bb54e9e30f63 : FAILURE in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777078a8d2c8446d9264f08eb253479b : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd0c9a1bf41b42c1b6ae595afc8c5d4c : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2cc8824711e45e094b26cb53a620b6e : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3f3166f77974e73aedb8cf7bd116be3 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff398760b984edf827c618b5dc22dab : SUCCESS in 6m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57b5fc89fda7421091633038d2a6bb09 : SUCCESS in 33m 03s","accounts_in_message":[],"_revision_number":1},{"id":"88f8595fc295bcf1b6e18fcd2f745cb56ba13b81","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:53:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6e8b3c2fe0fda93e79c66664dad3b0642cd3322","revisions":{"c6e8b3c2fe0fda93e79c66664dad3b0642cd3322":{"kind":"REWORK","_number":1,"created":"2020-11-27 15:01:10.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/68/764468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/764468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/764468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/764468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/764468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/764468/1"}}},"commit":{"parents":[{"commit":"1ea6d44a51df01f5a15eca2d380c849b8e297941","subject":"Implement secure RBAC for zone transfer requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ea6d44a51df01f5a15eca2d380c849b8e297941"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-27 15:00:12.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-11-27 15:00:12.000000000","tz":-420},"subject":"Update tld policy description","message":"Update tld policy description\n\nThis change updates the descriptions of the tld related policies to be\nmore verbose.\n\nChange-Id: I3f6b5d9871304bb2e4d9b91759b3ba64076fe09a\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6e8b3c2fe0fda93e79c66664dad3b0642cd3322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6e8b3c2fe0fda93e79c66664dad3b0642cd3322"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
