)]}'
{"id":"openstack%2Fdesignate~766843","triplet_id":"openstack%2Fdesignate~master~I81e7ee3243af11ebb3589f530533731b87178a96","project":"openstack/designate","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"I81e7ee3243af11ebb3589f530533731b87178a96","subject":"[goal] Deprecate the JSON formatted policy file","status":"MERGED","created":"2020-12-13 01:03:32.000000000","updated":"2021-03-09 21:56:06.000000000","submitted":"2021-03-09 21:54:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"766843-policy-json-to-yaml","meta_rev_id":"f7c383d3ad63655dad54c5430e3e43aff428acfa","_number":766843,"virtual_id_number":766843,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 21:54:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-03-09 18:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-03-09 15:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2021-03-09 17:58:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-03-09 17:58:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-13 01:56:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 02:35:28.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2021-03-08 16:20:19.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-03-09 00:30:21.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-09 17:58:05.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"9d2dc39b57f7d6c8a93e1b0b033f4b8b557b52d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-13 01:03:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83e420eca67e70eeb6a9523035c298e0cc0f5355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 01:56:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/db5660b104d44bdb80afaf0153dc42ac : FAILURE in 33m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b34d7d3e517a4778b4ec68ed65ec28ee : FAILURE in 33m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3c54e37eec554702967f9fa305ce5073 : FAILURE in 52m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fc126456b45947f18bcc79011e688d43 : FAILURE in 30m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/30811fcf2d10442eb105d8af6cecbe65 : FAILURE in 31m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd571839bd9b4eaca007500d28acbe34 : FAILURE in 3m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/945dfc274b6043508d254be966026e35 : FAILURE in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e4aeeedceef44c6b7581b22013e1a93 : FAILURE in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ff5da0f9e2541e7936f3dd9752c2807 : FAILURE in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/894aeb8264d54b939141b8091e0da200 : FAILURE in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6a3404763884148b51851ae7e057373 : FAILURE in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f6c0cb280942aea45bb447847aa9a1 : FAILURE in 4m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e6bf1408cc65453c8c76c8c736372d12 : SUCCESS in 1m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e5170efaf2e49a2ba7e916d1d34c04f : SUCCESS in 4m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a334b9921293421c8614f1f57476ed01 : FAILURE in 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"995864b70adbfe6a2aa525941b4eca39811576ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-14 22:49:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bbbed02071bc474317af2a4e98ef942b77e48236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 23:52:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/19892695c0864021b607205d41317e7e : SUCCESS in 34m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ea17c08fabfc403fb9f37d267e4bc24a : SUCCESS in 35m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ff50dc851fbf43058b0d9c1252b98997 : SUCCESS in 58m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d3142ee54e984ae38f48965e084a8b72 : SUCCESS in 37m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1fc96ce422cf477dae5e1ab52d720d59 : SUCCESS in 41m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/115dee6e797f4a8fada49a541a87e615 : FAILURE in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f7a7cd2acfb495a9c2ea663df7058af : FAILURE in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a28516eb2414ae9a35ee2b1ae0e95b9 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b80d118ede049f98fe90cf8917969c3 : FAILURE in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2893e7826b8a4a2289de954bb35e0aa2 : FAILURE in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f40c057a622141b5b6ad79436576d177 : FAILURE in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c3812e37184bc8bfe2f7c3391c2911 : SUCCESS in 7m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b1ae3402e9bd4810b8f5f1623093ad4e : SUCCESS in 2m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a148d640c701404eab9c6cc844ac6f60 : SUCCESS in 4m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fec315d268cf4538a6973a99d60cf400 : SUCCESS in 29m 58s","accounts_in_message":[],"_revision_number":2},{"id":"0871c1d3c1e7592e8288eb0d61e48b3f12989723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-15 18:31:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1aad6dac4027e8deae1c178293128547c7a6ba98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 20:00:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/69cdf08f92b24e288cf661ff1ebcadcd : SUCCESS in 35m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bb887ed6a63942f5b5f4ed037338848e : SUCCESS in 47m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3a31a6863e794f718d0e4bfb2dbab4b2 : SUCCESS in 1h 01m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/02610abf8c144d22a9e2bba79ab55600 : SUCCESS in 42m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5b256617e9ce44929a098014a1336305 : SUCCESS in 46m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dce4ba684ebe4694bd12ffe1194ce2be : FAILURE in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/224b39cc91734ffe9c8c63ee7d4aadac : FAILURE in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cdadf7a6a84462f9fe9807d958aa85c : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab443a15ed474250b5fe2e78c5ce20e3 : FAILURE in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed6f7f70b5144c8a8e338467e2112d28 : FAILURE in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dcefd9a8c94491dbc40090bcd3bdb03 : FAILURE in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5149703fd8343d3877c30616624650e : SUCCESS in 7m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d17e14ee64f042ce874c464250fbf14a : SUCCESS in 2m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/071830d2973d43dea2a8e36a3391a3bb : SUCCESS in 4m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4263047758c945cb8e0c843b922161dc : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":3},{"id":"64afc42ba57e134f4a7594542e618519bb6c62b6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-16 17:16:57.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"08f4b954f20fa0b60e4c21af9742efed4ca01d01","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-14 22:51:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"126c0a3cb8e9f3d68f8bd8436b8aeb1c3167a79a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:33:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ae63c10354b943efa0cc0b90cc689e18 : SUCCESS in 35m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/85803e97390f409a817f7173ee552829 : SUCCESS in 38m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/929ba1d30196442cac0e341bff01d7d7 : SUCCESS in 1h 31m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/51e90f55a25943ddb61b3ac035f2e9a9 : SUCCESS in 44m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/44d621af105940679ebb3c95db49a671 : SUCCESS in 47m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/299976831d6245eb9575cda0b652ddfa : FAILURE in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b71db23bb27748b68357263354f1ee95 : FAILURE in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/980bd4d4b8f64ac4bf259e071e6c17f8 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbf00cecdb28422b87602d9cb34e7b8a : FAILURE in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d8e34a182a74b3991e444608fb44a42 : FAILURE in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/458bebf5b50f4ed284394e714f05325d : FAILURE in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0cb5634f444451ebf67812e9058fc69 : SUCCESS in 7m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7ffc000c1bb443b291e8309afe1f1aff : SUCCESS in 2m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79ee018cc4df405aa9fe2db55a898178 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89547b81aa4448c99bbb5193aad165af : SUCCESS in 30m 25s","accounts_in_message":[],"_revision_number":3},{"id":"5554f0618147e34f7efe8e876298057456745802","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-07 00:57:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a5281231b0445b297bc6eade08db1776d450e66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 01:58:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a8b89bb2e9354ce391b2a99dc503c020 : FAILURE in 32m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bc2cd431614d43f9a928647b85299146 : FAILURE in 27m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b597294c955b405bbd8fec53de0d3139 : SUCCESS in 57m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/46fd3b4960c74b5ba37c5cb7a2bcb875 : FAILURE in 34m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f1ca8df036eb423eb2ee8f0413345b48 : FAILURE in 27m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f86e071150c745c4a2b2198fbbd3ad7d : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cd8ffabb1d3499bb58913c2819d4839 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e76cabd624b94beb913cb61b00ada397 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9beb634c394745ff9851e1ed86f314d6 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d76ff02ae19d482aacdcf8c359a495c2 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/848e8e193154431b94ad570a3d11fa9a : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2522abbfe56542cfa4494d78e3f96d5e : SUCCESS in 7m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa926b798c8a4ab386109462c59bff89 : SUCCESS in 2m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edcedfef90d14eb3ba9436a80467101d : SUCCESS in 4m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd50019fc2af4f0dacd24099cbfee9e0 : FAILURE in 23m 29s","accounts_in_message":[],"_revision_number":4},{"id":"a9fc00e63e825b51b3100237c233a4ae745c17ee","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-07 23:38:09.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"53415c22b3fe9e6b0f27d8139cc5314b151ff8da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 00:22:01.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1b9d462dea2e4a9aa63b85a9b8f69619 : FAILURE in 26m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b9f7c14078a8412f972aabf0e26df63f : FAILURE in 20m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0b96d620172e40ce8874559b3e88dfda : SUCCESS in 43m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5b235d32c141446abcafbefa71bca100 : FAILURE in 22m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0212bbcf22f74c289a8f4ff5b0abb1e8 : FAILURE in 27m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d90cef6b59a24a409585a2a6ce38936c : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd42e739847b4df99c6ff7b072deaacc : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0368e2c140fb4166bc60184b78f2c7c0 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cb9d6b8a56e41298492cd15db1964f0 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50883371401144b3a6bd77eeb2854c7b : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/406b562862a0452d9f6964eefd64ae85 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e2d6c3399b4c768f458c9fb5f5b9e2 : SUCCESS in 7m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a3dc4dc2c85447e39f9116965603faf3 : SUCCESS in 2m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d2bb4baa80046b6847d15c06f185ee2 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6f0c6804ec4640908055a503a51b0be0 : FAILURE in 19m 49s","accounts_in_message":[],"_revision_number":4},{"id":"d6d26980b1f2d13e6de9302035a8013aa4a59db2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-08 01:16:47.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9fcfb6c1b7e0fe633ee806132fbfa25291468623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 02:18:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e9185e47d6574190a42e1274ea8ae9bd : SUCCESS in 44m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7df950cc3fbd4c1b94fd8422f037ac25 : SUCCESS in 44m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9a80812db01a4968b3d5762224286876 : SUCCESS in 1h 00m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5db2525bf2ad41e9a62137ecf25d44db : SUCCESS in 42m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/110d40fad838415891b3b4f3ab9fc22f : SUCCESS in 31m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/040663c8e0934a43b50d8784cb6dbf50 : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e5ff6a980f6486c8ba83b988e8540e2 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9686a3a0a9cc4d378d029eea9dc86886 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8ca0024e9504940b28415853eeb13a4 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd232bec61994357934b60eee9797324 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de3840582a5346f08d4b454c7f5ab886 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c339c96e86a44ae94940b288789ee9b : SUCCESS in 7m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5e7d5247843f4210a4ab51c71b1e9540 : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b4204266d1f4efeb7888b1f272791a5 : SUCCESS in 4m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1a42e9ec499a4156b8f69a3bd28937a0 : SUCCESS in 36m 52s","accounts_in_message":[],"_revision_number":5},{"id":"c7171f6204a3c05182e837c03b9df1d7fa23481f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-08 02:35:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"682ed5a6f0a393142d9943f4ea45ee171ba6d7e4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-08 02:41:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"00a74e0194db301358b26c7f6146a85383be8091","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-08 05:48:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b2c37e7be5898e42c4999e94707a2b60918ab142","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-08 20:04:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"62270bc86317157b1955dea440272a0f00f22204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-08 22:55:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ae7ffe18fe485c81f21ce5aaf58b77aa6b8e807d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-08 22:56:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ade38575c47abfb2d5d31c55f67a06ed37751cc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 23:49:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5339ab04e9ed430191578446402f8ecc : SUCCESS in 37m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ebe6fa58e2a74d8199d393d0447bcbf1 : SUCCESS in 35m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7a2ae309442a43a99f9caca2e60f3266 : SUCCESS in 53m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4abf66cdcf9b44ddb8dd871b7593ae9c : SUCCESS in 37m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9bc826a3d4004866963cdfa9b891d2e8 : SUCCESS in 30m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fde71e4a2e34c998890ce76dc70291e : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/195224729e9446468a836e24522a655e : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38da7eb0ab2746339704c717b7130c3a : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14c65647781a46dea42e6fa66f5ab980 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3647c27c9fbc43739889e054b43df1cf : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90e4415fdcd8491899c00bc9b371dd63 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9e93dce2ac43a89e5aa80dbb963803 : SUCCESS in 7m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b2e3c6e6ce034d29b92150cb4f6ebd90 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbcfab048887431bb64123e0f2b62f25 : SUCCESS in 4m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0b229b9cfb7e405781b9c41e6996e934 : SUCCESS in 30m 11s","accounts_in_message":[],"_revision_number":6},{"id":"5a4ae1e7e4edf3e627742b2c3a6d3f65883f4525","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-09 00:05:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0512a59ca4a82cac12d2211100b86791defe84b1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-09 00:19:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2cbdaac1e7719e42f87e01a0b27fadbe6797e197","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-09 00:30:21.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"836107e2ef32472bdae6136cdc229ebf4d58eeb4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-09 01:01:21.000000000","message":"Patch Set 6: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"13a95df43ac7560fe024c83462b2a9669f8b874f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-09 01:28:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0898d1ad5d776713ac84f25b15d51ad4e3ab086d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 02:29:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ae8bfd350d334b61b58374ddfd214114 : SUCCESS in 29m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7b0eea611ef648648bcdc4d1511cb69f : SUCCESS in 39m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/32defafe1fe0470ea0d08085dc009424 : SUCCESS in 1h 00m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7ad3a1a6b3d44d35a9f3c051ff351219 : SUCCESS in 47m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/64576a526fec4c7097af6fad04160dc0 : SUCCESS in 36m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd5314f0933f44089aa8d7408302cfeb : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e30c3e664bff41e7add9dcdb73099235 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0853918ac0e47b0994403ead058301e : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97b7af5e33194a9fad547aac2ecd9e34 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3559cd1ef3348d2a7543bef79af863c : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c77e49171e2438ea00713330452d9d8 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d793316264442869ac2002c5a7c3b37 : SUCCESS in 7m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7630951307364c9eb74249ba2cebf64c : SUCCESS in 2m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46ba85d8f4354fcebc6a48feea96c366 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ddb0bd14828474493ed403c1d0b5a38 : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":7},{"id":"0b8d8019baf3510595c541c17bfe6a0da1a4ad1c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-09 15:02:22.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"53932ff20a35ccc1216f26cd35734f90bdf9911f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-09 17:58:05.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"2480430eaaf4b70e77eda6dcfae09052534e6bc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 17:59:50.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c583424afbd9e93b41d788d34b65b8820f70d640","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-03-09 18:52:07.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"3eaa650f8bee422381b54eb33dae9b4c37858ea2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 19:10:45.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c0ec6c7699b840248814aafff42ff613 : FAILURE in 20m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/55a879e645a648a8ae1236035e5381a8 : SUCCESS in 42m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/414c6e5380504d8899c7e4dd1be82ba9 : SUCCESS in 1h 04m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4cc8f500c1a04a59a0616182422e86a1 : SUCCESS in 39m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e48cd1d65c3d482791dfec8f8978389d : SUCCESS in 42m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d592c3248ab940a28d8aa15bbe4fb6c0 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e8baba74ea43d7974b073834688664 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c527eb9b51e4c17823bef374e9efd2d : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc852bf51e0949ba8f8291d35549aad5 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98eae9aaac9c4359bb30475979c389b6 : SUCCESS in 8m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ae72852200f7463ca1cc874892f5cec8 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dda4c158c94c4a6f98b65f5962430b00 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":7},{"id":"5f9c3cd0d405dac2e91d518c6bd991c48505fb13","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-03-09 19:13:38.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"eb7bdcf637462c060517874b4eff1e185787ec32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:01:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/679443c6e1f74588b77ec4ded7a71098 : SUCCESS in 36m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4615436a6ffd405cb981be8731ccf572 : SUCCESS in 39m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a0cb5e5806ed42bb910377baed35c11d : SUCCESS in 45m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c5b630ca2eb244bdb18705c574fcab1b : SUCCESS in 38m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0f9eece30cab4fcf928cba29409f3f5e : SUCCESS in 30m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b336e74fa8ba427db43472c63ceb75f8 : SUCCESS in 7m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/888ef87c7314497182ccf8fb10f45066 : FAILURE in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fddcc10cbb4763a18ea755d8ac2b84 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ec981312fc84eb3828a61a1710b59a1 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ed003f1c4d1491a87b71b4932e709b8 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4be4d3e2cd0b4b03a0222bbc1a0f561c : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/357ed45bf5bb442fb447b2d53aaeecb3 : SUCCESS in 8m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/758673e17005465baf1c351759d12b20 : SUCCESS in 2m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e9778365b734a2d8ddfbb057c1f9b73 : SUCCESS in 6m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1cacb59ec5b7497bafd661d356bb192e : SUCCESS in 26m 29s","accounts_in_message":[],"_revision_number":7},{"id":"5470ae705dc9f73433d7c8d30d056d4330e522f3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-09 20:05:18.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"e0ae9ad73a0a867df88b4e9599634f3fec52cf74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:49:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f57adbfffc6f45cb870bfe7cdc9f9a54 : SUCCESS in 35m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/78c58f205c78425599ad7ef7cf58dbf7 : SUCCESS in 38m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f9c644813e4644d2abaa7e8ff3ba2483 : SUCCESS in 40m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7543e2a0d8324efe9473a614d0608545 : SUCCESS in 38m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2eca10a1be6b46e3badf296ddc84195f : SUCCESS in 31m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f5696219b394a95bcca2702d40863ca : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/030842cafc6240b98c9372992efb11b2 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0346b17d267f425481c4d6422482c0c4 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1ee0ec5f9e749189533f32a5859cc40 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4469c8d241574cb6a1d06d180c2cec5e : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4c470a5b0c245d0b8fa8dbe43f3b923 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b9200a6a72418aac7c82d9b38b00cd : SUCCESS in 7m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/028a3168149f4c4fbc8958ad0124edbe : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba016b4ab5e94d4084423e56459a4175 : SUCCESS in 5m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6274d2ed80d141ffbd65341617e789ac : SUCCESS in 37m 31s","accounts_in_message":[],"_revision_number":7},{"id":"cf2298fa9f34b14c16a9115d8e83ee23c59bd2d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:54:33.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"6b122c3a55f5d831e554e33e0e845739a82c60c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 21:54:27.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cd607d36ad2f424db41af5d31976834b : SUCCESS in 40m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6f4e99fe3877461aba3ef724400d3708 : SUCCESS in 43m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/be94b56845484fdcb67cacfd78d927ba : SUCCESS in 54m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a43d900a81d44f89b0790288fdacc572 : SUCCESS in 39m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/99c258e3bfea448f9bdcdb2a83cd68f3 : SUCCESS in 38m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/235601e2acd447f29e6ca345b90af24a : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd8c23bdfb74c72ac5059252f6e6506 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/111261bd16e4478aa4f5e660e4355a6e : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cb6606ae4a04df988211ba9c19ab594 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50d47855ae64983be1415d18172c11d : SUCCESS in 12m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd6b4e713cbc4ba797d3b2a33e7062b4 : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b01357e45d254702b8b43a28b434c2c7 : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":7},{"id":"825af6397f99be1ae1509e98fa2cfed4854f2323","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 21:54:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"f7c383d3ad63655dad54c5430e3e43aff428acfa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 21:56:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/821b34d232ca40938157e32af11591fe : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22170a18f2d44f008d88d2ae1719064a : SUCCESS in 54s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1c0bd99c08e76fdfc0b573322dc858a175b61dbe","revisions":{"9c147f01d1c61ec93a39e32f3ca39ec50fbcb48a":{"kind":"REWORK","_number":1,"created":"2020-12-13 01:03:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/766843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/766843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/766843/1"}}},"commit":{"parents":[{"commit":"1ea6d44a51df01f5a15eca2d380c849b8e297941","subject":"Implement secure RBAC for zone transfer requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ea6d44a51df01f5a15eca2d380c849b8e297941"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 01:03:04.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 01:03:31.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I81e7ee3243af11ebb3589f530533731b87178a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9c147f01d1c61ec93a39e32f3ca39ec50fbcb48a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9c147f01d1c61ec93a39e32f3ca39ec50fbcb48a"}]},"branch":"refs/heads/master"},"c1b1c2e7b7517c7df7071bc15196f50142ca7fb6":{"kind":"REWORK","_number":2,"created":"2020-12-14 22:49:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/766843/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/766843/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/766843/2"}}},"commit":{"parents":[{"commit":"1ea6d44a51df01f5a15eca2d380c849b8e297941","subject":"Implement secure RBAC for zone transfer requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ea6d44a51df01f5a15eca2d380c849b8e297941"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 01:03:04.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-14 22:48:57.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I81e7ee3243af11ebb3589f530533731b87178a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1b1c2e7b7517c7df7071bc15196f50142ca7fb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1b1c2e7b7517c7df7071bc15196f50142ca7fb6"}]},"branch":"refs/heads/master"},"e3439c6d411bfa9f2045fde21ea754fa188cc4a5":{"kind":"REWORK","_number":3,"created":"2020-12-15 18:31:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/766843/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/766843/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/766843/3"}}},"commit":{"parents":[{"commit":"e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 01:03:04.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-15 18:31:18.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I81e7ee3243af11ebb3589f530533731b87178a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3439c6d411bfa9f2045fde21ea754fa188cc4a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3439c6d411bfa9f2045fde21ea754fa188cc4a5"}]},"branch":"refs/heads/master"},"cb7580cb27cf902f3a5cbf7af35e0e494b07fae7":{"kind":"REWORK","_number":4,"created":"2021-03-07 00:57:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/766843/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/766843/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/766843/4"}}},"commit":{"parents":[{"commit":"e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 01:03:04.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-03-07 00:56:48.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I81e7ee3243af11ebb3589f530533731b87178a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb7580cb27cf902f3a5cbf7af35e0e494b07fae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb7580cb27cf902f3a5cbf7af35e0e494b07fae7"}]},"branch":"refs/heads/master"},"15dd5620f60a464f89fde274a817ae3e7d29582d":{"kind":"REWORK","_number":5,"created":"2021-03-08 01:16:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/766843/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/766843/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/766843/5"}}},"commit":{"parents":[{"commit":"e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 01:03:04.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-03-08 01:16:44.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I81e7ee3243af11ebb3589f530533731b87178a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15dd5620f60a464f89fde274a817ae3e7d29582d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15dd5620f60a464f89fde274a817ae3e7d29582d"}]},"branch":"refs/heads/master"},"e2286ce770ca67293b183ee3cfc36e5a2c02ba03":{"kind":"REWORK","_number":6,"created":"2021-03-08 22:55:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/766843/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/766843/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/766843/6"}}},"commit":{"parents":[{"commit":"e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 01:03:04.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-03-08 22:55:15.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I81e7ee3243af11ebb3589f530533731b87178a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2286ce770ca67293b183ee3cfc36e5a2c02ba03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2286ce770ca67293b183ee3cfc36e5a2c02ba03"}]},"branch":"refs/heads/master"},"1c0bd99c08e76fdfc0b573322dc858a175b61dbe":{"kind":"REWORK","_number":7,"created":"2021-03-09 01:28:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/766843/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/766843/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/766843/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/766843/7"}}},"commit":{"parents":[{"commit":"e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-13 01:03:04.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-03-09 01:13:17.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I81e7ee3243af11ebb3589f530533731b87178a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1c0bd99c08e76fdfc0b573322dc858a175b61dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1c0bd99c08e76fdfc0b573322dc858a175b61dbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
