)]}'
{"id":"openstack%2Fdesignate~767923","triplet_id":"openstack%2Fdesignate~master~I02c79a8f16b2d7098d419fc46d44d9c04731610d","project":"openstack/designate","branch":"master","topic":"oslo_lc_drop","hashtags":[],"change_id":"I02c79a8f16b2d7098d419fc46d44d9c04731610d","subject":"Dropping lower constraints testing","status":"ABANDONED","created":"2020-12-18 17:29:45.000000000","updated":"2021-12-24 04:32:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9bf7861bae86949e9b427f1a8c81e5470f1f07fa","_number":767923,"virtual_id_number":767923,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-18 18:30:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-18 18:30:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85b4762cadb11415f1057cbe4a9f23501946d9cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-12-18 17:29:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15c68c63a23787070a12081f34f1befe829328de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 18:30:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8371eb055c1b4c25bb921ce471e5adab : SUCCESS in 43m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7f09d27ebc234687b029c4aa98d9f4e9 : SUCCESS in 46m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/21a9c9af340d4ca1866d72c73961de85 : SUCCESS in 57m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/55e463165e994a20b6161db167a73895 : SUCCESS in 37m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9a1100fcb48f4197b2de05eddf119aa3 : SUCCESS in 37m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1a6e550e39a40c0a7b3e01ab6771c1d : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/810ac38145444575b00f94264b60c354 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57c0572fe79447e0a4ea13a005d44c7e : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2346cf1b1b149cea8fd6974db2e52e7 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/468922d5699d40c9ba5bf403bea17067 : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee99f9be2fb4a20bc3d977d8f6f23f8 : SUCCESS in 6m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e358a26e6ee54a81a80370c7743dca62 : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9bf7861bae86949e9b427f1a8c81e5470f1f07fa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:32:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc723d186709e88e810dcd9b02be5e4094732ed9","revisions":{"bc723d186709e88e810dcd9b02be5e4094732ed9":{"kind":"REWORK","_number":1,"created":"2020-12-18 17:29:45.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/23/767923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/767923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/767923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/767923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/767923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/767923/1"}}},"commit":{"parents":[{"commit":"e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-12-18 17:28:55.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-12-18 17:28:55.000000000","tz":-420},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this topic was\ndiscussed on the ML and QA team proposed to to test lower-constraints\n[1].\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nChange-Id: I02c79a8f16b2d7098d419fc46d44d9c04731610d\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bc723d186709e88e810dcd9b02be5e4094732ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bc723d186709e88e810dcd9b02be5e4094732ed9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
