)]}'
{"id":"openstack%2Fdesignate~771060","triplet_id":"openstack%2Fdesignate~stable%2Fstein~If8738f5005e60cf46ed93edbefa272bc2611b53f","project":"openstack/designate","branch":"stable/stein","topic":"cap-bandit-1.6.2","hashtags":[],"change_id":"If8738f5005e60cf46ed93edbefa272bc2611b53f","subject":"[stable-only] Cap bandit and fix lowwer-constraints","status":"MERGED","created":"2021-01-15 18:38:40.000000000","updated":"2021-01-16 19:30:15.000000000","submitted":"2021-01-16 19:29:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771060-cap-bandit-1.6.2","meta_rev_id":"96c00ffa964e040006ee7158927dda385e80a001","_number":771060,"virtual_id_number":771060,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-16 19:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-16 16:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-01-15 23:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-01-16 16:22:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 18:46:17.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"CC"},{"updated":"2021-01-15 19:58:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-15 20:42:24.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-01-16 16:22:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"2256598e44af4203f2f022794cfe7a3473e4cf6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-15 18:38:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c1db9f8f32b38ff31f46687f0340adf36081ebc","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-01-15 18:46:17.000000000","message":"Patch Set 1:\n\nCould you backport I35780af8c858d44c3c5cb780af72d28e85693352?","accounts_in_message":[],"_revision_number":1},{"id":"60084127ad679ba44dce6d1fc6a3c3fca576a618","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-15 19:56:39.000000000","message":"Patch Set 1:\n\n\u003e Could you backport I35780af8c858d44c3c5cb780af72d28e85693352?\n\nSince this patch is more like a CI fix and (except bandit constraints) the content is quite different from the one in train, I did not see the need to backport that patch, hence the \u0027stable-only\u0027 tag in the commit message. Though, if you think cherry-pick would be better, I can reformulate this patch :)","accounts_in_message":[],"_revision_number":1},{"id":"7e2c4a4af4726fee00b2dfae645d1a2c2bbdc24e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 19:58:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/602cb2b17e1b44d58774f413f5859428 : SUCCESS in 37m 45s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/0f56d32eba0545f8bc4e5d80b6f020fa : TIMED_OUT in 1h 13m 16s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/25fbca9530ae4043b94e293a1457dd49 : SUCCESS in 45m 35s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/80f6d09743244458b68465225401600f : SUCCESS in 33m 16s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8612346ae5e14e41976c4f85ab105b72 : SUCCESS in 34m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4b86a533f8e6480c9103e97a56ba643e : SUCCESS in 1h 00m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b84d0797f78416086b3ee6221f6ed94 : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f58456a01f6a4837b9466daaa3958ad9 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f751f1abd21b4b798b03d347385b0904 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e2d0a1a714a4b38b4a169d866d6f0de : SUCCESS in 7m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cac66a2e6efd4886920c52bdaa234646 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8799b975a9c444b1bab90f686f3f0063 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40158c740f404278b14c62c6a344db7e : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/974220bfb7934039a1db0902d50a9ee2 : SUCCESS in 4m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1daf63077ce34eb4bbf51297786b2cc8 : FAILURE in 2m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e3e3b7fb71c4773bb2d1021fb87aba7 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f6ba96a657c56230e93e84271dccabe892700cd6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-01-15 20:42:24.000000000","message":"Patch Set 1: Code-Review+2\n\nHmm, next time I should really read both change sets before jumping to conclusions :smile: Yes, the change here looks good to me.","accounts_in_message":[],"_revision_number":1},{"id":"82b3ebffc536b417a8737cef88c90e0f44aa56a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-15 22:06:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"784a0aef1b0b5f2ed8a1df916f63cb21a9ece484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 23:15:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1aa7f049c31842a8b21074a8cbf7220d : SUCCESS in 33m 58s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/88a0c3dfae324e7e9dff83b50093a540 : SUCCESS in 41m 07s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/7c2486524fe547b2b61f84d286a4bd2a : SUCCESS in 38m 09s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/11213d9893a846f5b961ad28082c5ada : SUCCESS in 37m 01s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/cba13f47c7334a80b749d6321cdd2354 : SUCCESS in 36m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a8ac6b59873b499799aa8be1be3f9039 : SUCCESS in 1h 03m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/405493c0f35d4340a94c88f3cbe4f2b2 : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a59be9975c184cd185a843448cc0cd3e : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8dcd85583234aaa90b8b7ac2fb31484 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c29128edd0cb4ff0bb431c18308f000e : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/657ae39888724fc392f4ffd4c15366f2 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d64356d5dfc04e4fb03b9abac921b93b : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/308e13c6e7c74555857bed5dd5685dc1 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71bf48f2c3ed4323ac48a4b82cfbeda0 : SUCCESS in 4m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90631716c80a4510b144635580996083 : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2b5373d2e98404589e11ca2190b41a5 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b989cfea51d42f9f0c264b4897dfac9efdccc411","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-01-15 23:42:26.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"8412639ac2e85feeea1767e047a5c2811183bb5f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-16 16:22:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b948171b74b1011942aa47389d5657e3cc31c5c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 16:22:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1949e29469a21fe64bf351649a5f08dedd791325","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 17:35:16.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/79268aff94bf41009bee4e8885099555 : SUCCESS in 59m 53s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/f4ed2edbfc324a029f83cb9b873ed3fe : SUCCESS in 53m 59s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/898263eaa8264b1987a53198ed058328 : SUCCESS in 1h 00m 57s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/2f6cc63f5a19415db1533be578542618 : FAILURE in 1h 06m 53s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/db5e08c4aba44e2c8f7ec15d768b4658 : SUCCESS in 32m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/460f3ab4059e478aa564ab29316d35c2 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60b08cf80e364eb6ad94e2a31e29a4f2 : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19580e6add5a4132996910b5d37ab925 : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d278680a2bb640429bc0def41c7bdada : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/279b43b6b60b489d9cc43731aa6e8439 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c65cb6567a044922ba0bd3098b7bf0dd : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57accf6989944284b11261426bb57f07 : SUCCESS in 4m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fb8052802bcf4c8397c0eb910158b680 : SUCCESS in 2m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec0c0dcd2b804d7ca55c87264480b1da : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"84a2a2dc74ff001e633b3ade0888a53972890d86","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-16 17:46:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c24ee32be651586cbf8b33fed7dadd626096743e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 18:44:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d8332b2c5e994dc4bf51878acfed2480 : SUCCESS in 41m 03s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/98213b9e08bd4f71b64e247543411983 : SUCCESS in 39m 19s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/25dfb0aee5564afbad2af2e9ab6c5c08 : SUCCESS in 38m 07s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/b8cd33af38864934831b102bbde3afce : SUCCESS in 33m 35s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/f4b23fd8e0824b1fa56aca15828ba6cd : SUCCESS in 35m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/765da0b910144561a10a4d30de5fbbe4 : SUCCESS in 52m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/954c2f3ba8ff40569f167cba757cf2e4 : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/415eb8ef0db94348b61e7f95d3cb540c : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29fb685a5a534e23820d22d3730c9c33 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e67db9da10444d38b435742bdafaba6 : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/244fb128599f47ff975839dda4b81a9e : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a814d8feac146a292bc316dca73ab9d : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d26323aa714b40ab9a3182b57d722a5a : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f8e647a85bb4f5ebfb8cb67b75b7a02 : SUCCESS in 4m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/abb225ad44d9424b9f793fd94c170058 : SUCCESS in 1m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84a75c289b5a410a8ce9453d2dc5aa04 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"0ac4dd5a7ade7430076cd8cefc208fe6c606fab2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 18:45:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1261cab0a5f1af097c04a79abe97e128a0c8cdca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 19:28:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e6ae20de5ea74d41994a34a398e4cce3 : SUCCESS in 36m 08s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/d77b476b3844484cbc955dfb7555f36e : SUCCESS in 38m 15s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/cd58416a7d004bb092d94719ae9a4ddb : SUCCESS in 37m 49s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/dbec783702494dcf80d7e9287b9b0058 : SUCCESS in 33m 31s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/90dea470a3ee43d8b99427a3ba9a1517 : SUCCESS in 34m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bf15820db294d69a942d4d6225c12a8 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb17b8132d7648f6bf394122d5613770 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afd51099403745d09662ab8b6929ac46 : SUCCESS in 4m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/af0e2c1efd1d4a2a8485cb62785356aa : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51f418c7754540eaadce293239cbd85d : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd59264df2f54c318231bb2dc073a0ab : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/501779f5fb384c84bb3aa4c7dd4b1c59 : SUCCESS in 4m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2154376ff2764d4cacd447a027667b3a : SUCCESS in 1m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55a3fa52dbaf412b992035cf92d72217 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"ca841e48bdd7c64a5fa28fe9b4ead5b69b22fc6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 19:29:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"96c00ffa964e040006ee7158927dda385e80a001","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 19:30:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fdde04656ae4e18bb66c11f81feb327 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e9abb3bc21d4a0890e81b6319ec6066 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"86544150913bcd819b4ee8d77e83d2566142fb41","revisions":{"42fa348a2fc9cd317c60e9ea1c1ce046fd678d8f":{"kind":"REWORK","_number":1,"created":"2021-01-15 18:38:40.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/60/771060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/771060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/771060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/771060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/771060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/771060/1"}}},"commit":{"parents":[{"commit":"6307c84e83530ab27124ab9e6a8117ee44069614","subject":"Make greande jobs n-v for EM and oldest stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6307c84e83530ab27124ab9e6a8117ee44069614"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-15 18:32:16.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-15 18:32:16.000000000","tz":60},"subject":"[stable-only] Cap bandit and fix lowwer-constraints","message":"[stable-only] Cap bandit and fix lowwer-constraints\n\nbandit 1.6.3 [1] release has dropped support for py2 [2] but the\nrelease is faulty and pip still picks it up for py2 [3][4], so cap to\n1.6.2 when using py2.\n\nWith the new pip dependency resolver (introduced in pip 20.3) the\nlower-constraints job started to fail. Problem is here with the\n\u0027install_command\u0027 in tox.ini, which uses both the upper- and lower-\nconstraints files, causing the job to fail. This patch adds separate\ninstall_command without the upper constraints, so that only the\nlower-constraints.txt is used.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: If8738f5005e60cf46ed93edbefa272bc2611b53f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/42fa348a2fc9cd317c60e9ea1c1ce046fd678d8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/42fa348a2fc9cd317c60e9ea1c1ce046fd678d8f"}]},"branch":"refs/heads/stable/stein"},"86544150913bcd819b4ee8d77e83d2566142fb41":{"kind":"REWORK","_number":2,"created":"2021-01-15 22:06:13.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/60/771060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/771060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/771060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/771060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/771060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/771060/2"}}},"commit":{"parents":[{"commit":"6307c84e83530ab27124ab9e6a8117ee44069614","subject":"Make greande jobs n-v for EM and oldest stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6307c84e83530ab27124ab9e6a8117ee44069614"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-15 18:32:16.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-15 22:05:33.000000000","tz":60},"subject":"[stable-only] Cap bandit and fix lowwer-constraints","message":"[stable-only] Cap bandit and fix lowwer-constraints\n\nbandit 1.6.3 [1] release has dropped support for py2 [2] but the\nrelease is faulty and pip still picks it up for py2 [3][4], so cap to\n1.6.2 when using py2.\n\nWith the new pip dependency resolver (introduced in pip 20.3) the\nlower-constraints job started to fail. Problem is here with the\n\u0027install_command\u0027 in tox.ini, which uses both the upper- and lower-\nconstraints files, causing the job to fail. This patch adds separate\ninstall_command without the upper constraints, so that only the\nlower-constraints.txt is used.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: If8738f5005e60cf46ed93edbefa272bc2611b53f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86544150913bcd819b4ee8d77e83d2566142fb41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86544150913bcd819b4ee8d77e83d2566142fb41"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
