)]}'
{"id":"openstack%2Fdesignate~771562","triplet_id":"openstack%2Fdesignate~master~I815aabbcfaaaf4fce7e29ef7735fd4725d1f9aad","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I815aabbcfaaaf4fce7e29ef7735fd4725d1f9aad","subject":"add openstack-python3-wallaby-jobs-arm64 job","status":"ABANDONED","created":"2021-01-20 06:58:00.000000000","updated":"2021-12-24 04:32:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9640649e55f626b8ee4e6a3d9f6eea756ce4188b","_number":771562,"virtual_id_number":771562,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-20 08:14:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2021-01-20 08:44:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-20 07:10:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-20 08:14:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-20 08:44:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"6c7328f6ed8d767eef42d1a6e3d827437361e008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-01-20 06:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66a233ab9a83ff8b7a1a949b409b7993813a69d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 07:10:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a822fb26952e454cb7c02f5b5c8bfded : SUCCESS in 10m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfb8c8c158d64aa29375648dd505784e : FAILURE in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8fa530bc826e05d67f49407f36fb9a11beb20061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 08:14:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/70c84198450f4be9b3d709d4937c8943 : SUCCESS in 40m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ee7328fa33144855ab958668c29b2580 : SUCCESS in 37m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a2c75880316b42129060f665abea7e4e : SUCCESS in 1h 09m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/afd904f9a30a47279e066eb1cdfd9055 : SUCCESS in 37m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/600a25a58f4d4c8ba757d4b70e5950bc : SUCCESS in 42m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5caad2e330164abab7a65003f4010bd3 : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad72a9c6a292475db7609fa20440159a : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae919f8f0a6e495799a87609c370324f : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/812bc5e82d244c079fc212b308db8120 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97870e22ea784baf86484a3450ea9e59 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d09222de9274609a5134740dcbcee05 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac112345eeea4acbab0af5b5e127f5dd : SUCCESS in 7m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/56328c0b021f4b34b7a5969f1b310243 : SUCCESS in 34m 05s","accounts_in_message":[],"_revision_number":1},{"id":"541cd831bd33ef9f37b958b501b246e24e5264ac","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-01-20 08:44:06.000000000","message":"Patch Set 1: Code-Review-1\n\nWhy do we expect unit tests to behave differently on arm64?\nExcept for failing to install some wheels, which could be tested by some global requirements job possibly?\nI for sure don\u0027t want to add another non-voting job that is always failing.","accounts_in_message":[],"_revision_number":1},{"id":"9640649e55f626b8ee4e6a3d9f6eea756ce4188b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:32:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e5ad4e283630e6d5b4e2d90f5e70182e5a59da8","revisions":{"5e5ad4e283630e6d5b4e2d90f5e70182e5a59da8":{"kind":"REWORK","_number":1,"created":"2021-01-20 06:58:00.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/62/771562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/771562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/771562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/771562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/771562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/771562/1"}}},"commit":{"parents":[{"commit":"e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-01-20 06:57:30.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2021-01-20 06:57:57.000000000","tz":480},"subject":"add openstack-python3-wallaby-jobs-arm64 job","message":"add openstack-python3-wallaby-jobs-arm64 job\n\nThis is a non-voting job to validate py3 unittests on ARM64\n\nStory: 2007938\nChange-Id: I815aabbcfaaaf4fce7e29ef7735fd4725d1f9aad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e5ad4e283630e6d5b4e2d90f5e70182e5a59da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e5ad4e283630e6d5b4e2d90f5e70182e5a59da8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
