)]}'
{"id":"openstack%2Fdesignate~774389","triplet_id":"openstack%2Fdesignate~master~I76e3f7e3c6878af32e1ca641670b17df838ee0f0","project":"openstack/designate","branch":"master","topic":"bug/1914934","hashtags":[],"change_id":"I76e3f7e3c6878af32e1ca641670b17df838ee0f0","subject":"Allow update status with all_tenants\u003dTrue context","status":"ABANDONED","created":"2021-02-08 04:24:49.000000000","updated":"2022-09-24 05:53:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"06ff470705696a329f799014f45fae8fe0df364d","_number":774389,"virtual_id_number":774389,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-08 05:15:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2021-02-08 10:06:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 04:25:49.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2021-02-08 05:15:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 10:06:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c0aaba4bb2cd5930a4ae7acd94eac6fbceb398a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-02-08 04:24:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1708494e3e0a87f8c4c5067674ae16456c81f36c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 05:15:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cf00b9395cd747188fafb227ca673b4e : SUCCESS in 42m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/182240eabe60494aabceea9cc5836e3e : SUCCESS in 32m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ecdd8feafb6340719e905f08a79f6005 : SUCCESS in 49m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3050b124c28f401f8eb9c2d9b347db28 : SUCCESS in 46m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/375e88fcdd6a4db2872191b8ee5d7b89 : SUCCESS in 29m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6605818cbfa24dd2bac2bbf25723c7bf : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b6e77b5de4e4a06af4a59a3ac2a7c3a : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e88324c1c76c4401a319443d0eadc157 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afd98e5c6d30424d9c04d8389461651d : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a348dd37430649b5b595ffcead601c84 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81871af7c4a34ec89792951328af0031 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6c0a033cca439e962aec31a732a375 : SUCCESS in 6m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44d67852f10040e5b4ed4c4ae06418db : SUCCESS in 42m 42s","accounts_in_message":[],"_revision_number":1},{"id":"786ea275008bb7c0e6662d2218549538b06d170a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-02-08 10:06:48.000000000","message":"Patch Set 1: Code-Review-1\n\nI think that this fix is much too broad and may cause unwanted security issues.\n\nFrom the commit message I would expect to see a patch that only affects updates after a transfer, not all updates.\n\nI also think that we should have a test (likely a functional one in the tempest plugin) that exercises this scenario in order to verify that the issue is actually fixed and doesn\u0027t regress.","accounts_in_message":[],"_revision_number":1},{"id":"c4e9ea6eafbfaa2fee898031889c0f63a205d6c8","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-02-08 14:27:35.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e I think that this fix is much too broad and may cause unwanted security issues.\n\u003e \n\u003e From the commit message I would expect to see a patch that only affects updates after a transfer, not all updates.\n\u003e \n\u003e I also think that we should have a test (likely a functional one in the tempest plugin) that exercises this scenario in order to verify that the issue is actually fixed and doesn\u0027t regress.\n\nAre you suggesting that I should track in update_status whether a zone transfer request is created and only then search with all_tenants else do regular search(without all_tenants) ? \n\nAlso, I think if I add policy_check on update_status, that would make it secure too.\n\nFor test, Are you suggesting something similar to https://github.com/openstack/designate/blob/master/designate/tests/test_central/test_service.py#L3375\n in functional ?","accounts_in_message":[],"_revision_number":1},{"id":"06ff470705696a329f799014f45fae8fe0df364d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:53:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84ab48b4fa4548e070b9a588e4fc216b2e77a653","revisions":{"84ab48b4fa4548e070b9a588e4fc216b2e77a653":{"kind":"REWORK","_number":1,"created":"2021-02-08 04:24:49.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/89/774389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/89/774389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/774389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/774389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/774389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/89/774389/1"}}},"commit":{"parents":[{"commit":"e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-02-08 04:07:24.000000000","tz":60},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-02-08 04:07:24.000000000","tz":60},"subject":"Allow update status with all_tenants\u003dTrue context","message":"Allow update status with all_tenants\u003dTrue context\n\nWhen zone create and then transfer request create and then transfer\nrequest accept happens in sequence all within less than 5 seconds,\nand also update_status is triggered with zone previous owning user\nand project - the query of find_zones return zone not found. Thus when\ntransfer request is accepted by target user and project, zone remains in\nPENDING state for at least around 5+ minutes, until worker threads\ndetect zone in PENDING state and move it to ACTIVE state.\nThe above bug will be fixed by allowing the update_status with\nall_tenants, i.e. find_zones will return the zone in query and its\nstatus will become ACTIVE as usual.\n\nCloses-Bug: #1914934\nChange-Id: I76e3f7e3c6878af32e1ca641670b17df838ee0f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/84ab48b4fa4548e070b9a588e4fc216b2e77a653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/84ab48b4fa4548e070b9a588e4fc216b2e77a653"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
