)]}'
{"id":"openstack%2Fdesignate~775663","triplet_id":"openstack%2Fdesignate~stable%2Ftrain~I2b32cec8d6570e990d19046dbbbd764cb4c4e5cc","project":"openstack/designate","branch":"stable/train","topic":"drop_lc","hashtags":[],"change_id":"I2b32cec8d6570e990d19046dbbbd764cb4c4e5cc","subject":"Drop lower-constraints test","status":"MERGED","created":"2021-02-15 14:59:23.000000000","updated":"2021-02-16 21:11:48.000000000","submitted":"2021-02-16 21:06:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775663-drop_lc","meta_rev_id":"88066cbd3f143c68239f606e923fce6c83674ec6","_number":775663,"virtual_id_number":775663,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-16 21:06:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2021-02-16 18:41:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-16 18:34:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2021-02-16 18:43:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 16:10:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 18:34:13.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2021-02-16 18:41:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"de5410473bd308530bc5f7ed9d850b8723afceb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-15 14:59:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb58d1e104a6ede67a274ad267e93b785c5a2c4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 16:10:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dfa919bb8c644dc68f7ae921b7ad1ff9 : SUCCESS in 36m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dada601f6f6d4a1c8d07eb884891994b : SUCCESS in 36m 36s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/e0f8a641df9346c9accf5806af6b3008 : SUCCESS in 45m 37s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/f90f29318b074721baf26bd1223dd660 : SUCCESS in 37m 54s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/0827aed043074c09a09c4612a5b3abb5 : FAILURE in 28m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9c45050738df43b78b857f7165ca19f7 : SUCCESS in 53m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/430f60327265450a853b95ae1fbfac64 : SUCCESS in 36m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d2c469346ca64d768a72a3ceefb81e5c : SUCCESS in 35m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58fe1a34444f4d3699d8d8bacb4f40ae : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/306ff5da52bd43f9afcc27105e976e43 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/684d3a2a81e848a9872459a512359962 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3b8304660064bef96632366f901b4f8 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f977db1228434132833492a8a44dc5a8 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e5fab35d0f4e1ea9aa6deb503f3aba : SUCCESS in 7m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cda1b3d4369a48058c7da43b18901bdc : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be1e5159c774758b34f5ac71ecf9cc4 : SUCCESS in 4m 40s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/89e4e28ee60b47ff8d11a83598706938 : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b0e6c81894c7d46751804e58ab626801cd569487","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-02-16 18:34:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27b7d39a6822a08c217bbcfade132eb1d70ef778","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-16 18:41:13.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"25129410c5babd9cf1fc5a763f722c6a3afdd21a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-16 18:43:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7c4eb5a94f71a925f6c56bfb8b173409f56ce62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 18:44:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92d83f08df761621287e5449ddadc29d99d9c66a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 21:06:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1efb0f25d74a4f4ba6348d8cb2d0e839 : SUCCESS in 39m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f86b8b5ef845414ea138a4ad305d3b1c : SUCCESS in 39m 21s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/4a20bae0361848968223a7bddc5518a2 : SUCCESS in 40m 42s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/b6591719d2be4180aa505ec5dade4538 : SUCCESS in 44m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/40e8b129f53f43fb82c767c3294fe2a5 : SUCCESS in 54m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7a0f0820a11841cca87cc8223255c1c1 : SUCCESS in 32m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9f2cac1dfa004ca58d702285c99b91ae : SUCCESS in 33m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fcbcc548f31495e95351b7bd882336b : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57ac5b81b6be4224a01aa422efabd410 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35fa0b3337ad472895b4a2d704d711e1 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afff9507dae24dc68cea09ffabf5ba11 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7834e6406d4bbaa91b27f1544b685f : SUCCESS in 9m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/74f7614b6ccf4df7aba0057c737d08b2 : SUCCESS in 2m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50a941ba64a845739e8cdbbf35373e88 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1248950facb1c09bd872df8a9d009f0775e3ac8d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 21:06:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"88066cbd3f143c68239f606e923fce6c83674ec6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 21:11:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace1075c9a3e4ca9ba9ca148675c9961 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c418f14fc4146f89f86a826adaae499 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbf90fe5df83b48aeaa28946c93e606933c99d8b","revisions":{"dbf90fe5df83b48aeaa28946c93e606933c99d8b":{"kind":"REWORK","_number":1,"created":"2021-02-15 14:59:23.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/63/775663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/775663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/775663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/775663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/775663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/775663/1"}}},"commit":{"parents":[{"commit":"f9035b251e7de50d398d1a240923862237164f1b","subject":"Merge \"fix database sync always get the wrong  value\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9035b251e7de50d398d1a240923862237164f1b"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-15 14:58:48.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-15 14:59:07.000000000","tz":-420},"subject":"Drop lower-constraints test","message":"Drop lower-constraints test\n\nThis change drops the lower-constraints test.\n\nChange-Id: I2b32cec8d6570e990d19046dbbbd764cb4c4e5cc\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit d08dd621c988fe13b9844c5c44e143907aa74d0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dbf90fe5df83b48aeaa28946c93e606933c99d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dbf90fe5df83b48aeaa28946c93e606933c99d8b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
