)]}'
{"id":"openstack%2Fdesignate~775664","triplet_id":"openstack%2Fdesignate~stable%2Fstein~I2b32cec8d6570e990d19046dbbbd764cb4c4e5cc","project":"openstack/designate","branch":"stable/stein","topic":"drop_lc","hashtags":[],"change_id":"I2b32cec8d6570e990d19046dbbbd764cb4c4e5cc","subject":"Drop lower-constraints test","status":"ABANDONED","created":"2021-02-15 15:02:18.000000000","updated":"2021-02-24 16:21:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ba8622122ace0ad2f5a425da4afdfdd12bd92a48","_number":775664,"virtual_id_number":775664,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-15 16:25:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-16 18:41:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2021-02-16 18:34:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2021-02-16 18:44:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 16:25:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 18:34:18.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2021-02-16 18:41:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"6e2dbcc1f56d8c4ddeefe0c6d23e11bafcebdf0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-15 15:02:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f1876518c223fc5b050ffe56570d49be4f45174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 16:25:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/27bdd8b1963b406094aa6910793923a0 : SUCCESS in 38m 36s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/bdc1173b1d7f4fae973ff9f30e646f72 : SUCCESS in 37m 54s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/a307f07c43814d27b6cbe901fa8ba869 : SUCCESS in 41m 11s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/0955d07cad4944878a9a57047de55ae2 : SUCCESS in 37m 42s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/e376318f2fdf427e9476e2fc44e9bed0 : SUCCESS in 32m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3f1c9300d242477e9937f4f8a859fb6c : SUCCESS in 52m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d06e96c20df4ec89a68643bea99943f : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46837ae872a44eff9f01c586b60f8b68 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9927116353f4478ad796462935364c1 : SUCCESS in 6m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7f55d292a9034fc18a89a8bf74127626 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0757c7bb5bfb4fcd8abf0555a16842ea : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31979e796b324087b63e4ff397ca4cdd : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef6b6c8dab041a29eb09d7a147661d7 : SUCCESS in 5m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0f940f779de74e8b9f4d107a97aa97ed : SUCCESS in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2cb6d9b22d648d591ec2c3f3dd014d7 : SUCCESS in 6m 38s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/916ce0f84f8c4952bd9f9d81a23b2b4c : FAILURE in 30m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c8f162e34251ab41a7a97cfc9d161f9cb8fb1dd2","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-16 18:09:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b539bc8e62e66a9e6902fe443072670038596b69","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-02-16 18:34:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a33637057cafc34962fc3660fd24af73fd01109b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-16 18:41:31.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"02fa707880321ba038bc6e31699fe5af5a7d87a0","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-16 18:44:22.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83ac565225b81eb9d9b89f30685fd2413198b8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 18:55:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c7e85ef55b3a4278aacd4f3faf9c3ce9 : SUCCESS in 32m 01s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/dc204bbc7b064f038e05a93e2f691859 : SUCCESS in 30m 36s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/a95d446bbab941d88d6e66060a9dd366 : SUCCESS in 32m 13s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/e2ab19880c8546cf8336cd88fa7f7bf8 : FAILURE in 26m 02s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/27e2bb6ae59641ae98dd161f2a170ff8 : FAILURE in 29m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6ac431f8696547ada6f6ebc6f1e8930d : FAILURE in 44m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8813d2c4cd5648fea2e0dcdd4f3f77f9 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b87e215ed34e6a8b163ceb57e284ff : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a068ce9198345d6928f225db558952e : SUCCESS in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/12456394f599423398b681994ceaa129 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6044983d361d49ef9d50ee48ea5c6115 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a217d7351db4c7cb4e38af741e009f2 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/187ed8d3ea594d60835de50f1ca677d9 : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/567dc45433a3469ca700537f09a3c7ee : SUCCESS in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a8fb42720bb46d8b2b57b7a3ce67467 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/da05d7a488cd4f098f9e1159a22277da : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":1},{"id":"128e434872a793f179e5f73d513a7065d86fdb4c","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-16 19:42:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fef006da5245019f4e80fcc8312fc90200163aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 20:36:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/210729a0dd484c9b823b85ca5a685949 : SUCCESS in 36m 50s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/890eb8456972498e856cfe65091d0970 : SUCCESS in 37m 09s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/0ed4ffaadf854d1d8e00636c73733abb : SUCCESS in 40m 24s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/faad6ed029814cf386c50d1338f14ded : FAILURE in 29m 31s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/ef5bfa780939469faa16efaec223b14c : FAILURE in 28m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cf9613cf14744ddc9b01142951f0ea3f : FAILURE in 46m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf8986985d154789a272953773aeb0bb : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a0ced3f12fa453b9c541fb60736b7cf : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ed2fcfa6a114d21b9a10a511fad775d : SUCCESS in 5m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88f05e166f6646a8a06d3df5afb8d88d : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aca96d675e7d463eb24be8bb3ef260fb : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1ece2a5b5f64a16b617db7277220ab7 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b16839d3ed43e69fc1de7e7e7d04e4 : SUCCESS in 5m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ff450ddbdd4435c9ec91acc6b4a9119 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bf16b7ca27342b68cdb3873095a0788 : SUCCESS in 4m 30s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/fc2cd21583f248f3990ae3886c29129d : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a4fb142760c65dfb9f27d7e61770049679ab1a43","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-16 21:15:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4ee329acfd87b1335fafb563c83297ddf454f5f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 22:15:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2d0371eada364f749daf0055139340cf : SUCCESS in 39m 08s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/272c7577ac7e4020b244f6c1099fc571 : SUCCESS in 38m 38s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/3fcfe47156744c4a9a6d60a13ffeb6d5 : SUCCESS in 32m 01s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/dee7b112ffea4d17ae4b28fa840914f5 : FAILURE in 25m 51s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/d47662ce5e4c41699503ecb25a1a469b : FAILURE in 33m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/68a37c0995de44fab4daeeb76e1377f8 : FAILURE in 53m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0579d5a17988473ea668272934efa1bc : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/118a82dfed9044d8ac65c249bbf04c5d : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c73e0c0d4774feba6cef033d1a266ca : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9ea6b0add71a43c4a4b19a2592f6c937 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ad2898a755a451f9db9ef7828a6feeb : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac6c4213f3164971a43c95dbb8a667d7 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4e6dbba3a7474c8d6fa73a8e7221e5 : SUCCESS in 4m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/783523ab328745ce9d1b84cbd14c2b2d : SUCCESS in 2m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79b40b41c6d1449999fde7e193558d15 : SUCCESS in 4m 37s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/07983273778544d58f394dd34be4f7ae : SUCCESS in 29m 07s","accounts_in_message":[],"_revision_number":1},{"id":"451864868866523cc666f18be144c2199a7056cd","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-19 22:30:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"69a9d3509d3487132dbdf2691dfc358fdbba82c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 23:22:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8f231218c9a740d8a2c2c99e30944d69 : SUCCESS in 40m 44s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/9d89e89fdb944989955e72d61703a3c6 : SUCCESS in 42m 57s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/2c95a9e0010049da80f3ca57edf56238 : SUCCESS in 41m 52s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/bae42db334ca4b69ab3f4b1c440bcb6a : FAILURE in 23m 27s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/e3d34f16e82a4cb9a4fc02ca897d0957 : FAILURE in 23m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/65072788cf7e475696970a28219d101b : FAILURE in 48m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f76e632f96a4bd3b5b6c0462f9f01d3 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0829433fc5ee41189a02d90c600894f0 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61d81b81e98341e494bdc59c0c3c54ee : SUCCESS in 5m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2ef7e5d9412a4321a3467c2e72d9c2b5 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99befc9f03734341a6aba1a8d96a24c9 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bffbc489b8564101a4df892690c5135e : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa33383ec2848e8bd8cec095cc1aa80 : SUCCESS in 4m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a032494d18143bbb9da54ee8f497fce : SUCCESS in 2m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b7097546fd24c1db31c953e4e49f812 : SUCCESS in 4m 16s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/9ef329521a704ce0aaa370f545b2bb3f : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ba8622122ace0ad2f5a425da4afdfdd12bd92a48","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-24 16:21:16.000000000","message":"Abandoned\n\nMerged as part of I90af3a092296f943509833608f25522b6f8e9ab6","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfb3c74e38f71a695413750f6aac511caa02baae","revisions":{"cfb3c74e38f71a695413750f6aac511caa02baae":{"kind":"REWORK","_number":1,"created":"2021-02-15 15:02:18.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/64/775664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/775664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/775664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/775664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/775664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/775664/1"}}},"commit":{"parents":[{"commit":"28f74c81e76ce4e1f84920f52a3ae5530f0097c2","subject":"Merge \"Update vagrant config\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28f74c81e76ce4e1f84920f52a3ae5530f0097c2"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-15 15:01:38.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-15 15:02:01.000000000","tz":-420},"subject":"Drop lower-constraints test","message":"Drop lower-constraints test\n\nThis change drops the lower-constraints test.\n\nChange-Id: I2b32cec8d6570e990d19046dbbbd764cb4c4e5cc\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit d08dd621c988fe13b9844c5c44e143907aa74d0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfb3c74e38f71a695413750f6aac511caa02baae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfb3c74e38f71a695413750f6aac511caa02baae"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
