)]}'
{"id":"openstack%2Fdesignate~775903","triplet_id":"openstack%2Fdesignate~stable%2Fstein~I90af3a092296f943509833608f25522b6f8e9ab6","project":"openstack/designate","branch":"stable/stein","topic":"fix-pdns-stable/stein","hashtags":[],"change_id":"I90af3a092296f943509833608f25522b6f8e9ab6","subject":"Fix pdns4 devstack plugin and drop lower-constraints","status":"MERGED","created":"2021-02-16 23:07:40.000000000","updated":"2021-02-24 16:20:50.000000000","submitted":"2021-02-24 16:13:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775903-fix-pdns-stable/stein","cherry_pick_of_change":754604,"cherry_pick_of_patch_set":2,"meta_rev_id":"afec04ec8500c25de6d1c0b32443c046fd0b78eb","_number":775903,"virtual_id_number":775903,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 16:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2021-02-22 16:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-22 20:46:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2021-02-24 08:24:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-02-24 16:20:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 23:07:40.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-02-16 23:59:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-18 17:10:59.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2021-02-20 20:18:24.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2021-02-22 20:46:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"7b2cfc1497124864cde97952493c7364be3305c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-16 23:07:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"66c14ccaf6e88ff4388cc05d94e1ad46812dee11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 23:59:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8f3a5dfa2a1458196058e000e364ef4 : SUCCESS in 32m 33s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/033425be41aa4582a5aace3af8d0a3bd : SUCCESS in 33m 56s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/3826e0242775454581530001408e8520 : SUCCESS in 33m 32s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/ae00d6f18d494f1e9cdc70bb55894431 : FAILURE in 21m 51s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/a55c5e6c4f86463a866993b92873afb8 : FAILURE in 24m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0d061352c74f44118c4c67c36a965a34 : FAILURE in 45m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43f231f2e9b3490c8c2e049245940049 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19d67027eaf540fe856b111d0d4e0be8 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bb7652253d740e19a3bf8091e9b85b8 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9724c8a9670647c0bf97027ce7db7317 : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c73804e4b982455d89b62973120487a8 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c968db93f7e34ca7ab18afe8b2557cb1 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc41c1ddb76e4cfdaa18a14ed3335a37 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5721966a90b4247b0be4f49f3c2e5ad : SUCCESS in 4m 34s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/12ba4cd8e2df4350a78b008daa4c226c : SUCCESS in 25m 10s","accounts_in_message":[],"_revision_number":1},{"id":"72a0a27ec283749a211897f9b475d9d63b16d030","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-17 00:00:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1ef3cf063ebe6a65c4b5c4cb29b62edb5b25f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 00:52:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/34710c0fbfa546738a841c705409a358 : SUCCESS in 31m 40s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/da090aa321a84bb0a4ac71999f411173 : SUCCESS in 36m 25s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/af656f57159a447fa1f67fbda09c1949 : SUCCESS in 38m 13s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/22935ac2e0b24cc28657dfe095c982ad : FAILURE in 29m 30s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/0f8243db704f49b8bc3c07283ba8c7af : FAILURE in 34m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/40630b3e384f496da1298829a71f493b : FAILURE in 45m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c7a8a4c7ee34162ba5421112e017df7 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92087d17c14145c685d3d2dfba11d712 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dadc80a2c5648bebc2c20456db95ced : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fe04601adef4fec857b85ef2b97103b : SUCCESS in 4m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ebd55819ab5f4ab3abf452c395346415 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72e19d71f08e48649237fa17c9e26860 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f90cce8c42f496cad3e47241bc42685 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b66743c0e249d2a754f4eca73b5d54 : SUCCESS in 4m 39s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/dc5762ce334246e89d9003174e982099 : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0aca294bdd77745193b6a7fa39851322b4319628","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-17 01:15:52.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"abfadb1545b86ba787f63ea24fbdab33d4a2694d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 02:05:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/286b06c3f3e34ba297ff4e803aa41204 : SUCCESS in 47m 25s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/2f73655dffa14c3fbca8ed06f7255a21 : SUCCESS in 31m 15s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/98ca516dee874869bacc2983aae17739 : SUCCESS in 29m 52s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/ccf6292714cb4b26afd41b04e16e6528 : FAILURE in 24m 53s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/28329eca83de4cd0ae5e3961f0fea5ec : FAILURE in 22m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/35311797777449038a9904fa1dc62c33 : FAILURE in 37m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/912ebae399c84b30b4b58d12a4540760 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/367c4acadbbe415691485e078edbefce : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab7242ad91fe491eb6e41511276808f3 : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/290da256827f4c7db5b0707932f81f92 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63f6b2f686414157a044eedb07cf0a19 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cba339cad6e843aa9186bea706982210 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9030bb9f957412cbd2f609966da49b1 : SUCCESS in 4m 22s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/8e9fb489559d42c18bc777a50b0e1707 : SUCCESS in 23m 48s","accounts_in_message":[],"_revision_number":2},{"id":"70b5cad8ae080adfa8dadcc138dde3b55fea2bd1","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-17 13:42:52.000000000","message":"Patch Set 2:\n\nThe failed tests are all failing with\n\nThe specified regex doesn\u0027t match with anythingERROR","accounts_in_message":[],"_revision_number":2},{"id":"f3f9bd6748e37440eb99b192c78ae02821692c9e","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-18 04:48:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9e9be481545f6ee2fb5de766d64a6d1e1d54179d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 05:50:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7a1040ace69b4245a11d8589d4a256f8 : SUCCESS in 41m 16s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/0b2e5642b3bf4960adb00725a93bccca : SUCCESS in 40m 14s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/7da9fcc12fde44299f1089984b363774 : SUCCESS in 27m 27s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/f77e4308a8c846b38200ac55f1b6d5f5 : FAILURE in 20m 34s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/75a630f09b9f4ff1b982ec9d072c40b1 : FAILURE in 21m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cbbcd159c516477d9835de19280f3e92 : FAILURE in 34m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9ec14e2b93f479889e03983d1783571 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c483dcfca684e2e9014cb67053570b8 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/020d21aea57a4d03ba9f2a80517c66cd : SUCCESS in 5m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/73f94edbb21845aa8c184bcd02111967 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4c13f0e9a834266a9b0aef1117cca49 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9151a87d9cf4998b57928159556a5fb : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f359fea8336b4e679615178c0f7d4ea1 : SUCCESS in 4m 33s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/b3c74d39316048cd8a52ce3e84ba7b53 : SUCCESS in 33m 00s","accounts_in_message":[],"_revision_number":2},{"id":"141ff0c06c84f964e091142f1b0c9621159e2cd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-18 15:54:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50e3e8ee7a7b87e897a4a42a0aa148966d9bfd52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-18 16:31:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2c1c9316f65c7cb4e52d362ce73c9c05dd92e887","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2021-02-18 17:10:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8220cca78c6e6f411164f23b393ec1c18d550070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:45:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4f377f8aa4ad45e598c8daaa71118fe3 : SUCCESS in 36m 02s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/5815cf4d27174ee28622a84a6e026735 : SUCCESS in 43m 37s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/01cf50003ee9404b8f80a5bcfc53fb8f : SUCCESS in 39m 13s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/a6a5c2d3a1cd477e9c0e49c9906811c4 : FAILURE in 25m 12s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/787ce53e33d0452a9f6e10896abf4c2e : FAILURE in 31m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f2f3f0132712405ab1aaae16331e0f1a : FAILURE in 55m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd6abf057cf648249b321e243b9ae67f : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d745a3c2944e2ba8da6e1f5b1faaed : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25c5b0f839394a0da7ea117b16b1cee6 : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/273d14d50ae141c184c0b7078ee81370 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90c7a0a4141a4b16bb756ff5669f4595 : SUCCESS in 7m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fba828726fa4918b39a955af7a6e357 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea6e7997a4f47de9ac3ca6fcc59ab8a : SUCCESS in 5m 24s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/c89985d175124790b5b6263083adbe6e : SUCCESS in 31m 34s","accounts_in_message":[],"_revision_number":4},{"id":"1c53f6f2c651c4bfde8bfd5f3c76706ca89c7c8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-19 22:49:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"550d7da2db2d2f14ff247c1cead88cf8ac90e752","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-19 23:15:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"026c04c36d0311951d7c80611ee2657a2aa6560b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-19 23:54:35.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"a524313f6b7dedc0a5e8a12761b16f7631633b7b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-20 20:18:24.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"22333164aa2b7c744aaf576597883dac88de01fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 21:08:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/083e7ef5a39c4497adac8a1dfcf51948 : SUCCESS in 40m 57s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/73d95d4af8774be99014c2a27b0a5950 : SUCCESS in 40m 11s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/16c708b64f8b49819529aec392b8551f : SUCCESS in 39m 05s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/35fd0a4c76e0482e88ed32ef0840291d : FAILURE in 30m 20s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/9038beeaa80944bf8cf4b2a15f7bc985 : FAILURE in 31m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2b27eb1d5787411dbd76e40fafac4655 : FAILURE in 45m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e10cae2983184c419ca37df5ba7ed293 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87309fab49dc4bf08c7c6555b5d2cf6c : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21b46e7f6a814668ac2abed50e19f494 : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5be80e1901ce433c83e3d1a6a7ff1708 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33e758e8f8cc46e7b27010bd899f72cd : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ab068bb07504acaa1397e2022681a38 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbcaa8d6875144d5aba7ee654871c2b2 : SUCCESS in 5m 15s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/b62cbaee0c604fe984a395146343b3e4 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":7},{"id":"8d4bdd5514f297d76bf1fc35d9ceff13c752582e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-20 23:08:36.000000000","message":"Patch Set 7:\n\nwith devstack-gate fix, it use the same constraint as it was during designate-tempest-plugin installation in tempest venv but somehow tempest venv gets recreated and remove the designate-tempest-plugin installation - https://zuul.opendev.org/t/openstack/build/35fd0a4c76e0482e88ed32ef0840291d/log/job-output.txt#2424","accounts_in_message":[],"_revision_number":7},{"id":"8715e847eab23243a271c5c77fa5f2f6e3098fa6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-20 23:17:26.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\u003e \n\u003e with devstack-gate fix, it use the same constraint as it was during designate-tempest-plugin installation in tempest venv but somehow tempest venv gets recreated and remove the designate-tempest-plugin installation - https://zuul.opendev.org/t/openstack/build/35fd0a4c76e0482e88ed32ef0840291d/log/job-output.txt#2424\n\nI think I got it. devstack installation use hard coded constraints from tempest tox.ini file which is different than what d-g set and it end up recreation of venv\n- https://zuul.opendev.org/t/openstack/build/35fd0a4c76e0482e88ed32ef0840291d/log/logs/devstacklog.txt#8289","accounts_in_message":[],"_revision_number":7},{"id":"441c7539bfaff928648085a5111bab3866477885","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-20 23:25:20.000000000","message":"Patch Set 7:\n\nrecheck deps updated","accounts_in_message":[],"_revision_number":7},{"id":"0ecec5fbe7925c2f40e30399c8600cdc06a0b63c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 00:12:57.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"feefc24ca59d7afb1f3bacc69fb8b10bc31559ee","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-21 00:13:35.000000000","message":"Patch Set 7:\n\nrecheck deps updated with https://review.opendev.org/c/openstack/tempest/+/774060","accounts_in_message":[],"_revision_number":7},{"id":"a25993bfa23f7217510cbccc6da19b2dad3812af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 01:12:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dedfa696930e4360b2db6434ed557bf9 : SUCCESS in 39m 28s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/492a88b9f1f94b1098738d9ee3729b09 : SUCCESS in 42m 05s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/70050078d0c34020b429db07e30db4da : SUCCESS in 39m 02s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/096c72c6c6c5448da77501c3116c2223 : SUCCESS in 35m 56s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/b40ce97e01db4152aef93ad41835a8ae : SUCCESS in 35m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9285e8b712d45169bf2ec669e0ec5ed : SUCCESS in 54m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74b350138e07468f816d592a79693e8b : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116194394ce14973b80fe81770c6aed8 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4553932d0134d448d94d466dee7e073 : SUCCESS in 5m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b92405e7bdbf4f6c93d805a58272cf31 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b38a6e475864c2d9965821c4c9ab84b : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95dd9b984bc94d81b21306edf3f4e141 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f552e24067420889b83f3a33feea08 : SUCCESS in 4m 50s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/305e7053926c48e1b58466bab563e31a : SUCCESS in 32m 18s","accounts_in_message":[],"_revision_number":7},{"id":"b7e9dac5acb879c9f26123793cfd50d7f34f4dbe","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-22 02:19:45.000000000","message":"Patch Set 7: Review-Priority+2","accounts_in_message":[],"_revision_number":7},{"id":"867889eede709c370062e6241a617241de83f3ed","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-02-22 07:42:19.000000000","message":"Patch Set 7:\n\nSeems this needs to get merged with https://review.opendev.org/c/openstack/designate/+/775664/1 in order to fix all tests in one patch.","accounts_in_message":[],"_revision_number":7},{"id":"be73aa62ed4c6fab56c386cce59b103b72dd96a0","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-22 13:55:48.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\u003e \n\u003e Seems this needs to get merged with https://review.opendev.org/c/openstack/designate/+/775664/1 in order to fix all tests in one patch.\n\nDo I need to squash the two changes into one?","accounts_in_message":[],"_revision_number":7},{"id":"57abd89821e812266c55ec883f826b104210da3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-22 14:21:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"55d3f5e2e9f5339bd5accf87343452939c5c9b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 15:30:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d480aceacbef40ec94335edf9afdd6b7 : SUCCESS in 45m 18s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/38e37bf872ef497295379520ccee0788 : SUCCESS in 39m 27s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/ab989b4183f24caca7397dc43abc1b25 : SUCCESS in 32m 19s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/421a09549a8146c8a1f29d68cb670584 : SUCCESS in 34m 17s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/9c405ce87bf3489db0c53abe5bac4c87 : SUCCESS in 27m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8ff826361ffd45ae97736da1085b7fb3 : SUCCESS in 49m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/289a7651950b436c81ce39cf9e33162c : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b24f4e62cc2846a8bb2c61c3d794c16b : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63ba30e8c05f4b6db91c0f38a100b411 : SUCCESS in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cee59f4ca870415b9158344f7987364f : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a70ddca3d8e40319783e2ebb9df6087 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/635aafe07da24a57917b93ce5cf5eeb3 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4156103f7494b4b99f06115104dad9a : SUCCESS in 5m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a4dd60ba05ea45afb697d474be3f3a64 : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/160525bf430b44088dcedb6d3831198f : SUCCESS in 4m 32s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/8996ee9ee9d9422386c4e87efee1b387 : SUCCESS in 29m 15s","accounts_in_message":[],"_revision_number":8},{"id":"8c95b9cac50b2e47645e7c256259785ed370aebc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-02-22 16:07:12.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"e178956c20a576f65c5569da36918afe8a3cd590","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-22 20:46:16.000000000","message":"Patch Set 8: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":8},{"id":"82a4ddc426ed6353a14a190515ec5e7f301a9b78","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-24 00:48:26.000000000","message":"Patch Set 8:\n\ndepends-on is merged now.","accounts_in_message":[],"_revision_number":8},{"id":"45955403865de3cbec709c09984dc3354e42467c","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-24 03:08:31.000000000","message":"Patch Set 8:\n\nThanks Ghanshyam! Is it customary to update the commit message and remove the Depends-On?","accounts_in_message":[],"_revision_number":8},{"id":"3abe10219d93524c30e3f769554ad0e7ffb45e3d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-02-24 08:24:27.000000000","message":"Patch Set 8: Workflow+1\n\nNo, it isn\u0027t needed to update the patch. When the depending project doesn\u0027t share the same queue with the current one, the merge wouldn\u0027t automatically be triggered when the dependency is merged, so one would usually make sure to approve a patch like this only after that. Though I\u0027m actually not sure whether designate shares the integrated queue or not.\n\nAnyway, single approving this now in order to unblock things.","accounts_in_message":[],"_revision_number":8},{"id":"86bb2ea6af9bcec3378e4cd6cbe89491525e6f5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 08:24:43.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"35d8220527514cb733f38f69817a6d4d1e9469fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 09:08:59.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5511678bf84946f09d6205e14100e2d9 : SUCCESS in 42m 14s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/6ca54d9ba8f448858e05a531a7900509 : SUCCESS in 40m 13s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/315c59f50af249b6a2e8b067a8b33cbb : SUCCESS in 35m 45s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/06389b6c422e4d78a5ee8f7a653bf101 : SUCCESS in 36m 54s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/d130bc688d4d42aab3acd49b8fe6777f : SUCCESS in 27m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9969c8a7ca2a4727887c5a8681e1f8e6 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f26f1adcfba4e23bc61216a1592a8f0 : SUCCESS in 5m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a478820ee91d475f878b424cf082fe8b : FAILURE in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/377c8438cf4f4a28b57310b1bcb2124e : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f85ceff9a4346588ed5f28ee3ea71ce : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb8c592339e49a398b0a131ffce47e1 : SUCCESS in 4m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9bf89be69523437b885a9b2238f1ef26 : SUCCESS in 2m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f9abcee487b45e69bad09d912cf01fd : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":8},{"id":"650971b7adfd4aa5e002a189f73e08b3c0339aa2","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-24 14:17:44.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"166e5d97b73d4ea50f6facb6da619e6d6fd99099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 15:23:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5296d8f9f15b4179aec4580c6dda6ba8 : SUCCESS in 43m 46s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/736f78ffb40c4782ab620ca0f3a16417 : SUCCESS in 43m 09s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/932f3bf086b84b8faf31c57bea313749 : SUCCESS in 44m 11s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/55d45b241e7b4362a50fa17188abfdb0 : SUCCESS in 33m 37s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/c3ae7b66c8ee4205bda6e2f54361503e : SUCCESS in 45m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/00c89c47e0f14246be174f0ef15cc43a : SUCCESS in 56m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cd0d043a94a482c9b3afa3de50d8b3a : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfae2e7368b84701b2cb3019d8eeb2ad : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ccafebd1f2c415692fbb0f9a9736ad0 : SUCCESS in 6m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5da694cee2434899b2f8b6639a5dacef : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2a83eebbc7a4defb45cd73d007527fd : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da2b37b1404347b48f84fffc0254a120 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0749dbb8e9c4abe9a9186c9bb1e0cb0 : SUCCESS in 5m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7d2184ad2a1748b7aa39cfa5691e6fea : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/369b2cbc37c64c999aa93a41c02546e8 : SUCCESS in 5m 08s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/77dd77ef229649cab98da3e46c3ac026 : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":8},{"id":"c592964b5a12dd9e0cd88f86209858e912faf9bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 15:23:38.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6134c98516ddb3a3f4daa686b44096b98f798efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 16:13:05.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3739c4f48d944ca995418f18f9d615e0 : SUCCESS in 44m 19s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/5317524e13454dee829a23d1231aa23e : SUCCESS in 41m 49s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/b05bbb1f330841dbafca675a50f87432 : SUCCESS in 41m 00s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/fa677954db894d08bf11bed8e53585ef : SUCCESS in 37m 20s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/6171bda57cd34704b03d7332608c863b : SUCCESS in 32m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37849907503646758cd7ecf0d84abab1 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19813ba9f9924a60860600ec2d492152 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/53bc4c8aa2794553865f60298de40ff9 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c3c1f26c6c84b15b01e38148ece2ed4 : SUCCESS in 6m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/196b24876e3340c18cd9a4db7342979b : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ee0f0e701a4b53bdc69b1e20068f85 : SUCCESS in 6m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e60f40fdde7f4194a21cf446cb46a240 : SUCCESS in 2m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb90876c5f124a69a8bf350b471cd01d : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":8},{"id":"f41204a02f163a869cf46f76576ec285f50e29a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 16:13:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"d6a8df5954e912f93a88f03fe7b4c26632311e9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 16:20:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe35c6f9df54653849c6125a5ae89db : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc84941e58a34f82a2b9e386b5f62ea5 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"656a26c45ce4c9bc7931867c7db0883ff763cf2e","revisions":{"11b733afaac41aa99347bacd5cec00b679f62842":{"kind":"REWORK","_number":1,"created":"2021-02-16 23:07:40.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/03/775903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/775903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/775903/1"}}},"commit":{"parents":[{"commit":"28f74c81e76ce4e1f84920f52a3ae5530f0097c2","subject":"Merge \"Update vagrant config\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28f74c81e76ce4e1f84920f52a3ae5530f0097c2"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-09-27 10:23:37.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-16 23:07:40.000000000","tz":0},"subject":"Fix pdns4 devstack plugin","message":"Fix pdns4 devstack plugin\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11b733afaac41aa99347bacd5cec00b679f62842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11b733afaac41aa99347bacd5cec00b679f62842"}]},"branch":"refs/heads/stable/stein"},"fd616f259802aab05f7168abb8f9ed56b738a910":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-17 01:15:52.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/03/775903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/775903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/775903/2"}}},"commit":{"parents":[{"commit":"cfb3c74e38f71a695413750f6aac511caa02baae","subject":"Drop lower-constraints test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfb3c74e38f71a695413750f6aac511caa02baae"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-09-27 10:23:37.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-17 01:15:11.000000000","tz":-420},"subject":"Fix pdns4 devstack plugin","message":"Fix pdns4 devstack plugin\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd616f259802aab05f7168abb8f9ed56b738a910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd616f259802aab05f7168abb8f9ed56b738a910"}]},"branch":"refs/heads/stable/stein"},"d339823654f269d4a5d9c8dd1b587c39b0f40c43":{"kind":"REWORK","_number":3,"created":"2021-02-18 15:54:46.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/03/775903/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/775903/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/775903/3"}}},"commit":{"parents":[{"commit":"cfb3c74e38f71a695413750f6aac511caa02baae","subject":"Drop lower-constraints test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfb3c74e38f71a695413750f6aac511caa02baae"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-09-27 10:23:37.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-18 15:54:07.000000000","tz":-420},"subject":"Fix pdns4 devstack plugin","message":"Fix pdns4 devstack plugin\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d339823654f269d4a5d9c8dd1b587c39b0f40c43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d339823654f269d4a5d9c8dd1b587c39b0f40c43"}]},"branch":"refs/heads/stable/stein"},"eb0c4413d78bdab45ac2d78f974d06e22d07f711":{"kind":"REWORK","_number":4,"created":"2021-02-18 16:31:53.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/03/775903/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/775903/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/775903/4"}}},"commit":{"parents":[{"commit":"cfb3c74e38f71a695413750f6aac511caa02baae","subject":"Drop lower-constraints test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfb3c74e38f71a695413750f6aac511caa02baae"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-09-27 10:23:37.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-18 16:31:37.000000000","tz":-420},"subject":"Fix pdns4 devstack plugin","message":"Fix pdns4 devstack plugin\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb0c4413d78bdab45ac2d78f974d06e22d07f711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb0c4413d78bdab45ac2d78f974d06e22d07f711"}]},"branch":"refs/heads/stable/stein"},"8ef7912bcaa16e10d68312aff15e89e11b17d4b2":{"kind":"REWORK","_number":5,"created":"2021-02-19 22:49:15.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/03/775903/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/775903/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/775903/5"}}},"commit":{"parents":[{"commit":"cfb3c74e38f71a695413750f6aac511caa02baae","subject":"Drop lower-constraints test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfb3c74e38f71a695413750f6aac511caa02baae"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-09-27 10:23:37.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-19 22:48:54.000000000","tz":-420},"subject":"Fix pdns4 devstack plugin","message":"Fix pdns4 devstack plugin\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8ef7912bcaa16e10d68312aff15e89e11b17d4b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8ef7912bcaa16e10d68312aff15e89e11b17d4b2"}]},"branch":"refs/heads/stable/stein"},"fb679459c8cd379c6ac2e39f46ccb61b4f9da5b6":{"kind":"REWORK","_number":6,"created":"2021-02-19 23:15:03.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/03/775903/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/775903/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/775903/6"}}},"commit":{"parents":[{"commit":"cfb3c74e38f71a695413750f6aac511caa02baae","subject":"Drop lower-constraints test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfb3c74e38f71a695413750f6aac511caa02baae"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-09-27 10:23:37.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-19 23:14:50.000000000","tz":-420},"subject":"Fix pdns4 devstack plugin","message":"Fix pdns4 devstack plugin\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb679459c8cd379c6ac2e39f46ccb61b4f9da5b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb679459c8cd379c6ac2e39f46ccb61b4f9da5b6"}]},"branch":"refs/heads/stable/stein"},"7d2f9dd0e0e1d6ad7eb08f5b915350aca43b284a":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-02-19 23:54:35.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/03/775903/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/775903/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/775903/7"}}},"commit":{"parents":[{"commit":"cfb3c74e38f71a695413750f6aac511caa02baae","subject":"Drop lower-constraints test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cfb3c74e38f71a695413750f6aac511caa02baae"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-09-27 10:23:37.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-19 23:54:13.000000000","tz":-420},"subject":"Fix pdns4 devstack plugin","message":"Fix pdns4 devstack plugin\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\nDepends-On: I36799ee9d4db20104377e2772c86229b8083448d\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d2f9dd0e0e1d6ad7eb08f5b915350aca43b284a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d2f9dd0e0e1d6ad7eb08f5b915350aca43b284a"}]},"branch":"refs/heads/stable/stein"},"656a26c45ce4c9bc7931867c7db0883ff763cf2e":{"kind":"REWORK","_number":8,"created":"2021-02-22 14:21:14.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/03/775903/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/775903/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/775903/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/775903/8"}}},"commit":{"parents":[{"commit":"28f74c81e76ce4e1f84920f52a3ae5530f0097c2","subject":"Merge \"Update vagrant config\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28f74c81e76ce4e1f84920f52a3ae5530f0097c2"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-15 15:01:38.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-22 14:20:58.000000000","tz":-420},"subject":"Fix pdns4 devstack plugin and drop lower-constraints","message":"Fix pdns4 devstack plugin and drop lower-constraints\n\nThis change drops the lower-constraints test.\n(cherry picked from commit d08dd621c988fe13b9844c5c44e143907aa74d0c)\n\nAnd fixes the pdns4 devstack plugin.\n\nWe no longer need to deal with xenial or older distro versions, so we\ncan assume that we can always install pdns4 from the distro directly\nand don\u0027t need the distro version check any longer.\n\nDrop a config option that was removed in pdns 4.2.x.\n\nUse mysql credentials correctly.\n\n(cherry picked from commit 6f17fe10bb37d71944e16a422606ccf0f78c6771)\n\nChange-Id: I90af3a092296f943509833608f25522b6f8e9ab6\nDepends-On: I36799ee9d4db20104377e2772c86229b8083448d\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656a26c45ce4c9bc7931867c7db0883ff763cf2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656a26c45ce4c9bc7931867c7db0883ff763cf2e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
