)]}'
{"id":"openstack%2Fdesignate~776173","triplet_id":"openstack%2Fdesignate~master~I72f0e82ecd69a1877c628cd6bc7a032a187606b4","project":"openstack/designate","branch":"master","topic":"zone_serial","attention_set":{},"removed_from_attention_set":{"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2023-02-17 20:48:57.000000000","reason":"Change was abandoned"},"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2023-02-17 20:48:57.000000000","reason":"Change was abandoned"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-02-17 20:48:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I72f0e82ecd69a1877c628cd6bc7a032a187606b4","subject":"Increment zone serial instead of using time","status":"ABANDONED","created":"2021-02-17 13:33:31.000000000","updated":"2023-02-17 20:48:57.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a908a5ad9c0f978dbca4f8771b28292c7162f4a5","_number":776173,"virtual_id_number":776173,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-04 02:23:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 14:44:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 09:45:42.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2021-03-22 09:45:42.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"state":"CC"},{"updated":"2021-10-11 18:56:43.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"182fea1d458b28288369f94725c90915cd45f9c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-17 13:33:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"273572f50541936f6e49c83a6f258cc7c290bd0f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-17 13:33:53.000000000","message":"Patch Set 1: Workflow-1\n\nStill WIP","accounts_in_message":[],"_revision_number":1},{"id":"a814a04d20d8bf3ee04f0fe2679ed894f7ae40a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 14:44:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/180ea9f5512a420ea17e7895d8042ce3 : FAILURE in 38m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3a7568ea754547f39d218e0708e8f5fb : FAILURE in 33m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b70c1a4967c54fd68c16dd28a20bb1ed : FAILURE in 53m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ac282218fa604996bfc42e11f0607772 : FAILURE in 54m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/762c458cf52046baa33da0e8526e5add : FAILURE in 37m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a68a79dba47c434a981a1f281c463038 : FAILURE in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6467a6cdbe3a47668ab512795f620897 : FAILURE in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bc145270ef34616ba78214dc54f470f : FAILURE in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3b1a08962f74a499699a47983acdcbb : FAILURE in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/925dd61a14234cf19a6b8cbff74cf43b : FAILURE in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f05e9ce09f847a291a876645349a85f : FAILURE in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057a836bc4834c81a3b62efcf8b1a017 : SUCCESS in 7m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e65fddd140e74310a84b1edf8040775d : FAILURE in 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c4814f436778ca959b15a1c38ffc548d1fc2d9c9","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-17 20:57:59.000000000","message":"Patch Set 1:\n\nWill implement zone update batching so we don\u0027t get ridiculously large serials.","accounts_in_message":[],"_revision_number":1},{"id":"244939e908c8dd7d43f538d88de2e7147326ea40","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-29 07:27:24.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Will implement zone update batching so we don\u0027t get ridiculously large serials.\n\n\n@Nicolas: Thank you very much for this change as I believe there currently is the risk of race conditions on zone updates, especially when using IaC tools which do update things concurrently or in very rapid succession.\n\nI don\u0027t know how far along this line you are yet, but even though batching might improve the performance or avoid large serials, maybe fixing the race condition should have a higher priority for now?","accounts_in_message":[],"_revision_number":1},{"id":"27c11d25229538873a3bd35e809c38937554c5c5","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-29 07:31:22.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Will implement zone update batching so we don\u0027t get ridiculously large serials.\n\u003e \n\u003e \n\u003e @Nicolas: Thank you very much for this change as I believe there currently is the risk of race conditions on zone updates, especially when using IaC tools which do update things concurrently or in very rapid succession.\n\u003e \n\u003e I don\u0027t know how far along this line you are yet, but even though batching might improve the performance or avoid large serials, maybe fixing the race condition should have a higher priority for now?\n\nArgh, sorry I meant to add a proposal:\n\nHow about combining timestamped serials with a counter?\nThis would create a unique serial and also avoid a duplicate.\n\ni.e $timestampWithSecondsPrecision$zoneUpdatePerSecondCounter\n\nEven if the $counter might be two or three digits, but there certainly is an upper bound on the number of updates that could happen per second.","accounts_in_message":[],"_revision_number":1},{"id":"f59d914e00dc109690ec1227fc0864655d22511c","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-10-11 13:37:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84d91ac85b0436eb6a62489a237e86ddb179f4dd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-11 18:56:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2e8f69ecbf5d826018f3b7d4502b1a31896c7a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-04 01:11:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"66e8965c5349a2f920b017f8fcd4a8ca7f08732e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 02:23:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e83285dcdd8f4ae1a82a02d9fa14ea1a : FAILURE in 44m 31s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/1597750ca0614c2c9fb695d02b6ce663 : FAILURE in 52m 39s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9c3a37f77d1940eb94c87a6a2f52cfc5 : FAILURE in 46m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/07c0c58eda7b4ee3be355b5029a29956 : FAILURE in 40m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ee04bc26e48f49c390edce82c478e3a6 : FAILURE in 1h 03m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/09c687ef157d42699b406fc7180de94b : FAILURE in 49m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aedbe29397f74ebfa32005ac46b2c8cf : FAILURE in 45m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f2b734325fc4d0b9c58156aa62bd918 : FAILURE in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/316329568e8648ceba8303fd2f9df32d : FAILURE in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec81f686552f405692eda89156425777 : FAILURE in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/084c8b909aa943cdb449534ea2350db9 : FAILURE in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d327a0f6bfd44d9c967128392228113f : FAILURE in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b734cf94c00340138d2a5060972fcdb0 : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4f7266b6be2f4a53bf72f474c7fa85f8 : FAILURE in 27m 14s","accounts_in_message":[],"_revision_number":2},{"id":"a908a5ad9c0f978dbca4f8771b28292c7162f4a5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-02-17 20:48:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4622fd596000aa94119354d19c3e6e1d55c1ed2b","revisions":{"056cb62ce8dd73bccce63c7fe307219d04fe041d":{"kind":"REWORK","_number":1,"created":"2021-02-17 13:33:31.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/73/776173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/776173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/776173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/776173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/776173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/776173/1"}}},"commit":{"parents":[{"commit":"e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8c901c32318a4f6bec7cb4f6d95fcc48c57ad73"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-17 13:30:59.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-17 13:33:00.000000000","tz":-420},"subject":"Increment zone serial instead of using time","message":"Increment zone serial instead of using time\n\nChange-Id: I72f0e82ecd69a1877c628cd6bc7a032a187606b4\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/056cb62ce8dd73bccce63c7fe307219d04fe041d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/056cb62ce8dd73bccce63c7fe307219d04fe041d"}]},"branch":"refs/heads/master"},"4622fd596000aa94119354d19c3e6e1d55c1ed2b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-04 01:11:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/776173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/776173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/776173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/776173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/776173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/776173/2"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-17 13:30:59.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-04 01:11:38.000000000","tz":0},"subject":"Increment zone serial instead of using time","message":"Increment zone serial instead of using time\n\nChange-Id: I72f0e82ecd69a1877c628cd6bc7a032a187606b4\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4622fd596000aa94119354d19c3e6e1d55c1ed2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4622fd596000aa94119354d19c3e6e1d55c1ed2b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
