)]}'
{"id":"openstack%2Fdesignate~776288","triplet_id":"openstack%2Fdesignate~stable%2Fstein~I98f7f80ce365cdee33528f9964c03274f62a795a","project":"openstack/designate","branch":"stable/stein","topic":"central_lock","hashtags":[],"change_id":"I98f7f80ce365cdee33528f9964c03274f62a795a","subject":"Adding distributed locking to central","status":"MERGED","created":"2021-02-17 22:42:25.000000000","updated":"2021-02-26 10:01:25.000000000","submitted":"2021-02-26 09:59:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776288-central_lock","meta_rev_id":"decb8a88b676f02c7b76482008cc10e1afe98f6d","_number":776288,"virtual_id_number":776288,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 09:59:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2021-02-26 09:14:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2021-02-25 20:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2021-02-26 09:14:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 22:42:25.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2021-02-17 23:41:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 16:56:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-02-25 09:48:12.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"13409c8792f65b3e6617632969bd8259af7ec693","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-17 22:42:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18624cfc06a3d9ddc4455297f4b2dc0f61bff0f6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-17 22:54:16.000000000","message":"Topic central_lock removed","accounts_in_message":[],"_revision_number":1},{"id":"0d2d698b68528f10e09f9cd837b1a03b0b8d5f37","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-17 22:54:19.000000000","message":"Topic set to lock","accounts_in_message":[],"_revision_number":1},{"id":"9b04adcceeeb0db2e2fe8be11eee2b3d28592a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 23:41:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a05b266828f142499e1c5949ccb18564 : SUCCESS in 39m 34s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/e83860c62b854185925f551150c6f993 : SUCCESS in 38m 43s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/ab27feffc5d24781b33ce07405d12b95 : SUCCESS in 38m 51s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/f5b7d5bad51f45ea849bac350d6b9948 : FAILURE in 31m 18s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/067b7d17e13443d9a7afbdf7d2308174 : FAILURE in 31m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0685dac0fd9949df932b8625de1c9caa : FAILURE in 52m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5274ce85f6af4877b38ae773c277e915 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83fe2cee8a1245cf8a2d11d3e3a3b0a4 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4e4dd28d16f4779a037ffdaf7f4e0b7 : FAILURE in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78649a78649942bba132a80954684653 : SUCCESS in 5m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ffd6734f3ed743d0888d4ad3bf8cf5a7 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51c2053508644607b0d359faf0b1c744 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cdc95394cfb4d4d97bf80360ffdc50f : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a4a4e9020841b294dde3f1601b145c : SUCCESS in 4m 42s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/200a4b4fbe244bd98cade13e25f10313 : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":1},{"id":"772f54f8d12c19a998739ce277f9325b7c6d56a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-18 15:52:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e7a79fde5d7ea234efd867b7eb965a22ec3fe9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-18 16:06:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f71172117c524ef0b14bc4439302d1437caacd32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:07:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3b822c1f5d69421e9be815bc61d03eca : SUCCESS in 48m 55s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/f4271f44b00e46468c31dceef2bd7d61 : SUCCESS in 41m 29s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/cad4a23fb0b74967a05456eacf9b6fc3 : SUCCESS in 30m 38s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/dd8098fc1e464417ac8ddf432fdf144e : FAILURE in 26m 44s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/c5adfc6b0f984f13852baec558d4f8d9 : FAILURE in 25m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/965c03ffcdec418f84aef74d1537bfda : FAILURE in 48m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df41868251784703bf5bc031c8aa06c3 : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fb82ada51f04019a55376c1f97845dd : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e5ca878abf464491fb3fa4c1f35b04 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3941b5ac8834599b9e0d82369bcaa75 : SUCCESS in 5m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b45386f288c411a86140dc64f5edd08 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ee13698b54b4b88abf03c393f7c3f7e : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ae075d1719a453f998835ec29087d02 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2259f5e2790040cb9c8d45b82ae31189 : SUCCESS in 6m 30s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/12650e0e97d64b329c4328e885e1c94b : SUCCESS in 30m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a1bf114664fde93caedc7379b0a6071cf71c67e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-24 16:25:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"69a14c7ed61a30935d3c783e802ae72cf47b622c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 16:56:03.000000000","message":"Patch Set 4: Code-Review+1\n\nI am ok with this. I would like to see a release note attached however.","accounts_in_message":[],"_revision_number":4},{"id":"f7bc00d63b6e2efe866bbb640d247607b27b2a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 17:15:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9367fa9f37144770a2b45787b6ec38ea : SUCCESS in 36m 59s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/b65d941b3cb748fd8db63387d8c890be : SUCCESS in 37m 50s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/c3ef87a246b14b13806acc6c40406786 : SUCCESS in 38m 32s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/8bbc97d9d9564463ade6ad1103d70043 : SUCCESS in 31m 54s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/fcd33e61b38b4310bdfaf8d5d56e55b8 : SUCCESS in 37m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ae7568ac5db04e23a14af60b2b7adb0f : FAILURE in 34m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/052f6ce6a4434098b9d8d389d5297955 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b5b44f099d84bd8989881852c0cfeaa : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6958177cb3ec4a91a72fcfdfb92a2ddc : SUCCESS in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/21021af103f54670a17a5ebb068f4692 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2de0208d62f0466aab7e59b9c1b7634b : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c68b99d6c2a0493dbf54303bc2da1bbf : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49af6ff3ebeb4870ae169a6268a9a5dd : SUCCESS in 4m 55s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/a6167a242e2e48b892c0754080cab2f4 : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":4},{"id":"06baaa11592a765b68b0b79bd91701150dfb71d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-24 19:12:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae2528b09ec74405cb7894b2e3a941662f00256f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 20:12:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/18277387d66f4b219b443043563fc862 : SUCCESS in 37m 21s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/3134cd743b5c4454b6de1fa94058d0c1 : SUCCESS in 40m 19s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/05e841d61ca3436d8f96896dde1fcf82 : SUCCESS in 41m 36s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/6fe1d14cc5a2415493264543b3a436df : SUCCESS in 25m 54s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/93d0c143150a4622984c265946158aa1 : SUCCESS in 32m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/29a772ceabab4f0494808874d97eaf9c : SUCCESS in 52m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13e30565b2fa4bf89d674dcfc7fc6bae : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0bcba13357540bcace9ec2f69be26d6 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3237f6a6b3d94daeb139c2bb2c48809b : POST_FAILURE in 20m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e587c341ec6f4db3a2436cd6e8404984 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b62a4e70820741e6958b3e88f13da37a : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/459c8e7a88fd41f494641e450917509d : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fdff151aba7476dac147cadf1f96f02 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72255bcda31d4bc7bed56c5488abd24b : SUCCESS in 5m 06s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/4857514b98234fd1bef684a5c5319093 : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":5},{"id":"455f4849090a34c32eaf5cec906b83636bd1dca3","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-24 20:29:22.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"60b904559c6fd2049ee3e1528510bb35a039235a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 20:51:51.000000000","message":"Patch Set 5: Code-Review+1\n\nThank you, LGTM","accounts_in_message":[],"_revision_number":5},{"id":"bb1587ab06fe5d52685d6d48616e9389dafcc656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 21:27:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6adb8f6142ef4ef68aad4a7e24dffb93 : SUCCESS in 36m 37s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/702382a6e53a49eaa3e0ea34dfd8a949 : SUCCESS in 30m 59s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/75d16ddb62c34bc89901770381542547 : SUCCESS in 51m 22s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/80a0aa695acd4296abf16721a798ef77 : SUCCESS in 44m 03s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/f2a25110d72c4cc6996a4ee84791611a : SUCCESS in 31m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f529cc2a3f0b47ee8c13a4efa83a985a : SUCCESS in 51m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad323ae4b11d49d1882892797d8cf915 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/492436c44f2940b7a95e342eff918256 : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4066ee6950474f0bbcc3479ae5855a9e : SUCCESS in 5m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1c450552b2c04969b03107599fd366e3 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c26d1760a1343de99c2a5fd01701157 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c796f7e055f467dacba60007e0f0d69 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/159043882dbb44238a316a951f1576e2 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deceb805f84c45daa3a525a7e66aa3d0 : SUCCESS in 5m 38s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/f3581a96e0444aa89bad6a5be58c8418 : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":5},{"id":"bcc4952b2c1145d79f50de0f1d66016fedb93686","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-02-25 00:48:33.000000000","message":"Patch Set 5: Code-Review+2\n\nI like the addition of the release note (not sure why I didn\u0027t create one originally), but might be worth adding that to all releases.","accounts_in_message":[],"_revision_number":5},{"id":"a4d84cc7187cb5fa1da4a35f50539602834f048b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-02-25 09:48:12.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nNit about the release note, though I\u0027m not actually sure it is worth respinning.","accounts_in_message":[],"_revision_number":5},{"id":"2ed6b0e6b47beac40d386721aae1e87ab49a9255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-25 14:08:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3a03213815211109f438dca4e57cf6e8fa9ba3cc","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-25 14:09:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"61f1aae801690a232ba37f1952c56e231796d8d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 14:56:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f25c448686e24c24b89ae0f0cdff9bb7 : SUCCESS in 35m 17s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/987a67352c6c494bb699fc9da3835a81 : SUCCESS in 37m 48s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/436d44693d8549dfa34ac561f5f1ff6d : SUCCESS in 42m 52s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/a0a206cb0b6f4e7caae728ca1247cf90 : SUCCESS in 37m 45s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/e180bb871ad8499998b805730a8783ad : SUCCESS in 30m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/642e1eb76c144c80a0a9cb22afa27ceb : SUCCESS in 44m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42c2bb4da5a24c11a32804921dc0f59d : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f4ea3e53654921ae471b56964db320 : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50860c38e9b14946b1aaab0ca7e0ee2d : SUCCESS in 6m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/21dc94be72d24ddca8cddf9edbfc6984 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfcb70add95342729d7da49604921b0a : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44d239771ca043a497abd42762abde38 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15ef944093c74816a457462f9277c749 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b57e501a14cd4cb99e27171b590fe9bf : SUCCESS in 4m 19s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/51360fe88f50413ebd5280bcfc32b55b : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":6},{"id":"2b68e334b4ad2a77cd7e797082278124d04d785e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-25 15:08:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e04803208987f5350a52a83d1714ccf529bb569a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-02-25 15:38:04.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9c3809fa339d9689f7d208a7636ee40de0660f06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-25 15:42:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"69c9086662272d4d7a130b0c8cebf19cc0be0173","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-25 15:42:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ac4281c39396d76e1fb863ee385de665f2dce618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 16:45:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d127da5960cc4f8099d7662dc78b7f72 : SUCCESS in 44m 38s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/6e279ea2708742ac8acb923e7b6a7de8 : SUCCESS in 37m 49s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/26ffd70e134e44acb43ef1873f38858a : SUCCESS in 40m 19s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/6f876d18c9084318965ea7554c91e8d9 : SUCCESS in 36m 36s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/6ebefb9add82403294660a23b966eae9 : SUCCESS in 37m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/95660b1bf6c94fada2e01d8423f4c96f : SUCCESS in 57m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b1bdd39a71a4d4097eea1e01019d9a9 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31280b4920fb4787af90f2a752cdaf82 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbf9ab7fc8414ebba84d79da4f29ca3c : SUCCESS in 6m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c03bb03fedc447faeb9b485d69180d0 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1329e67a581d44fa9f8d62f6d3a6c3ac : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db7716ea6d234ccba3cb8a6d1678c9a0 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f371903de449f093518447278d5edb : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d06fb47e74fc4a81bc0e9e89124ec3d1 : SUCCESS in 5m 20s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/01b911117cf147309bc47ca56fdba38a : SUCCESS in 27m 53s","accounts_in_message":[],"_revision_number":8},{"id":"efeb36ed801c1a40d3122454007740f7ee7ebcca","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-02-25 20:53:52.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"9bea85ed31a823f866a6e7c9dbacb4a6f3c93582","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-02-26 09:14:05.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"e141ef6274e47cb633f1fe4d77c3a14bc7b4e6ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 09:14:20.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"87afbdcb71f11ff98f5107b9e34779f8e2fd096c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 09:59:57.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2cc71fd9d8c74b5381e91f06042139f2 : SUCCESS in 41m 39s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/581b132f29214a77b1386971a98b45c2 : SUCCESS in 42m 01s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/11f33f9a62c94f5698102a5604a772c7 : SUCCESS in 43m 35s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/1a1964bde11b473498fbb82301356339 : SUCCESS in 41m 52s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/a8dc1fb3593d401bb65a9c4c1f2dfebd : SUCCESS in 41m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11cc6ab9997b43e0bdd4d62a7d17be3f : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/214e10eb35f444f2a269bf12766dfb74 : SUCCESS in 7m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/56656ee10743422495a250dd952e09df : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22e2d1c6b06f4f71931d8db31e3d5c7f : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/756f45ee544c4e938b4f70215bc34453 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b0fa1bd5a1047f8b02fea2b1c1fe9da : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/370bc53505fd4e71981c4666f25154ea : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":8},{"id":"dc0875f5b3f9f12ff1bf0cd62b9b17e33f45c713","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 09:59:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"decb8a88b676f02c7b76482008cc10e1afe98f6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 10:01:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5900c63b9279413b991b5bddb1b9c6d4 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f239a9f38e354929b636dc1429517edb : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"acc82755f1e8df8bb8f0ab7b4524f691637a312e","revisions":{"29bc92f9518d0800ef29bf743ddc4842b11c7767":{"kind":"REWORK","_number":1,"created":"2021-02-17 22:42:25.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/88/776288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/776288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/776288/1"}}},"commit":{"parents":[{"commit":"28f74c81e76ce4e1f84920f52a3ae5530f0097c2","subject":"Merge \"Update vagrant config\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28f74c81e76ce4e1f84920f52a3ae5530f0097c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-17 22:37:28.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1871332\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit f6090d885c8ebe00a89307a1e2f5621de75dcda8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29bc92f9518d0800ef29bf743ddc4842b11c7767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29bc92f9518d0800ef29bf743ddc4842b11c7767"}]},"branch":"refs/heads/stable/stein"},"9d28d3cbc50867ff1118375da461ea302f54bef6":{"kind":"REWORK","_number":2,"created":"2021-02-18 15:52:28.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/88/776288/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/776288/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/776288/2"}}},"commit":{"parents":[{"commit":"28f74c81e76ce4e1f84920f52a3ae5530f0097c2","subject":"Merge \"Update vagrant config\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28f74c81e76ce4e1f84920f52a3ae5530f0097c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-18 15:52:14.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1871332\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit f6090d885c8ebe00a89307a1e2f5621de75dcda8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d28d3cbc50867ff1118375da461ea302f54bef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d28d3cbc50867ff1118375da461ea302f54bef6"}]},"branch":"refs/heads/stable/stein"},"d6815979725eb46019ef9205c3fd568dde40e951":{"kind":"REWORK","_number":3,"created":"2021-02-18 16:06:47.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/88/776288/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/776288/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/776288/3"}}},"commit":{"parents":[{"commit":"28f74c81e76ce4e1f84920f52a3ae5530f0097c2","subject":"Merge \"Update vagrant config\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28f74c81e76ce4e1f84920f52a3ae5530f0097c2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-18 16:05:54.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1871332\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit f6090d885c8ebe00a89307a1e2f5621de75dcda8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6815979725eb46019ef9205c3fd568dde40e951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6815979725eb46019ef9205c3fd568dde40e951"}]},"branch":"refs/heads/stable/stein"},"2638c885475d15c0bf5faa4d65512276488c1768":{"kind":"REWORK","_number":4,"created":"2021-02-24 16:25:36.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/88/776288/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/776288/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/776288/4"}}},"commit":{"parents":[{"commit":"656a26c45ce4c9bc7931867c7db0883ff763cf2e","subject":"Fix pdns4 devstack plugin and drop lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656a26c45ce4c9bc7931867c7db0883ff763cf2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-24 16:25:24.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1871332\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit f6090d885c8ebe00a89307a1e2f5621de75dcda8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2638c885475d15c0bf5faa4d65512276488c1768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2638c885475d15c0bf5faa4d65512276488c1768"}]},"branch":"refs/heads/stable/stein"},"28bef8ad5e32b26a5563cf370e35f4ded5daebcd":{"kind":"REWORK","_number":5,"created":"2021-02-24 19:12:46.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/88/776288/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/776288/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/776288/5"}}},"commit":{"parents":[{"commit":"656a26c45ce4c9bc7931867c7db0883ff763cf2e","subject":"Fix pdns4 devstack plugin and drop lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656a26c45ce4c9bc7931867c7db0883ff763cf2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-24 19:12:34.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1871332\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit f6090d885c8ebe00a89307a1e2f5621de75dcda8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28bef8ad5e32b26a5563cf370e35f4ded5daebcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28bef8ad5e32b26a5563cf370e35f4ded5daebcd"}]},"branch":"refs/heads/stable/stein"},"95d3b62c617a2779b3bd421284d97c381b824067":{"kind":"REWORK","_number":6,"created":"2021-02-25 14:08:13.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/88/776288/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/776288/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/776288/6"}}},"commit":{"parents":[{"commit":"656a26c45ce4c9bc7931867c7db0883ff763cf2e","subject":"Fix pdns4 devstack plugin and drop lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656a26c45ce4c9bc7931867c7db0883ff763cf2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-25 14:08:05.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1871332\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit f6090d885c8ebe00a89307a1e2f5621de75dcda8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/95d3b62c617a2779b3bd421284d97c381b824067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/95d3b62c617a2779b3bd421284d97c381b824067"}]},"branch":"refs/heads/stable/stein"},"1e9c3e2873f957725ca84c26c5dbdcbfb8a22e0b":{"kind":"REWORK","_number":7,"created":"2021-02-25 15:08:54.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/88/776288/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/776288/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/776288/7"}}},"commit":{"parents":[{"commit":"656a26c45ce4c9bc7931867c7db0883ff763cf2e","subject":"Fix pdns4 devstack plugin and drop lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656a26c45ce4c9bc7931867c7db0883ff763cf2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-25 15:08:44.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1871332\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit f6090d885c8ebe00a89307a1e2f5621de75dcda8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e9c3e2873f957725ca84c26c5dbdcbfb8a22e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e9c3e2873f957725ca84c26c5dbdcbfb8a22e0b"}]},"branch":"refs/heads/stable/stein"},"acc82755f1e8df8bb8f0ab7b4524f691637a312e":{"kind":"REWORK","_number":8,"created":"2021-02-25 15:42:14.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/88/776288/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/776288/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/776288/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/776288/8"}}},"commit":{"parents":[{"commit":"656a26c45ce4c9bc7931867c7db0883ff763cf2e","subject":"Fix pdns4 devstack plugin and drop lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656a26c45ce4c9bc7931867c7db0883ff763cf2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-25 15:42:08.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1871332\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n(cherry picked from commit f6090d885c8ebe00a89307a1e2f5621de75dcda8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/acc82755f1e8df8bb8f0ab7b4524f691637a312e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/acc82755f1e8df8bb8f0ab7b4524f691637a312e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
