)]}'
{"id":"openstack%2Fdesignate~776294","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I9a0881c942eed3ccb5c364c3445907722913f20e","project":"openstack/designate","branch":"stable/ussuri","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I9a0881c942eed3ccb5c364c3445907722913f20e","subject":"WIP DNM more debug for devstack","status":"ABANDONED","created":"2021-02-17 23:28:31.000000000","updated":"2021-11-03 17:18:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9361571ce3be65ab9b3ae0398a23ea8debc3a07c","_number":776294,"virtual_id_number":776294,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-18 00:14:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 00:14:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5622f6bb7c91fc4b4435e8153c8e0dd3ea811f33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-17 23:28:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31eb71e6497d167c1261d1c642f827f40d8f2175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 00:14:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ff308ac0e34149509a3b94efe4fcf74e : SUCCESS in 39m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1789acf16b0649c8b830ea44bd5ae660 : SUCCESS in 30m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1f941a1a9b3f4b028fe6f17758bdc735 : FAILURE in 34m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c84a118e03a94d4fa3d788db96e68d6d : SUCCESS in 43m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b24677f8948a4249b474e35f842df543 : SUCCESS in 29m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47f82e838763476b99690a228c39f96e : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de569d8eb2746f9a564be6337e28033 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3829fcf81174422bae943b5f0340f98d : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24efa88bf7d241539ed16ff35e166365 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fde676b76e495c96105b04a0223b55 : SUCCESS in 8m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/821e25f57582465783ac599edbf74222 : SUCCESS in 33m 02s","accounts_in_message":[],"_revision_number":1},{"id":"87a8803fb1b06e0e6a7d74dfe86f77eb23dd33e6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-06-03 11:08:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b95c0bf12785617033196787fee243b8af95ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 11:48:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/97badc1a5bfd4187ab5797fe60927f6c : SUCCESS in 37m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7b07ebfd790d472ea853a03a823cb40b : SUCCESS in 37m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e68b62d6614548729c9f114f2aa3c5d3 : FAILURE in 28m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dd051ace069748cc914cd21a440b5bb0 : FAILURE in 39m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f45331ff1eeb475d96f059b8a5972b06 : SUCCESS in 35m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80fba41b65ff4aebb65c25c4321febf1 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135302a2cd174b3c868b3c735c351a1e : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d0b4e0e2b9e44d591935f5c324aebac : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88a850ed09f240269346692c08859771 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72a4d65ad5545a29d4900829da3089c : SUCCESS in 7m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e19c8346680d4f14b34c39aad230d74c : SUCCESS in 30m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9361571ce3be65ab9b3ae0398a23ea8debc3a07c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-03 17:18:11.000000000","message":"Abandoned\n\nCleanup for Ussuri EM","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a03f54e08a583a73a52c01498afc8222228696a3","revisions":{"a03f54e08a583a73a52c01498afc8222228696a3":{"kind":"REWORK","_number":1,"created":"2021-02-17 23:28:31.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/94/776294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/776294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/776294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/776294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/776294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/776294/1"}}},"commit":{"parents":[{"commit":"e47770bdf33a69df5fb9a6db01c404f30a10d299","subject":"Native Zuul v3 designate-grenade-pdns4 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e47770bdf33a69df5fb9a6db01c404f30a10d299"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-02-17 23:27:33.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-02-17 23:27:33.000000000","tz":60},"subject":"WIP DNM more debug for devstack","message":"WIP DNM more debug for devstack\n\nDepends-On: https://review.opendev.org/776293\nChange-Id: I9a0881c942eed3ccb5c364c3445907722913f20e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a03f54e08a583a73a52c01498afc8222228696a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a03f54e08a583a73a52c01498afc8222228696a3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
