)]}'
{"id":"openstack%2Fdesignate~777866","triplet_id":"openstack%2Fdesignate~master~Id4da1901b6e2e689c5962d8739695c8fa6f2c5bf","project":"openstack/designate","branch":"master","topic":"nameserver_records_in_parent_zones","hashtags":[],"change_id":"Id4da1901b6e2e689c5962d8739695c8fa6f2c5bf","subject":"Allow NS records for child zone in parent zone","status":"ABANDONED","created":"2021-02-27 17:10:40.000000000","updated":"2022-09-24 05:53:18.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"7707618be39136cff3b33de9bcad5da168c55602","_number":777866,"virtual_id_number":777866,"owner":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-21 02:48:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2021-03-22 21:28:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-27 18:05:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-01 04:14:53.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2021-03-01 10:34:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-03-22 21:28:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"31cc114838d01d3d8b511f0638303c1810373b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"date":"2021-02-27 17:10:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"272e555b6f909f75417b2e8195721bfdf6c393d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 18:05:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/118c30b5c51243e0ae76a7f4dea7087c : SUCCESS in 38m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a00d36f8a6e74e77bac7461239ca221f : SUCCESS in 46m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2fd55f23346742c19fd86b18cd522888 : SUCCESS in 53m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5acfc0a7197440e6b3ffe01ef80aac12 : SUCCESS in 47m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fa59884d91934c358cd533f23f7ee5b9 : SUCCESS in 37m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dfa2b7928664c11bcf2fbc2c7046730 : FAILURE in 5m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e331516a07349ed8dfac617186ca86c : FAILURE in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4541963994a14e69a4e91506f5ec3533 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e03dcb7d00b4203b2b6da5b623bae15 : FAILURE in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19af2eb5d2c349b3a82269050dc63eb4 : FAILURE in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9847d244e744fef8d5069e86eb52b37 : FAILURE in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23669647b86f4625844699f742ea032a : SUCCESS in 8m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c648a3499ee2428b8408b02c91464069 : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1d6fb10464bf3eaf4c1af777cbae4cfd1686c854","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-01 04:14:53.000000000","message":"Patch Set 1:\n\nTo get the CI passing you\u0027ll need to add values for recordset_type to all functions calling self.service._is_valid_recordset_placement_subzone here.\nhttps://github.com/openstack/designate/blob/75dc32815cdcac1f96c91fa0e2ca889166cb647b/designate/tests/unit/test_central/test_basic.py#L701","accounts_in_message":[],"_revision_number":1},{"id":"0e545da746770edb24ca19e6fc2ded5289d36746","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-03-01 10:34:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI\u0027m still seeing the associated bug as invalid, see the discussion there. Even then, you\u0027d not only need to fix the existing tests, but also add a new test for your use case.","accounts_in_message":[],"_revision_number":1},{"id":"2d869ae30dc89716682e018fb7af825f43553ca3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-03-01 10:38:25.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nOoops, the bug reference is in fact missing.","accounts_in_message":[],"_revision_number":1},{"id":"f6e69ca14c03133f7ef4de58f27d8e54fa6f38a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"date":"2021-03-13 11:13:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6a4abb0734142d695f94dac0a3049d60aa469fad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"date":"2021-03-13 11:21:35.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b72539f71d2afb666dcbad2caeaeb13907bc2bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"date":"2021-03-13 11:25:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c83a6c2a9bda284dde82f9cf5a23443f60b52177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 12:19:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e520b46d51df4c06a64c8ccb7043912d : SUCCESS in 43m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7dcd3a4828594da68eb410b1aebcdf9d : SUCCESS in 34m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/65953650a70348a09c1a83172044d971 : SUCCESS in 52m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5d86421675ab4f41bbde8118a03a60ec : SUCCESS in 36m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/13f0450e154b4b0fa04152659bec0d68 : SUCCESS in 39m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cf9cef134db49e39af4cdf76bbbdc9f : FAILURE in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a064a1d1f27b45c58a4b136974bdce2e : FAILURE in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd2f843d2884bcb9f84f05ecdb883d8 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6726cb7dfa664199aa7cad1f29fbe676 : FAILURE in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/defa27829fab4f4aa69bccb608d61eb6 : FAILURE in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e20b95b5bee142e2bccec0a229029f86 : FAILURE in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ccd0bdb81184117b869bab7915e0e74 : SUCCESS in 7m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dd33a0c2ae9b4f0fb24d066938173c35 : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":4},{"id":"a98a1a2a22a7b369e7c259d7747aa4cc228b3eef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"date":"2021-03-13 18:07:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5e5cc578cc0d3c463bce944a012891f71eef3043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-13 19:15:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1c6aa07f164b4c9ca5f8429af903621d : SUCCESS in 29m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/961431784e8241918eff1fda8a14c33f : SUCCESS in 38m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/415f385c3ccc4498a8589d068d975762 : SUCCESS in 53m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e0fa39686a234fcc82fe5dbe0f512b95 : SUCCESS in 31m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e9663d6c07d34b298949fe85709d7d5e : SUCCESS in 29m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01a811aceb39487598dbe4ce2764501a : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32612cb2386b47cc97721ff341169c72 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a46d386368594c13bacaecaa53f48e1f : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7713d0a5a6794c33a128adee26d861c0 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b88ab2fd55424e63af62b9b168a2680d : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2890ea1369d24086b1d4440edf47e9d0 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0574d16fea6d42aa84b4aa6d22699785 : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3fbc78aa24684c98862539cbfc159f4b : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":5},{"id":"1a3e68d29beb21bf7ea61ae2d89753d529faa2ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"date":"2021-03-21 01:37:35.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"c739405ef454ad61a34aec95747d4b61d68de6e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-21 02:48:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fe02e4c8a748498185c846e79fc39816 : SUCCESS in 38m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c1092716e41547008046df20afc591e4 : SUCCESS in 44m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2a7856fd66054316badd41de3f0f8db0 : SUCCESS in 1h 06m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5b51ab844c1a4ec38132f533512321ae : SUCCESS in 40m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1ef8753111bf4a4baf9cd476f4fedea9 : SUCCESS in 33m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbb01bad2b7f40b28bb06fa2b159b044 : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a797ce7cdd2c4c4aa161dd43653c08db : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06347d2866f24ad1a0bcd5a1dfddd067 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df1b8feef9524b4ba409f32739a82028 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/296190f527324e5f8a66dd527d785bd7 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb2188f54f4344c9bee10a8b9fc0f72a : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b86a6504a2c435e9b1362288382d88e : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90f755352f6a40d1b656deeaf9f9c26b : SUCCESS in 26m 05s","accounts_in_message":[],"_revision_number":6},{"id":"fabb5c9b19e803554e7f9d4f79d9e8745d7aa1a6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-22 21:28:28.000000000","message":"Patch Set 6: Code-Review-1\n\nI am -1 holding this patch as I cannot reproduce the issue reported in the bug. I am asking for more information on the bug. If I can\u0027t reproduce it, it is hard to review.","accounts_in_message":[],"_revision_number":6},{"id":"7707618be39136cff3b33de9bcad5da168c55602","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:53:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3c2d20f1acb88187bdbf57b36ce160633d6179bf","revisions":{"c2ec2cb16e40d95d50908556f9956e1ea844daf4":{"kind":"REWORK","_number":1,"created":"2021-02-27 17:10:40.000000000","uploader":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"ref":"refs/changes/66/777866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/777866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/777866/1"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Alex","email":"alex@alexcarder.com","date":"2021-02-27 17:04:24.000000000","tz":-480},"committer":{"name":"Alex","email":"alex@alexcarder.com","date":"2021-02-27 17:04:24.000000000","tz":-480},"subject":"Adding support for NS records in a parent zone pointing to existing   child zone","message":"Adding support for NS records in a parent zone pointing to existing\n  child zone\n\nChange-Id: Id4da1901b6e2e689c5962d8739695c8fa6f2c5bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2ec2cb16e40d95d50908556f9956e1ea844daf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2ec2cb16e40d95d50908556f9956e1ea844daf4"}]},"branch":"refs/heads/master"},"a46b757eca6c268d427de25edb454aa744259f7a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-13 11:13:46.000000000","uploader":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"ref":"refs/changes/66/777866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/777866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/777866/2"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Alex","email":"alex@alexcarder.com","date":"2021-02-27 17:04:24.000000000","tz":-480},"committer":{"name":"Alex Carder","email":"carderalex@gmail.com","date":"2021-03-13 11:13:46.000000000","tz":0},"subject":"WIP Allow NS records for child zone in parent zone","message":"WIP Allow NS records for child zone in parent zone\n\nIn order for DNS recursion to work properly in cases where a\nchild zone (foo.example.org) doesn\u0027t exist on the same authoritative\nserver as the parent zone (example.org), users should be allowed to\nadd NS records for child zones to existing parent\nzones.\n\n\nCloses-Bug: 1917099\nChange-Id: Id4da1901b6e2e689c5962d8739695c8fa6f2c5bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a46b757eca6c268d427de25edb454aa744259f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a46b757eca6c268d427de25edb454aa744259f7a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e8f183638941b01bb063c5d7d094497b33cdcc2d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-13 11:21:35.000000000","uploader":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"ref":"refs/changes/66/777866/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/777866/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/777866/3"}}},"commit":{"parents":[{"commit":"37b3fa2ea454183e4a5c6a099eca5129959d10e1","subject":"Merge \"[goal] Deprecate the JSON formatted policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/37b3fa2ea454183e4a5c6a099eca5129959d10e1"}]}],"author":{"name":"Alex","email":"alex@alexcarder.com","date":"2021-02-27 17:04:24.000000000","tz":-480},"committer":{"name":"Alex Carder","email":"carderalex@gmail.com","date":"2021-03-13 11:21:35.000000000","tz":0},"subject":"WIP Allow NS records for child zone in parent zone","message":"WIP Allow NS records for child zone in parent zone\n\nIn order for DNS recursion to work properly in cases where a\nchild zone (foo.example.org) doesn\u0027t exist on the same authoritative\nserver as the parent zone (example.org), users should be allowed to\nadd NS records for child zones to existing parent\nzones.\n\n\nCloses-Bug: 1917099\nChange-Id: Id4da1901b6e2e689c5962d8739695c8fa6f2c5bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8f183638941b01bb063c5d7d094497b33cdcc2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8f183638941b01bb063c5d7d094497b33cdcc2d"}]},"branch":"refs/heads/master","description":"Rebase"},"7f1729565426815be9a6791ad819ecaabbadba61":{"kind":"REWORK","_number":4,"created":"2021-03-13 11:25:41.000000000","uploader":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"ref":"refs/changes/66/777866/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/777866/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/777866/4"}}},"commit":{"parents":[{"commit":"37b3fa2ea454183e4a5c6a099eca5129959d10e1","subject":"Merge \"[goal] Deprecate the JSON formatted policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/37b3fa2ea454183e4a5c6a099eca5129959d10e1"}]}],"author":{"name":"Alex","email":"alex@alexcarder.com","date":"2021-02-27 17:04:24.000000000","tz":-480},"committer":{"name":"Alex Carder","email":"carderalex@gmail.com","date":"2021-03-13 11:23:55.000000000","tz":-480},"subject":"WIP Allow NS records for child zone in parent zone","message":"WIP Allow NS records for child zone in parent zone\n\nIn order for DNS recursion to work properly in cases where a\nchild zone (foo.example.org) doesn\u0027t exist on the same authoritative\nserver as the parent zone (example.org), users should be allowed to\nadd NS records for child zones to existing parent\nzones.\n\nCloses-Bug: 1917099\nChange-Id: Id4da1901b6e2e689c5962d8739695c8fa6f2c5bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f1729565426815be9a6791ad819ecaabbadba61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7f1729565426815be9a6791ad819ecaabbadba61"}]},"branch":"refs/heads/master"},"ef83174feb4f0872a8ba821cf05c38e195d42589":{"kind":"REWORK","_number":5,"created":"2021-03-13 18:07:39.000000000","uploader":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"ref":"refs/changes/66/777866/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/777866/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/777866/5"}}},"commit":{"parents":[{"commit":"37b3fa2ea454183e4a5c6a099eca5129959d10e1","subject":"Merge \"[goal] Deprecate the JSON formatted policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/37b3fa2ea454183e4a5c6a099eca5129959d10e1"}]}],"author":{"name":"Alex","email":"alex@alexcarder.com","date":"2021-02-27 17:04:24.000000000","tz":-480},"committer":{"name":"Alex Carder","email":"carderalex@gmail.com","date":"2021-03-13 18:07:26.000000000","tz":-480},"subject":"WIP Allow NS records for child zone in parent zone","message":"WIP Allow NS records for child zone in parent zone\n\nIn order for DNS recursion to work properly in cases where a\nchild zone (foo.example.org) doesn\u0027t exist on the same authoritative\nserver as the parent zone (example.org), users should be allowed to\nadd NS records for child zones to existing parent\nzones.\n\nCloses-Bug: 1917099\nChange-Id: Id4da1901b6e2e689c5962d8739695c8fa6f2c5bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef83174feb4f0872a8ba821cf05c38e195d42589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef83174feb4f0872a8ba821cf05c38e195d42589"}]},"branch":"refs/heads/master"},"3c2d20f1acb88187bdbf57b36ce160633d6179bf":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-03-21 01:37:35.000000000","uploader":{"_account_id":33167,"name":"Alex Carder","email":"carderalex@gmail.com","username":"carderalex"},"ref":"refs/changes/66/777866/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/777866/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/777866/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/777866/6"}}},"commit":{"parents":[{"commit":"37b3fa2ea454183e4a5c6a099eca5129959d10e1","subject":"Merge \"[goal] Deprecate the JSON formatted policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/37b3fa2ea454183e4a5c6a099eca5129959d10e1"}]}],"author":{"name":"Alex","email":"alex@alexcarder.com","date":"2021-02-27 17:04:24.000000000","tz":-480},"committer":{"name":"Alex Carder","email":"carderalex@gmail.com","date":"2021-03-21 01:37:35.000000000","tz":0},"subject":"Allow NS records for child zone in parent zone","message":"Allow NS records for child zone in parent zone\n\nIn order for DNS recursion to work properly in cases where a\nchild zone (foo.example.org) doesn\u0027t exist on the same authoritative\nserver as the parent zone (example.org), users should be allowed to\nadd NS records for child zones to existing parent\nzones.\n\nCloses-Bug: 1917099\nChange-Id: Id4da1901b6e2e689c5962d8739695c8fa6f2c5bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c2d20f1acb88187bdbf57b36ce160633d6179bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c2d20f1acb88187bdbf57b36ce160633d6179bf"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
