)]}'
{"id":"openstack%2Fdesignate~778117","triplet_id":"openstack%2Fdesignate~master~I80fe08238005a94161e2dbcc89e77c90cde0a715","project":"openstack/designate","branch":"master","topic":"daitan_ns1","hashtags":[],"change_id":"I80fe08238005a94161e2dbcc89e77c90cde0a715","subject":"Add NS1 backend","status":"MERGED","created":"2021-03-02 00:31:54.000000000","updated":"2021-03-08 22:49:19.000000000","submitted":"2021-03-08 22:47:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778117-daitan_ns1","meta_rev_id":"7d07b6ad3a31e2bed1f92a6385929db898e1f16b","_number":778117,"virtual_id_number":778117,"owner":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-08 22:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-03-08 21:30:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-08 20:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-08 21:53:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 00:34:54.000000000","updated_by":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 10:13:08.000000000","updated_by":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2021-03-06 04:46:40.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2021-03-08 16:29:12.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"cb0486f7e8cf84d5955ce6c7e11c82fe6002ff24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-02 00:31:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d4024f3ee8ea82787e762f05e0fbfed21aa99c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 01:46:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3570c2918f89468c88c0cbd8685bb661 : SUCCESS in 3m 05s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ab91c0ba921f435c92178caa7543299d : SUCCESS in 40m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f3e0627d340449f8afb927708234ab40 : SUCCESS in 34m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a129402dcf32488e9bee9f68901f211e : SUCCESS in 1h 12m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0b8707c7bf844430964c0e5e003c823c : SUCCESS in 37m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/07b25ef5ce564e58a6a5efde29294fda : SUCCESS in 53m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/369b7048cc6b499b8df9b0c1156bc03e : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0febeb4b024844198bb006a03a014bc7 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcbeb9063f3942f59cf5c1188e8b5d2f : FAILURE in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d35efe8813b84ea594670d8d796cc59b : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb58d96dba1a4787a51473a7f1eb15eb : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4caa7b176f247c7b5160b9704d8c5dd : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39baa48af66a4dccb3d82fa7c647ccd8 : SUCCESS in 7m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/06812970f74c4cb5bb5e81a0e811aed3 : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f1bbded72279278e332b6947c02d065c51156fd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-02 02:21:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d3adce204af1a0de9cb8b7fda8fe49cb492fd09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 03:17:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c5a4f04d0b58429a950d0b7d73d35390 : SUCCESS in 2m 28s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4b8ef64149d74c1ea8c4d735372eccbe : SUCCESS in 38m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eccb151f928b4d33ba636d3d6303e497 : SUCCESS in 37m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8197a5512aff42caa047fce3d139f70a : SUCCESS in 54m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/641f669dd1504c7d94db7c11b0d9da50 : SUCCESS in 41m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6e94c5bd8ca44bba9712a20d4ed42b78 : SUCCESS in 35m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20e60d47e57f4bd79f99c01a3b74df3b : FAILURE in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61f475acef7849a9b5137eac3cde9a44 : FAILURE in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c59e0d459ef4d32a14e4605f8507e3d : FAILURE in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce5e01b4660747d9b3427c88bb2697b5 : FAILURE in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3370d033b2fd43df94a2fe5cdd228a1f : FAILURE in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1720a5abad654618a58b5ec058ff39c6 : FAILURE in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab46016d1d14d808e060eb24be96ed9 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bbc0cdb653344e9b9496d785e40ff4ae : SUCCESS in 28m 56s","accounts_in_message":[],"_revision_number":2},{"id":"36b4b1d0a95af0c43dfdb22ffb9adf3d0d8d5fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-02 03:42:01.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fb3e3d760538ddcc650e5dc08a0e0f3170092714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-02 03:43:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"988251a154c454ab40899e8c47bb0df48b94c551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 04:43:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/21bf3ba8abd54f938a5be5f8097626b0 : SUCCESS in 3m 27s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/02db6c6a740142dfb3b77e7c57fb44ed : SUCCESS in 36m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2c99c1e7d38a4ba3a75c73ade7b22704 : SUCCESS in 41m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3eddde2de89d4ddfb310c2b61c8b5dad : SUCCESS in 58m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b04f63b179fa4adc864bdc739e922185 : SUCCESS in 50m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e6524fe74fc04b97b7bc0e97a470555d : SUCCESS in 32m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05b6c74a1f8b43d0b2e4d3b86396acb8 : FAILURE in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1619ff68d98948f18daff743caaaaed9 : FAILURE in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aacca4e02dcc455da59ca7015e70b362 : FAILURE in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3033a300c957416e8f78c7bdad36fa9d : FAILURE in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca2237ce6ebf415aa3abed64a3b5250b : FAILURE in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffd490732999489495bcdf1543ad2978 : FAILURE in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7450a4d6ece406daa6760f479a95d6b : SUCCESS in 10m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af45e292cd7348c8a68b3955d6b7ddbc : SUCCESS in 29m 33s","accounts_in_message":[],"_revision_number":4},{"id":"266e24931eff2454795ed620c59afd2bcc086f0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-02 04:55:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"58d169709b7dbe84d7da28a8bb70e91308b5b75b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-02 05:00:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e05ed2815675e414dda2de07dbe714154c67f0d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 05:51:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a5125f3ae88942498bd072df87261eb5 : SUCCESS in 2m 08s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b320e1e7cccc4804bd06cf42785a90f2 : SUCCESS in 37m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0c9fa1c79529408396c077d136056ef0 : SUCCESS in 35m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4c996f8350024e06b374929819229954 : SUCCESS in 50m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3c31d567b84f4629b214e2cb9686ea18 : SUCCESS in 32m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ae41ae31313d48789db95ce4b61f3e70 : SUCCESS in 32m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b59f7e9ac49a4872ac49a66be9ddc7ed : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ea74c4ce5624a5dad820511ab487588 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d92587c049774239b189d82b8ec14f02 : FAILURE in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8741b126fc54794922ca9f6c4d715cd : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a0a84f9aeea4f76b46a8704e501d278 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fca35fc4f73441bbaccbbec3a871b8e : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1bce63f17944689cfc3739d87f278d : SUCCESS in 6m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2811edbdfbc643efb995403b104945a3 : SUCCESS in 26m 36s","accounts_in_message":[],"_revision_number":6},{"id":"8f34efc6c497a856d15aa7c8e543f85005159e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-02 07:09:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"064c0e575e7472a1207c14699dab6e51427c06f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-02 07:47:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0a4c3780bd4086defd90678215eef142933419b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 08:39:00.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7bef05b53054b8bb61585803c39c946 : SUCCESS in 3m 50s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2b8d064b7c1e452681a91d1757de008f : SUCCESS in 30m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/309c87c64abb4438919b72db250e8f3c : SUCCESS in 33m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b26adb0429904f419d4ee1930f7e5cdd : SUCCESS in 48m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a64ef56941ad4504ad8f248390fa4a6d : SUCCESS in 34m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4215a08a0c104d108c74b9fe4ac8541b : SUCCESS in 29m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b240b0c2ad464e91b43b9fff5790aa85 : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/760d18f99d374f06829aa570d6a0b488 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a891456e7074aa5b07b69022a8111c9 : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e7a30009d8b41f685ebfa0123d9cdcc : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e49af44457b48d890fa695333748592 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bc4bab57a5144d7820f126f77f168de : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79251da7346c4f32abe65c3552c17c51 : SUCCESS in 8m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42f33747fca1499381bec8d346405a0e : SUCCESS in 41m 52s","accounts_in_message":[],"_revision_number":8},{"id":"4ff5e8cc965cbcd55946cc4642e3a9abf392c489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-02 16:13:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6275ba9e4dcaa090f8bd88a1737d88a1e9ee6710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 17:36:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9001f86dc02b47d9b1ef7931f2833db7 : SUCCESS in 3m 28s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/062503cce50e4dbf808bc4fb349ec8f8 : SUCCESS in 41m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6839027ae5274fed94081a3eafa5c619 : SUCCESS in 46m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b9ec3c5b125f4a7f8af0674e33919a12 : SUCCESS in 1h 14m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/856c041f804f4b7b83f1836d014a612e : SUCCESS in 48m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3c260e7275094ac28d2523f2caee7cf4 : SUCCESS in 37m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4287d27270924d68be471fc803aa041e : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aebf2c958ccc4efca4777b7cde7a4e3d : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aef6bdf09eae4d4a8a58212b0845021e : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52d4769d8eb44c2e82c22495a38d6533 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e424833b03bf438a82d2ca16b06d2a82 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45dc5f0c74c64652a50ee82f4f164400 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293ee4e2dbf743bab18e19b3d44f0cea : SUCCESS in 10m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8f24e863842f4b63a03545ccf5af4106 : SUCCESS in 46m 24s","accounts_in_message":[],"_revision_number":9},{"id":"dfa93933fb1ff876fd04d1ff8c65714e494fca29","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2021-03-03 10:13:08.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)\n\nPlease see comments below. Please also add a sample for pools.yaml configuration for NS1.","accounts_in_message":[],"_revision_number":9},{"id":"822af49199f6e54c04f67a42423d3e0830dcfd22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-05 16:04:51.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f0e43a2c7ef7af6209aeae736ef4c351a7d84053","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-05 16:06:42.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"e0b4fa2f968721db4cc20373f50e6ee79b355f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 17:22:03.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/adb5691182d847aca3d1b6d7d032b1fc : SUCCESS in 3m 23s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dd31d7be1b474e30acc3332992e2ad53 : SUCCESS in 1h 00m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e6cac8ea49bb4a749bae3b65c2091540 : SUCCESS in 39m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9acf07b1dd1149499daab770499ae4d0 : FAILURE in 17m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/81c8a2b958e54573afcd9842d34eee52 : SUCCESS in 42m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5cba0e5b80ee4960be44846774d99e5a : SUCCESS in 41m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ff90457bbf84f5188aae04dee093c94 : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5138a62499b42d880d1116c4bd13a4c : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98f2dfc195874f13827f9e44944dee57 : FAILURE in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e5da4d9b92546eaa84089c18e1f9380 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2fef008b9594f35a7edf79503892960 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a0b2e01a0944178b7387cc3fb4a590d : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec75fbd0fc58448ab22b7a5b26354302 : SUCCESS in 8m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5f0319aa0232456d8b2aaa2c7821a7c3 : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":10},{"id":"ba483b84411676c76f9dcab571a00672a2bdd4eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-05 22:41:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"78b5c245000c08f54566aadcc33cca0731816f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 23:42:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed5b2217e7a246e49f26efaa9f9fd150 : SUCCESS in 4m 46s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/00d2e88cc3174beba60b1e464ec333c6 : SUCCESS in 42m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/280f60173835486cbf98ea5599151c9e : SUCCESS in 41m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9debd47e57c74fb6b72e4e9a7d41778c : SUCCESS in 59m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/194bd8ba270043b3889824c6d634dc72 : SUCCESS in 40m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/90752f639be1457eb3d1ec78ca2181db : SUCCESS in 43m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9cbc922ccbf4e15adb1e74e70bd5728 : SUCCESS in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d86a696131f4557a7780db498d18e8c : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e930efc08bf048dda14b801f5608fb67 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8c6871d5f32450fb5812e5e065acfde : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04dfa8b27c4b459692b1ee7907bcc26f : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d20f0e18c3bd469cbd2a54d5b4954de5 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20ee0382ea742f18906e62241be4fbf : SUCCESS in 9m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57ec063cb25d4f808a0e7e976b533d69 : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":11},{"id":"f67a187329096cee091e45666da523d1ceb06b54","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-06 04:46:40.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)\n\nLGTM\n\nNit: I wouldn\u0027t mind seeing closer to 100% test coverage, but it isn\u0027t a deal breaker.\nhttps://c57b1905c4579ad0d19e-7bcd0c6ba9078d6047b64bbe393931c0.ssl.cf5.rackcdn.com/778117/11/check/openstack-tox-cover/d9cbc92/cover/designate_backend_impl_ns1_py.html","accounts_in_message":[],"_revision_number":11},{"id":"e4082fbd789858d44a8229838c6ad228a5e93acc","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-03-08 16:29:12.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)\n\nCould you address the formatting comments from Erik and myself? Otherwise this looks great!","accounts_in_message":[],"_revision_number":11},{"id":"3fd8916c1655afb2884ac618ffc0cb19664bdb8a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-08 18:04:09.000000000","message":"Patch Set 11: -Code-Review","accounts_in_message":[],"_revision_number":11},{"id":"fca46bf15e8f877cb4438799f1887a4f01aad993","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-08 18:55:27.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"cb1fae7d1d73bf4ddcf026992dec80694a4c2913","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-08 18:56:38.000000000","message":"Patch Set 12:\n\n\u003e Patch Set 11: Code-Review-1\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e Could you address the formatting comments from Erik and myself? Otherwise this looks great!\n\nDone, thanks!","accounts_in_message":[],"_revision_number":12},{"id":"2ae291e28ce4b4c7da6dd23cc75f3e05cb4397e3","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-08 18:57:26.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"e417408a07dad3e68259b2ea9e9c27c38f5416cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 20:02:31.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3fcafcb9d18945de8f3df2f64f998a3b : SUCCESS in 2m 44s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ca71c38a10cb49938912273ec6cf428d : SUCCESS in 47m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8ce78a145fe64a0f9b2779ddf06acb18 : SUCCESS in 34m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/034767e1013e42efb41f11bbe3935c7b : SUCCESS in 1h 00m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6ce4fffdf60743f9b00b48361c3c7d72 : SUCCESS in 32m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/33091c8a1e9b4cf398614e5888899b3e : SUCCESS in 30m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75ca3d80a8434715af2426aae8d86b9b : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4737a0b21724ed2a7eb1e6d9b7736cb : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0fd4b8d9d654b0e816073a45123c5c5 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3604d1a492444e648013ef8181ceb5c1 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/755507dd15324a7db985e61b75219f11 : FAILURE in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0801eb7894e84b8385dc559f4b9c1abe : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6e5fb76a1f4d5782fba39ae2aecfef : SUCCESS in 9m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/32e85a68f0904898aeff27c139c409e6 : SUCCESS in 26m 25s","accounts_in_message":[],"_revision_number":12},{"id":"dd22517b2467c26e5f3563b4b89755caf599ecda","author":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"date":"2021-03-08 20:13:30.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"b4f345b10ee73d3281b2f5079ec99969a879e10a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-08 20:48:00.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"bd354efab937ee35f9f092c72ba469d9d8e93ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 21:07:46.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e7b0bcfc30764fcb8dd68dc3bb82f337 : SUCCESS in 2m 57s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f88b6a008bea41399686182574f82275 : SUCCESS in 36m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ecb7377183344051acd44c4178a5183b : SUCCESS in 42m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f41bdba1d3e5461f9cfb015e94c1989b : SUCCESS in 53m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c5844946ec8d4757b226f288de3243cb : SUCCESS in 42m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cd0d27464b994360b59bed033c79f87a : SUCCESS in 37m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/464cc836b7424bf9b2eed7c48835d3d6 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8c8aaa0d21647a4bcd26a922499e89f : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af010a87bdc4a72a5beb1425a0aa0ab : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4309620532844019ead089a9242f66e : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/021cbaf905aa456aa8c3af8960126c15 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41a0895a61bc401b805479d7b3dcbca9 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4751cf735c468d93c8d3a7ab074a18 : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/743e0fa5200445b19d060c4d3807963b : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":12},{"id":"23f58622711724f2050a6cf0abe31265c6ac01cb","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-03-08 21:30:02.000000000","message":"Patch Set 12: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":12},{"id":"1893a51e39ac28ec2689d5679a59d1f65e6542bf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-08 21:53:37.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"79f2d2defd5182aa9f64e5d6ec775d7854c296fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 21:54:17.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"32e314a78047b1568535c7b1445fb5e74e056aa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:47:45.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/48d65a0243424acfad929eaaa52a664d : SUCCESS in 2m 39s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/286f30e462334200b40ca2ab2b0a131d : SUCCESS in 36m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/503bad568ae247709f1f10e9e585d34a : SUCCESS in 34m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/01e4c6cbc01547a8905d317d48b3762f : SUCCESS in 52m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f6430556afdc469d9ac16761c3f7ac53 : SUCCESS in 42m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b3227370e5cd426ebac8990c55e43f86 : SUCCESS in 40m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed8d68ca329c467b8271a8f5e328bdbe : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/179886a5233c4418a19ab8c8fbaf8364 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42c3fcd8614942c39b914b52e800f0d7 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1afd48a7e0c4c36984f6791a4db4f0c : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf6f3d44b4e46c190f1e22b24b7aa6f : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":12},{"id":"b181de1489bce197331aa77d10774d2186c641d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:47:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"7d07b6ad3a31e2bed1f92a6385929db898e1f16b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:49:19.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d3672a9c7554ae1852058526c85f35c : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"5aac48f08b441cbde8c2078a9ace0551d3d19058","revisions":{"e03bb9906bee279ac9b9340b412dd136821c2301":{"kind":"REWORK","_number":1,"created":"2021-03-02 00:31:54.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/1"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-01 22:45:52.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-01 23:35:42.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend for Designate.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e03bb9906bee279ac9b9340b412dd136821c2301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e03bb9906bee279ac9b9340b412dd136821c2301"}]},"branch":"refs/heads/master"},"7ca0afbc25074a3d24f963a0cf3f9e05d615237f":{"kind":"REWORK","_number":2,"created":"2021-03-02 02:21:21.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/2"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-01 22:45:52.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-02 02:21:07.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend for Designate.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ca0afbc25074a3d24f963a0cf3f9e05d615237f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ca0afbc25074a3d24f963a0cf3f9e05d615237f"}]},"branch":"refs/heads/master"},"738bb90edf7609e864eeea02adf809eecd9ea93d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-02 03:42:01.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/3"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-01 22:45:52.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-02 03:41:47.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/738bb90edf7609e864eeea02adf809eecd9ea93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/738bb90edf7609e864eeea02adf809eecd9ea93d"}]},"branch":"refs/heads/master"},"a8141fb683a766db590db1998caf0025d1f35dca":{"kind":"REWORK","_number":4,"created":"2021-03-02 03:43:45.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/4"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-01 22:45:52.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-02 03:43:40.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a8141fb683a766db590db1998caf0025d1f35dca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a8141fb683a766db590db1998caf0025d1f35dca"}]},"branch":"refs/heads/master"},"7b88a7cfe41ead73acd7c725607ec653c70b32ab":{"kind":"REWORK","_number":5,"created":"2021-03-02 04:55:42.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/5"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-01 22:45:52.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-02 04:54:26.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b88a7cfe41ead73acd7c725607ec653c70b32ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b88a7cfe41ead73acd7c725607ec653c70b32ab"}]},"branch":"refs/heads/master"},"d736dd1609421d422119e2c4a79e1932ad46cb76":{"kind":"REWORK","_number":6,"created":"2021-03-02 05:00:26.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/6"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-01 22:45:52.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-02 05:00:06.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d736dd1609421d422119e2c4a79e1932ad46cb76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d736dd1609421d422119e2c4a79e1932ad46cb76"}]},"branch":"refs/heads/master"},"7caa77c713968449cb1def51a3f936ea87e84a97":{"kind":"REWORK","_number":7,"created":"2021-03-02 07:09:49.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/7"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-01 22:45:52.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-02 07:09:39.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7caa77c713968449cb1def51a3f936ea87e84a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7caa77c713968449cb1def51a3f936ea87e84a97"}]},"branch":"refs/heads/master"},"f1a4490c8fd1756df33781d40782f18f9a3f1919":{"kind":"REWORK","_number":8,"created":"2021-03-02 07:47:45.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/8"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-01 22:45:52.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-02 07:47:29.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f1a4490c8fd1756df33781d40782f18f9a3f1919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f1a4490c8fd1756df33781d40782f18f9a3f1919"}]},"branch":"refs/heads/master"},"a0bdbde061ad7d0a8ec0829df6b1610733da29d5":{"kind":"REWORK","_number":9,"created":"2021-03-02 16:13:30.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/9"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-01 22:45:52.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-02 16:13:22.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0bdbde061ad7d0a8ec0829df6b1610733da29d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0bdbde061ad7d0a8ec0829df6b1610733da29d5"}]},"branch":"refs/heads/master"},"d2e9fcb70c2d501adc4c151d1845482154605a2d":{"kind":"REWORK","_number":10,"created":"2021-03-05 16:04:51.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/10"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-05 16:04:29.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-05 16:04:29.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2e9fcb70c2d501adc4c151d1845482154605a2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2e9fcb70c2d501adc4c151d1845482154605a2d"}]},"branch":"refs/heads/master"},"3423186f49029162121fdd54402476492d4d4e1b":{"kind":"REWORK","_number":11,"created":"2021-03-05 22:41:46.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/11"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-05 16:04:29.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-05 22:41:40.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3423186f49029162121fdd54402476492d4d4e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3423186f49029162121fdd54402476492d4d4e1b"}]},"branch":"refs/heads/master"},"5aac48f08b441cbde8c2078a9ace0551d3d19058":{"kind":"REWORK","_number":12,"created":"2021-03-08 18:55:27.000000000","uploader":{"_account_id":33166,"name":"Michael Hood","email":"mhood@ns1.com","username":"mhood-ns1"},"ref":"refs/changes/17/778117/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/778117/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/778117/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/778117/12"}}},"commit":{"parents":[{"commit":"75668d084c2a27fcc25c67741ca309bfe43b1322","subject":"New Doc intro section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75668d084c2a27fcc25c67741ca309bfe43b1322"}]}],"author":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-05 16:04:29.000000000","tz":-480},"committer":{"name":"Michael Hood","email":"mhood@ns1.com","date":"2021-03-08 18:55:06.000000000","tz":-480},"subject":"Add NS1 backend","message":"Add NS1 backend\n\nIntroduce an NS1 backend.\n\nSigned-off-by: Michael Hood \u003cmhood@ns1.com\u003e\nChange-Id: I80fe08238005a94161e2dbcc89e77c90cde0a715\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5aac48f08b441cbde8c2078a9ace0551d3d19058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5aac48f08b441cbde8c2078a9ace0551d3d19058"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
