)]}'
{"id":"openstack%2Fdesignate~786506","triplet_id":"openstack%2Fdesignate~master~I638626562038dd29e9a10510f07c4dcb1bae6224","project":"openstack/designate","branch":"master","topic":"dnm-uwsgi-debug","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-11-05 19:05:04.000000000","reason":"Michael Johnson replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2021-11-16 22:22:43.000000000","reason":"Change was abandoned"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2021-11-16 22:22:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I638626562038dd29e9a10510f07c4dcb1bae6224","subject":"DNM: Debug uWSGI","status":"ABANDONED","created":"2021-04-15 21:05:08.000000000","updated":"2021-11-16 22:22:43.000000000","total_comment_count":33,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e69af5983d2934eaee8ff9ea5de78c99d281480f","_number":786506,"virtual_id_number":786506,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-12 22:36:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-15 22:04:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-06 07:37:06.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"25da279ef0ad787053eb4a6d94056568e05e9d7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-15 21:05:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"981bdebe7fe57c3bf631ea8d6de7e7a67f6ed928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-15 21:25:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"894deba492a799ccf4e31729ef92ae0d6f6cd76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 22:04:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0bcb9cfcd0d44513a5f5fc5ae8bfe860 : SUCCESS in 35m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/51d0066e81e94400bb42950e2dd88468 : FAILURE in 35m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3cfd05831a024401a1929fd70cb2d7f9 : SUCCESS in 36m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ca266054f4824d328bffaf35fe529e9d : FAILURE in 35m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6aeb54d76c40cb708a4ded258bb89ab186c8fd4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-15 22:13:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3718b2a1caf8599e4ab327e6743a5aa53635f110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 22:56:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/554d449fb64144ffac17cd6467bc37b1 : SUCCESS in 41m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fad2437443434a9c8d6757283655ce30 : SUCCESS in 36m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0018a3136d544ff198882e51d8c316ff : FAILURE in 35m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9d5acd1d38404fe7abb058a55ab7cd75 : SUCCESS in 33m 55s","accounts_in_message":[],"_revision_number":3},{"id":"4b0717077ba70afb9e1ee29ce9cc7fce9da012fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-15 23:09:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cd6aed7c0737e864f5081b8e7b37d76ceb2f5b25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 23:59:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c18d4e32c8864674b61daf163fb8a613 : SUCCESS in 39m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8300f3c89270475c83e19962364c6b2a : FAILURE in 48m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3d44c87fd71349a987c7ca7de4621ec6 : SUCCESS in 38m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c94940d2ddea487f965732cb726bd531 : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":4},{"id":"8bb8a6318ae01c55081d87f5292a156cbb2673b7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-16 00:02:02.000000000","message":"Patch Set 4:\n\nrecheck\nwrong failure","accounts_in_message":[],"_revision_number":4},{"id":"3b997a2dfc33b9b0114faf7ec5623eb78aa498cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 00:46:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/beefd96065204a4589881f79c7e99224 : SUCCESS in 33m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d1f3b2f9f54c425fa950c88402a1e61b : FAILURE in 25m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/170a57ef87144d87bac2d7761310fe50 : FAILURE in 33m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f0a31424d9a048188d23864ba7bb8105 : SUCCESS in 42m 34s","accounts_in_message":[],"_revision_number":4},{"id":"5825d410f850101b3f5552fae566b6988ab8ec7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-16 20:45:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7e0f7194d944d4856fbfdbb006ce02bcdb3b8b07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-16 21:12:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a3610c2999214b8940de114d7d335a86bae50a2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-16 21:33:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c4101606dcd7ca4a075775a6792587bac0beb0f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 21:54:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0a3611297409440e90ce2b8da78315e1 : FAILURE in 13m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4cabfa83aa584721823eb4fa467e974c : FAILURE in 12m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/59366e4149194a9784a89c9c21469ed2 : FAILURE in 12m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/49bcd2da49704d6b8390f17ed826ae3f : FAILURE in 12m 46s","accounts_in_message":[],"_revision_number":7},{"id":"b3cf2fb69cc8fa587e74fdd7cbb33f7651900dce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-16 21:57:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c247f20fb8ccf9cc6b1e872599089006e6841a97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 22:18:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/49cd209973d441f7907d7be2e2b85872 : FAILURE in 11m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a687da93193d4c21831467a2bb52af19 : FAILURE in 11m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/57fca74f213e4fe7950767e1b9fa1a04 : FAILURE in 10m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4dc806c1b41345329d06262f2d6d3557 : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":8},{"id":"d023b5c250ad781a88cdbcd2afa8ada9fe26268b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-16 22:25:25.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"36a34289f7f9faa9621afb9c6fc47b70e6e53815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 23:00:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8e3b73cd42124370add44ef589ab65eb : FAILURE in 32m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7bd33f5b39dc40afb990707d1cb88311 : FAILURE in 31m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/242fa1a3b3654a8b815f8343498e175e : SUCCESS in 33m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ace0475470ee4e4aaa6b59c66958b2fe : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":9},{"id":"9a0d18cf44804db0a871b35899e54504c04b398c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-16 23:24:34.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cbdda824a7d6b4b2c49b7f262000fc8956b27125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 00:15:45.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/58db3bc540194c73b6b71ad8a6704dae : SUCCESS in 43m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4c745bc0d2ab46b4882862a44821aeac : FAILURE in 37m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/25148777e0854051a8d39efd76c44968 : SUCCESS in 44m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f69052e5bf594a33b8396c0d5dc1316d : POST_FAILURE in 39m 27s","accounts_in_message":[],"_revision_number":10},{"id":"0296917410e63fbb33e9913a7a689505666a60e5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-17 00:15:56.000000000","message":"Patch Set 10:\n\nrecheck\nNot the failure you are looking for","accounts_in_message":[],"_revision_number":10},{"id":"13761563e82e3b88d819a53638f3e858faf849ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 00:51:29.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fcfa9ff05f364bab90a8e92d295f7746 : SUCCESS in 32m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fb9d6f4d4fac49dda2b361b73867cb7b : SUCCESS in 34m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e06f4f79eac24485834bf760a95f924c : SUCCESS in 32m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f061bbdf4cdd455fbbcda24670ee1f7b : SUCCESS in 30m 48s","accounts_in_message":[],"_revision_number":10},{"id":"d0cb88f3aeb9058ea9cb5e576430472149f2f949","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-17 01:25:50.000000000","message":"Patch Set 10:\n\nrecheck\nIt knows I\u0027m coming for the bug....","accounts_in_message":[],"_revision_number":10},{"id":"1bd14fec7aa2c47b640b502e0afc072b92581f64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 02:05:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c8149541f89a434f80a5bb86e1f7987b : SUCCESS in 31m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f95279b4c8e441daa5b6a981b091bd98 : SUCCESS in 38m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/35d1747cf4df4498943b00dc9684ed15 : SUCCESS in 34m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/364768b5c3e047218c9cb6df7d8237b0 : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":10},{"id":"4e303388b3e54a8ae9ef0fe8f78ad19f56fe5cb3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-17 02:39:10.000000000","message":"Patch Set 10:\n\nrecheck\nStill hiding","accounts_in_message":[],"_revision_number":10},{"id":"b7f4e57b65128031c0ae1438424b456bc3807dc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 03:17:22.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/936c055d6fd74e268ef09595c9fcd90d : SUCCESS in 36m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/276bededf845452bb9c3b6d4d8855b35 : SUCCESS in 30m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/86fb2b55a6c742b2a6730a3e0ec22cce : FAILURE in 33m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d5c488f792884e808ea80ccb1a2591ca : SUCCESS in 36m 49s","accounts_in_message":[],"_revision_number":10},{"id":"37533d169457c4de4090c93474dd864ad5f11945","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-10 17:51:33.000000000","message":"Patch Set 10:\n\nrecheck\nmore uwsgi debug","accounts_in_message":[],"_revision_number":10},{"id":"dd370cb58be1127020e146baa72290fdd89b6c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 18:45:46.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/012aaaf4112f47269b8fd2f549f133a9 : FAILURE in 40m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9b646e55ff9d4c4fa55da6dad5f2ab2a : SUCCESS in 41m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/725c8f23ebc34dcab8338979899c05ae : SUCCESS in 41m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f6054680b0d34be18c901cffe863144c : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":10},{"id":"30794b0c1daf169c0ba6135cf02809154f081171","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-11 05:00:59.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"6eb6796bb6479aeeabda943bf148a85287ab9f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 05:33:43.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4f1e939d1d294dcdaf223d130202aaff : RETRY_LIMIT in 10m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2e0aede49a70462dadbfda9020218cb5 : RETRY_LIMIT in 10m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/88eca507412f4ebaa7c68b59fea5795c : RETRY_LIMIT in 11m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7e03b8862c1e423a9fad3d1360eee38d : RETRY_LIMIT in 6m 51s","accounts_in_message":[],"_revision_number":10},{"id":"a7746cfb63807e8abb906451f53727742951181f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-11 17:44:06.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"83216e4704c27ca17e0da17c7ea7dfa6691bb5e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 18:19:51.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/49d0ba3dc17544298ad45ac046f84a06 : FAILURE in 33m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9c75a61a1c484d58b84031669be8d94a : SUCCESS in 34m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/db669d9db7ac47df9a2106ec14994257 : FAILURE in 27m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b2030f74f4c24c1e987cfcf9852cc77c : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":10},{"id":"6639e30c10fb9c22ab553511c6e7b45509461c79","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-11 20:19:02.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"e76154fb673e967948a9784cf8afd52f347cbac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 20:53:27.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3fc4b0677e5d434e9a2d4239c3ef007f : RETRY_LIMIT in 7m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1850c3c53ac4478f8b34db1d86afdd86 : RETRY_LIMIT in 11m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b87462fd35434b1588c115feafe82b21 : RETRY_LIMIT in 11m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eb983c4652244be698b7016c2d6a3d60 : RETRY_LIMIT in 13m 26s","accounts_in_message":[],"_revision_number":10},{"id":"a07e836ec3921c370200d7546f38437a69954954","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-11 22:09:01.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"bc2953a7e6db766bd352802009a9e443ba047adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 23:01:22.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/41e0ee6b3de44e588d789e49e01462c9 : FAILURE in 49m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6528636e25e34bdfbdf8e800e3415a46 : FAILURE in 41m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/425689c919c141abaaac44f01aa69ec2 : SUCCESS in 25m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/59904ac47ad14fc19d44ff407a4553d9 : SUCCESS in 28m 35s","accounts_in_message":[],"_revision_number":10},{"id":"e029f4f965d301ba6a43b2358b9b92a8c8b4668f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-02 03:50:57.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"74cd40f399bf5402c04affbf780da2d6d89fd99e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-02 04:12:06.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f0d86b9e69c7d63271126097fc31dd6438515b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 04:33:26.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8a081b18b984e08bc793c40f355fa93 : FAILURE in 19m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/97438023fbb54e0c80093f82d6f56fbb : FAILURE in 18m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7c0fd833ccf34d22885c1aa7522c536e : FAILURE in 18m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b123fd2e977c45af81846763f5710363 : FAILURE in 18m 57s","accounts_in_message":[],"_revision_number":12},{"id":"924be4b87fc495ac027245565136ddbff241ce07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-02 04:42:29.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"9f29aa3ba5280c164924e2c622cb9ea40be730ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 05:29:08.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cc4c45ab4e2141c0b11aef95e89a2ce0 : FAILURE in 32m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6a65b1e7a7694f2a925339a3bef64c8b : FAILURE in 40m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6ca07665ef994f72a4b6112c1c3b7e22 : FAILURE in 31m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4fe1fc3bffc84952996e138f90c28393 : FAILURE in 31m 44s","accounts_in_message":[],"_revision_number":13},{"id":"b2e771091dd82f263b0bad9009a05b3ac2b3c77a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-02 05:49:53.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"21cd10b31fbaec103c24522383cd56e597456054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 06:18:45.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a5a55365326049019b14076b4e18bdee : FAILURE in 23m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a67e287c9bc34997a4e274b05692f0ef : FAILURE in 27m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9433920b4d264e64847bd59bd8ade764 : FAILURE in 24m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/650ecda7d54c4997afb536101b060483 : FAILURE in 22m 43s","accounts_in_message":[],"_revision_number":14},{"id":"01950bc4d1b2b1f40f557598bb43837e4439b008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-02 07:36:27.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"8f05550856815ef0d6447aaf7964d0875da59372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 08:08:47.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ca5bd7be500749e6a1b62bf1261afc76 : FAILURE in 30m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bbea33352fcb482baddcc9f94e49fbc9 : FAILURE in 23m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2c08acd2763247b8aa20148c43e9ae90 : FAILURE in 27m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7449bdc3d65c4ae7b18022ed1ae11d70 : FAILURE in 22m 06s","accounts_in_message":[],"_revision_number":15},{"id":"6264cf5fefdbff1356520187a843469e1a1bd1ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-02 16:40:22.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"552a3ac537ac7ebe5a1908803b5d68f4c5232653","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-02 17:07:52.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"9db2ad9e0b95305766529086ee6812380963c22c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 18:00:08.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a82390e639224f59867344c1d7f79101 : RETRY_LIMIT in 8m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8880007da45945ce90040d73dae2a256 : RETRY_LIMIT in 9m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ec583992078a4bc4919799e794191fba : RETRY_LIMIT in 9m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2cc9f88ddaff43499037ccadf8693894 : RETRY_LIMIT in 10m 23s","accounts_in_message":[],"_revision_number":17},{"id":"468244e971063fe66fdb47305685dffc163740ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-02 18:33:49.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"103786cf2e9fde9ca09fbeb1e4121704f30af7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 21:06:10.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/045a582a1acb45a181f6e1f74f15b84f : FAILURE in 48m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/217c1d50f52545278b4050e281c0961f : FAILURE in 52m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d2ef5cbc6b4842359c9b4c7d5a1fad32 : FAILURE in 32m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0cc40282f3624b1690c693a7c60bb3b1 : FAILURE in 34m 39s","accounts_in_message":[],"_revision_number":18},{"id":"8070a546d6ccababeb1247d3e9002608c130bd12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-02 21:54:33.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"2a98a786e1cfb902a1b233090b9474e91291197c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-02 22:33:07.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"4f89bd14e13249260a97317a26cd167b282a7471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 23:06:55.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/832fdcbcd4544d28ae1a60e3d7b839dc : FAILURE in 31m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d497659702004e26802659ecd422111b : FAILURE in 28m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b5e77658114b45fd9ca668f25a4ff4c4 : FAILURE in 29m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/35d1bfc29a414012a5662ef5bb33eb51 : FAILURE in 29m 47s","accounts_in_message":[],"_revision_number":20},{"id":"51706247b1bbcf799c49fb81c2808c81e28d2ea4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-02 23:17:24.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"e8a57ba42c2b07e2e2a9f2b7ee6eaecf41119bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 00:02:54.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f39f42995ac8410c815ed29f53615f5f : POST_FAILURE in 27m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4a9a717d133e4a0cb8ae8d8cb636a194 : POST_FAILURE in 39m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5fd98300fbdb4cb2a99690a958357508 : POST_FAILURE in 25m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e67a2a1a92d5486c886e0a12d588807e : POST_FAILURE in 27m 33s","accounts_in_message":[],"_revision_number":21},{"id":"ab5ae5b3dcc4b264a3672898f1704938b85c212f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-03 18:43:45.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"0bf2d3267f857ec7ab5bdfc84bd40c97d9e67f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 19:20:02.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cbbd59a8833e4cfd8c57ae1f5cb1791d : POST_FAILURE in 28m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/232175b1f968452ab9b030983b2a5062 : POST_FAILURE in 30m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/23e36b9487354f7da7ec4ad347486a52 : POST_FAILURE in 34m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8b3e6bdba087431bbf71f1092bb8ee2d : POST_FAILURE in 29m 06s","accounts_in_message":[],"_revision_number":22},{"id":"4a5dc47b951309a95592c127bfa2d200624aa4cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-03 21:50:10.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"87d59d177f9a63727963d3cfbca2b918121b7b1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 22:22:29.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f4765119703a4981b0fb90087f146483 : POST_FAILURE in 21m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4bd57e3041f245a9b94d69b248dc4925 : POST_FAILURE in 22m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4434863e4ff1455e9e3b274b26c6ac12 : POST_FAILURE in 29m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9d5b04b948124ee5be35cfae46f3562c : POST_FAILURE in 24m 11s","accounts_in_message":[],"_revision_number":23},{"id":"8a0143461eb010a1f3f8805c6d5632b2ab393f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-03 22:47:36.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"e59d17b524d8f15dbe894ce12d576131d858b6c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 23:11:11.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b47c55edc2f74b92ac6074b58f1d57d3 : FAILURE in 20m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4e0a20408c6a48bdbe08ba3b07da667f : FAILURE in 21m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c6cf753632a34b9ba3b5366d7347c983 : FAILURE in 19m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c6b5a4355c0e435f9d0f3a3c25dd9f9a : FAILURE in 21m 42s","accounts_in_message":[],"_revision_number":24},{"id":"01e50df4bd8c4c4aeb3ba5e36c01dd4ea5f1af51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-04 20:07:39.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"1fd8da3fd7be6d28084ad5cb5eec93ac109d46d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 20:44:13.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c4cdda3dd6e045ba9f793c679471e410 : FAILURE in 31m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5799816a5aa24f4d9046e2f93a95bdfb : FAILURE in 24m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0f06b24afc03439a89e9a33a8d91bef6 : FAILURE in 28m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4116eee2e07b4b83b1c45be9a3b7abcb : FAILURE in 27m 07s","accounts_in_message":[],"_revision_number":25},{"id":"ca9aa1f8a1419794c191fcc1ff644579154f4eea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-04 21:03:36.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"946e70369e667470f931dfafaf131c00359e9dee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-04 21:45:27.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"e60cd4bf9118f7d2e3ff77d8361debc091763f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 21:45:35.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4465def57aa5482793461264a3411a2e : FAILURE in 27m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fa183d6935c34e1e8e274e65c1321227 : FAILURE in 36m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3270ade7aec1413d840d5fae8e01960b : FAILURE in 34m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/70b3d249c8154b12b5c4fd85b883352e : FAILURE in 29m 02s","accounts_in_message":[],"_revision_number":26},{"id":"3831084e731aa88e5568000f52ac4904ed1cb035","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-04 22:09:40.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"6119c8a81c26901c32ad3e7e50f862e329131d9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 22:56:37.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f43dae11a4894c3f8f9a5dcc0ef7fb32 : FAILURE in 38m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f2971de3796640738923d7910a401d90 : FAILURE in 38m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d859229baa5a49f29943f431c961c613 : FAILURE in 32m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5dcd4e27db034e86b21f8a421c0c4075 : FAILURE in 36m 52s","accounts_in_message":[],"_revision_number":28},{"id":"0dd7d8415cc858b96f729bf85d66a8a763128ed7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-04 23:17:07.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"8149adc867112d537b76183f9a0665d4405c1f39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 23:50:11.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7295e5a794d945559d8a900cc0c3d909 : FAILURE in 22m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4125ded3edc540549fdc773e33928ed9 : FAILURE in 30m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/99c7a1e230474889b4a29e9da325b5e5 : FAILURE in 26m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2625093f065c46a88552c7311e2a059f : FAILURE in 25m 52s","accounts_in_message":[],"_revision_number":29},{"id":"eada02fab63032ae2597ccb549b9437cc878cceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-04 23:54:06.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"960da938cf4dc333765d1eeaeee15b9ccd2e71c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 00:24:48.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4df649d8d562490483faa419ab4a0ee1 : FAILURE in 25m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d4821f2d7cae4cf18345b3419c959f1c : FAILURE in 24m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c1a7e77bbe044e9f881a76d958b8daa1 : FAILURE in 26m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eb85dd9f1db543428a1c6993677eb051 : FAILURE in 26m 21s","accounts_in_message":[],"_revision_number":30},{"id":"a21e3d53dbc78c05446362d02cd83d72188ffd78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-05 00:33:11.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"1e5727cda9307a1faa5ea47e51934f5fb5c177cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 01:06:59.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dcddc112aa0a4bacaff6a6807ec49422 : FAILURE in 26m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/97abcd3d31a24a0bbe267f4b12cb8d90 : FAILURE in 25m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a44af9d5a1be4734aeebde463cff8666 : FAILURE in 31m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e1e3b5bc1754447da6a201e794e72e0e : FAILURE in 28m 20s","accounts_in_message":[],"_revision_number":31},{"id":"b9a0dd217d95d333892fb8aae1739cd21c21ca6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-05 16:19:58.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"63c972b706e153db16f4ee5f8bf68c8e431b4e5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-05 17:00:07.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"cf6562e9437ff9e5361d9f704396b20fe2a9cd4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 17:37:43.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/90617a1ab8404b638f660a4cdca735ee : FAILURE in 28m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/898a2d2ad6d74e1cb63ee317757a811d : FAILURE in 29m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e1943a4d0b3848648303479c9e8fac0b : FAILURE in 25m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/56e60142caaa46a2b195e3004c4498bc : FAILURE in 34m 38s","accounts_in_message":[],"_revision_number":33},{"id":"0280ca7590704cc5aa2b7ce6187d3caab3aeaccb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-05 17:45:53.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"a3ccea743ba258c679d14cd08d547cfcced16de5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-05 18:19:58.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"8796fae2b0f0241e0017bd29796e776e6d447fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 19:01:03.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f4a26f0fbcae45b8b76f2d3597ab379d : SUCCESS in 39m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2c7f137cf2eb46c2ac4ce0967dfba818 : SUCCESS in 39m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4971785bfb9d4963a3944e7664b28492 : SUCCESS in 37m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/871f463b1fa347ceadbd929d7b4b6818 : SUCCESS in 31m 34s","accounts_in_message":[],"_revision_number":35},{"id":"fd899454b88d313e35e69d3a74f9883622c31460","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-05 19:05:04.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"431dc4cb761fce7e36e1719686e5e342e6e79ab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 19:46:58.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d4ea4cb87413463c83b19f6d833c6bb4 : FAILURE in 13m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9614acf3acb54e528457a38bf90d1ad7 : SUCCESS in 38m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4bfdd27ea5a84873ab6f92f81797d6b8 : FAILURE in 38m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0b0cde46a30a4bb2bb77aca61c406cf1 : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":35},{"id":"bb314ae547b23129a8ac5c435cdf5a47ca7acd43","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-05 20:21:24.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"b4d98c851548bbaa1556abd627c3d04051b00c84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 21:12:30.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4eb6dbed3d6841adac5cfe6203197110 : SUCCESS in 44m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2097a0016de1425aa6fdcc53ecd153a4 : SUCCESS in 43m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cbeee8d01c5d4548ab3f57838d33a5ae : FAILURE in 40m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/59f7a4d1382c41408459bbcc4f2cd789 : SUCCESS in 42m 55s","accounts_in_message":[],"_revision_number":35},{"id":"60e75fb4d4bf2d25de3e940f52dd9e662f92d194","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-05 22:19:56.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"7d0e386ee56f938c56e4de8b9cae3268bf5eb2c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 23:17:27.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0702301e92bb4a65b63b2387c93315f2 : SUCCESS in 32m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c572a2a931924751b981b8e1b06ac1a5 : FAILURE in 37m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dc0ca683ac9b435fbc9a0f9561ff4b0c : SUCCESS in 34m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/111714bc953746a48b0ca2aac12752a0 : SUCCESS in 54m 53s","accounts_in_message":[],"_revision_number":35},{"id":"28f762825d9f881d1a245540c661c3018c86eb06","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-05 23:17:52.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"7b5b4686636e5e832470a6ec59b1b5f222a5a373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-06 00:03:15.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fbbc510f00d3458f9eeb32422e31a8ed : SUCCESS in 34m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8265197124f14de0b7aba0e402cb5591 : SUCCESS in 40m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ad56cf1bf66b4e60b80e60840bbd6e2b : FAILURE in 43m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5e923afafec2426397a3ab3784ae3fba : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":35},{"id":"c27ba0fa5477dcf5bb9880463b3cb8f21cead582","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-06 00:04:28.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"e9204573605eacde02c02f2e4cad13820bdf44a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-06 00:50:16.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d956e9af56b14ae3866bebd332a13709 : FAILURE in 40m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fad20a8e4ce24467ba14a2f9b43bd0dd : FAILURE in 19m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aa93c1176bb34f4b97a0a37694ef678e : SUCCESS in 33m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1ce9e6e4a4d44dda8a9273b62cebf745 : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":35},{"id":"c1622b137151ccf32c522fe419ec3f3d672bafbd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-06 07:37:06.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"a45f5e65da47e1ef6c6e925e1030891d1c5edd81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-06 08:28:58.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/767e4e81dd904a31a72d171d67947bf1 : SUCCESS in 38m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c5462a69bb9c4a8f9e54a3bca27ffc6e : SUCCESS in 34m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d1254c59b8bf4bf086fb40a080e05da9 : SUCCESS in 45m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3887ee5c7c9445eb8f07d817c0457f00 : FAILURE in 20m 37s","accounts_in_message":[],"_revision_number":35},{"id":"ccb810a8bf6a01ba99240765049f38f094163941","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-06 08:52:05.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"d7ed20476e35cad7cccec19f4932b84eb039f301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-06 09:32:15.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a6eb3cc32a8346298e5f6c258a9ff761 : SUCCESS in 34m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e261c784e6bc4531858bd79576f391e0 : SUCCESS in 26m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0396a8368a5f448aa1be30015a5df39f : FAILURE in 38m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e384fc7e88c2406ab3f6a923c14f7bb4 : FAILURE in 33m 40s","accounts_in_message":[],"_revision_number":35},{"id":"b4fe830ef62f533b788dcf5c9d17c35b4e5fbc11","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-07 05:19:27.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"eb4aec05d69a0247c0d20f9cff5627420b819c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-07 06:03:15.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/73174fa1b01c49af9553ee4b7ac837ae : SUCCESS in 42m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/606031355d154c878e0515b6102927ff : SUCCESS in 35m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7071fb7c7607493fa4db174722df9f7e : SUCCESS in 34m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/55ecdb9a51f8464cb3123d35ca724522 : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":35},{"id":"32fd2454df0a0449fa2bdc6265691dd9e2ec792b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-07 09:48:51.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"1ed028a6ee834b9114c39a936d4b2a18addece14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-07 10:30:31.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f75d21b8c80843439a1b67615be9b6d2 : SUCCESS in 33m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/609f426cf28a48ff864521e77832ae9b : SUCCESS in 33m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8ee18fc2777a46cdb66d8aa392c792da : SUCCESS in 30m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b5ce2dd89c424cb4bbbe28f131490235 : SUCCESS in 39m 38s","accounts_in_message":[],"_revision_number":35},{"id":"012222e5be738cc9b5befd6cb1be5d0454869e5d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-07 21:43:06.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"601aee75cd2d76b5260ba5b51c284eb3535fbc9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-07 22:27:37.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b51f2c0a1e724a2199cf3bbc7b46ac34 : SUCCESS in 37m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/950bc0336b174bb59d0e315a9c740065 : SUCCESS in 42m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1e82ea75fdf64d32b921a4163fc45e3f : SUCCESS in 38m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/04cf1bb6d36f4cac96fb08a7a606899d : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":35},{"id":"18000ed0b076119dd83f338be099603f207652e5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-08 00:29:01.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"8061050e6d6bf2e1ad0f72a62096d36069e90837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 01:47:57.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1f2cb6944dab46e7b3f8d31f38dae102 : SUCCESS in 1h 15m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/31e24342dac04349a7427bd44d31afd5 : SUCCESS in 36m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/58f04c8e9b2346a188cdfd57fbc5b726 : FAILURE in 57m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d5d80902a5714d38ba53678c03ce7fe0 : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":35},{"id":"f15e4c15146d8a693123769ed7856bdb717dd8c5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-08 02:38:34.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"527c38e49f657e4fe4549ced3e26068d7162f358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 03:37:30.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4062af895cd3428a9d4b64aeec0e7483 : SUCCESS in 43m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/87424eb282b74de1a304631d05324b59 : SUCCESS in 48m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/16585c4c301748d187fd3af8679c39b8 : SUCCESS in 48m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ce673f95ab9242ef8ae1d176dd99fda4 : SUCCESS in 48m 40s","accounts_in_message":[],"_revision_number":35},{"id":"231ba738e9dbf8a49d586b9c6626257298fc4879","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-08 06:38:36.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"44caf88b5e766f02c5ef6593392f2cad92683190","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-08 17:23:04.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"6754a7d7ccec4c4e358c30ad615fe36d497eb0e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 18:51:57.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6e2df7192f4f4407ae2e9c80f851e393 : SUCCESS in 45m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/78eb19f7332845338feb2914b99a6146 : SUCCESS in 45m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a3e5798b6d3e4af9b4a0b65283cd439c : SUCCESS in 39m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a911e21af897485a87cce8091f55a658 : SUCCESS in 47m 13s","accounts_in_message":[],"_revision_number":35},{"id":"37b0d1c6508afd99aacefb4321a31523ea271bb8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-08 18:52:29.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"80a2dc7c53b833ac8abefb085cf5e6f9a6a59f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 19:48:48.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fc110af2795c41faa80a1a02d1d0ff4b : SUCCESS in 37m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bcf1230ffcbc4d13b4c961eda927d9bd : SUCCESS in 45m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2c22ebdfeace4aa6bae86bbf37ec4e49 : SUCCESS in 44m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/40e00e5cac10435e8475d37fe49ea697 : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":35},{"id":"0fd17c3899f9395670fdebb53735989c0b1e2e13","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-08 20:45:32.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"3c47840654c6ebaba694d45a0a3dae2a0e2b8526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:46:25.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c7511f11a1b54e41bb9ca5804a380089 : SUCCESS in 43m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ab9d94af005349edb5fd35832f452db9 : SUCCESS in 52m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/81b07daae5fd4f5b97cd9532a4266a1b : SUCCESS in 45m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7df281f11593484d94cc8b71af7abe46 : SUCCESS in 44m 51s","accounts_in_message":[],"_revision_number":35},{"id":"7802942c50b454358a0ad5ae9aa46670b3a31ca1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-08 21:46:59.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"bb8334a2961df22d33a57dcc2f3a98ad5d9805fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 22:47:52.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/20eb8f4310124bf1acebee12eeab0980 : SUCCESS in 38m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e65fb8473a2746608fed9fdf6b7a0f0c : SUCCESS in 32m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a9db403c9fb543e6a3eb659d8196deef : SUCCESS in 57m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d9fd821bc62c44cb9274e9c8005c95b0 : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":35},{"id":"dc018953c4bd4fa67311227e89f474ac02adc646","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-08 22:58:48.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"abb8ed5b4f7aec93791eb69981342085cc9b5ff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 23:45:40.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4261392f4e5349a4bda8aa57e832dd1c : SUCCESS in 37m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ad2c0c7ad3a1461d9d993f3c0092b944 : SUCCESS in 42m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aa06a3d372a242589a7d12608384fb82 : SUCCESS in 43m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fa1e1f34cd374c9db12e26f4ea3d0486 : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":35},{"id":"dc693a3df67069b7fd31b073e7e5c0ee09ddb8c0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-08 23:46:10.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"1bb167ca619fbe40f101475c51b816dab11ac372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 00:51:25.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6d1a687e920846ef86e0ab8de7e50e44 : SUCCESS in 41m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4abf6c85892b425baad4f59905502df4 : SUCCESS in 47m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8197b6d8ac6947569139040a17f09e6c : SUCCESS in 39m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/df82f77396234e71838b0f81ee0bfcec : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":35},{"id":"f89bd9d3ddb560bc8e2bead0a9c371efec101b5d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-09 00:52:55.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"551fcd5218f983a87363d9196d2b31f979194e9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 01:38:41.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b27c65ae31864a879bfdc607ddfbb4e0 : SUCCESS in 41m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e5985e1a93564d0296c0c3f39dc181cf : SUCCESS in 37m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1d6cb3300bef4367b29959e66de37818 : SUCCESS in 42m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/531d79ecb1924972a82690ada3229499 : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":35},{"id":"5ce2a5bc8d82e52b9aa5c5c4352f8c3978bbaf91","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-09 01:46:04.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"5df496b985962c489061fd86e7b753c6085aec11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-09 01:47:57.000000000","message":"Patch Set 36: Patch Set 35 was rebased","accounts_in_message":[],"_revision_number":36},{"id":"4fabd1ec8b2a7ef9f1fa9b3d9b21cff8628877b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 03:55:00.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b4cdab0b8e554d33b31e63d1f1a825f0 : TIMED_OUT in 1h 16m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/638a1b5323e14292aadce293a42ef64c : FAILURE in 14m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dad0196b149c4444babc094e7a9a7da9 : FAILURE in 13m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/92800fe786bc45d7b3abe8e0536a2ec5 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":36},{"id":"6f93eecde2dcd0fbd664bab63d8694b52baed577","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-09 03:55:18.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"b131bc8f7edb17e53ed32ab921e62f3d9145eebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 05:02:19.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2396cfaf75384eddb0cb3079937053b5 : SUCCESS in 1h 04m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/efac694c3095461dae3aa37299c558e3 : SUCCESS in 43m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/58a8ec44f0514931ace29deb3c4796d4 : SUCCESS in 30m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e547e6421cec4ac88334e9bd2d5465e8 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":36},{"id":"5d1fedc27bf5fb0907986939f8ed872f30b723ff","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-09 05:19:09.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"b1b9f1b896df4cc21ef93cc13e82790e5bfad9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 06:16:34.000000000","message":"Patch Set 36:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ede38cc262984f438581f35a5d2bcbe2 : SUCCESS in 46m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b5866f5384c244e68948feac1a2ff3ca : SUCCESS in 41m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9ef3615633c54ae892c9e5a0de18f906 : SUCCESS in 50m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/abc4b39b3ec94e10926a3a6479753a94 : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":36},{"id":"6920735baf6fae5001ab3d89afe5a50620972b35","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-09 16:54:59.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"a10639bb0a76c152c84adbb52bfec084da429638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 18:29:56.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/933a0a1f005c45649cde1865da6467a1 : SUCCESS in 42m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5ce25dc3b72c48208cf5bde815847456 : SUCCESS in 44m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5b909ed69b984a9da92da46d28778cc7 : FAILURE in 1h 26m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/447552ce8581461f869d96074aeba046 : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":36},{"id":"621bc578265dc58ab90828a762d75052a573060c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-09 18:43:12.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"dfb484a911bb049349027d269928e2b8b0150d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 18:44:05.000000000","message":"Patch Set 36:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":36},{"id":"68b55635afcc11306903d332ff5ce93569fd4260","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-09 19:00:40.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"538111bcdb9cab98e0618ee12ab54ef26fa948a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 19:01:26.000000000","message":"Patch Set 36:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":36},{"id":"fad0a913015cffe3176fe3cd04e95cf51c97fd8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-09 19:06:16.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"56d041add6aea729da6361047984fb4d9342dded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 20:02:18.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/01c638eee2984a7b8a9840ac66f249e0 : SUCCESS in 47m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d13e9594c22c49fd87cda6899c97ad55 : SUCCESS in 43m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e6704aa0972c430bb25c4c2599897a79 : FAILURE in 45m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/69885b9444f54bdaafb5b41ab2cc35ce : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":37},{"id":"c9b640d7484ea448eb80d1ba4c05fbbaf1d0ed2a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-10 00:28:39.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"8f31c005b601f66fbdabf889b20cf2d515009043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 01:07:42.000000000","message":"Patch Set 37:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bde5b9613d71483b8f06afdefe7e5441 : FAILURE in 31m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/14facc11defa4ce18fc2b27adaad4335 : SUCCESS in 32m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c464b7515e74521aaa4c8846c874387 : SUCCESS in 36m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/50cbec4b33e249479632244db672e2a4 : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":37},{"id":"b70a520e5af6fb084b1b7aa664ef31396c93cd86","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-10 21:26:42.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"cc865d08e9740179b3105352604dfc30a3253bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 22:22:29.000000000","message":"Patch Set 37:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/420afabd723840fe9058f348ca7374c0 : SUCCESS in 45m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3a715520c9254b1989f38069644e3587 : FAILURE in 49m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0bf3eb0cb58943fbb9da174cc9547063 : FAILURE in 44m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/32fc9e6966984ef592e6b1e2ec4b45d6 : FAILURE in 48m 15s","accounts_in_message":[],"_revision_number":37},{"id":"a8893a60c5b0500864980f9bdca5eacb07271d1a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-10 23:30:33.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"74babed3feac196c0d588b89addaf41e14e0ec2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 00:19:06.000000000","message":"Patch Set 37:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bf74d54bf03e47a693fe6231345e64d7 : FAILURE in 42m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3577ffa1ec4841c3818d4250f2c7a853 : SUCCESS in 36m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9474a01a6a9e4ab6825eab5c35890fb9 : FAILURE in 37m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5b897d56a10d4ef0abf3a9447639eef2 : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":37},{"id":"9e7f1343b2cca2d6c7d41ea7a2224e483fbaf25b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-11 18:14:13.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"feaf192ebfb980c30a8f55c4cde6708ab5b29ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 19:07:53.000000000","message":"Patch Set 37:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dfe7586d679c4a578d89caf38979599d : FAILURE in 45m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cc0bd8066a8b469591e84b5fddb4b531 : FAILURE in 41m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3c67ca9b1e644fbb910eaba3c29f57dd : SUCCESS in 41m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2cb00111d6e34fbebbadd3107c2e386d : FAILURE in 43m 15s","accounts_in_message":[],"_revision_number":37},{"id":"c8f842e00cb98f6ec829bc010847dca9ec3b9904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-11 23:07:44.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"4be34d263897e85ebc13fa718d933f04332382ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 00:06:14.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e8a5db69e4964fe59d11a5ce58c984d3 : FAILURE in 40m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7380a60a9c1848dda0143260e82823d7 : FAILURE in 44m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/29723dbd56c448e69e5f2a20ea02f384 : SUCCESS in 49m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7642e09c2b5e4dfe940c23c7dc7cc1ba : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":38},{"id":"e6594d206fe58cef12ba010d25e1d92b15c9db2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-12 00:09:59.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"ea0fded03b101574536a82a871d1789f9b13a489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 01:15:21.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9971a97b88d54fdb8ba204f6a36d2a41 : FAILURE in 48m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0ac0d0c349da4f0c869c5ce5976f463b : FAILURE in 41m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8e042dd34a974c0b9f9935017950dd56 : SUCCESS in 38m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bb9a72a445dd4204b02af8832d1f06b2 : FAILURE in 56m 14s","accounts_in_message":[],"_revision_number":39},{"id":"e248ba97bb273ea9df23f23bb205c979f4e1fbf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-12 17:53:03.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"b44588d9b9c7a0e104190b284b43d0b5ae9e6972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 18:45:30.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c0e8a9495c944bd8b00c2d71c7b7dc3b : SUCCESS in 43m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5954971e255b4c6d81cbbe72b2b0393f : FAILURE in 43m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/704ee7b4146140e9a2e6dcedab1fd1e1 : SUCCESS in 43m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/187d1dd10021480d95ae0a42bccb1750 : SUCCESS in 44m 59s","accounts_in_message":[],"_revision_number":40},{"id":"4e34daacaf223f5d3e82e7a2bc1ef9a0a953b210","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-12 19:17:56.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"5f2edd736bfedb35a550b9d623a8dd8cd835da3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 20:12:26.000000000","message":"Patch Set 40:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1611feddc6fe41f899e109a3ee6c1b0c : SUCCESS in 48m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5f8346e019bf42c1b8ae62600f2c2003 : FAILURE in 45m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/625173a849da44f58c04e6566786e463 : FAILURE in 47m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6a122baa02464970a8f85f6806049d51 : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":40},{"id":"1afb636d246ce8f250a615706f5f1b5b6fa6d72d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-12 21:48:57.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"ce98b8e6f33b1dc356213b89df4b24a2c452b695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 22:36:11.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9cff897d5bed48129737222955e69e53 : SUCCESS in 41m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ad141c707972463eac273cfed2836517 : SUCCESS in 44m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c6346b3dbf5417cb5d907a095f661a4 : FAILURE in 41m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f6b0559f0f204a6ba4ccb1cfeeaa7ad7 : FAILURE in 42m 23s","accounts_in_message":[],"_revision_number":41},{"id":"9cbad6223046391ae7e98c209b758b261ad26868","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-12 22:40:15.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"2d4aba7835b7a8d5ab80ec4e6dc6e758c9b18944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 23:34:51.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bc81883d000847f0a4eeb3af75bb64cf : FAILURE in 50m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/711ce650fc254dcb88c4f87bd45310ef : SUCCESS in 48m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/60c4d7f955d546b088fb1e135b9d2aad : SUCCESS in 47m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a62bfebde3244231ab2abf0be9ee34fd : FAILURE in 50m 37s","accounts_in_message":[],"_revision_number":41},{"id":"98c285602565b75f1fb948050c3d3f9f7d910bfd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-12 23:37:59.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"9ea151dab85b7bace1b1ac4c460304bc69e045eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 00:32:24.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9659b9b4ccd34bf6b52a9e7830db7be6 : SUCCESS in 45m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e450e154e5ab4378b61e041bc75cc245 : SUCCESS in 42m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5d221924ffcd4e5c8f93b4bcbbca3307 : SUCCESS in 46m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9eea702ec531455ea50fc7a3aa0ae95c : FAILURE in 47m 34s","accounts_in_message":[],"_revision_number":41},{"id":"e69af5983d2934eaee8ff9ea5de78c99d281480f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-16 22:22:43.000000000","message":"Abandoned\n\nNo longer needed, https://github.com/unbit/uwsgi/issues/2368","accounts_in_message":[],"_revision_number":41}],"current_revision_number":41,"current_revision":"956ce104a4cdf25512425d28c28b57ca6795f6ba","revisions":{"d10d40cf2fafbab1f00b9f5cd51ad8f76623e55b":{"kind":"REWORK","_number":1,"created":"2021-04-15 21:05:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/1"}}},"commit":{"parents":[{"commit":"3285b92d6d5c032409734fab817e084ab4b027f1","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3285b92d6d5c032409734fab817e084ab4b027f1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:05:00.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d10d40cf2fafbab1f00b9f5cd51ad8f76623e55b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d10d40cf2fafbab1f00b9f5cd51ad8f76623e55b"}]},"branch":"refs/heads/master"},"fc29cd894194b39108121e47c47064b21eeeeffe":{"kind":"REWORK","_number":2,"created":"2021-04-15 21:25:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/2"}}},"commit":{"parents":[{"commit":"3285b92d6d5c032409734fab817e084ab4b027f1","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3285b92d6d5c032409734fab817e084ab4b027f1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:24:57.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc29cd894194b39108121e47c47064b21eeeeffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc29cd894194b39108121e47c47064b21eeeeffe"}]},"branch":"refs/heads/master"},"3e0d5a45138dc6a33a29e082a2343da914a5844c":{"kind":"REWORK","_number":3,"created":"2021-04-15 22:13:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/3"}}},"commit":{"parents":[{"commit":"3285b92d6d5c032409734fab817e084ab4b027f1","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3285b92d6d5c032409734fab817e084ab4b027f1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 22:12:55.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3e0d5a45138dc6a33a29e082a2343da914a5844c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3e0d5a45138dc6a33a29e082a2343da914a5844c"}]},"branch":"refs/heads/master"},"bb4b74471c7893217233b38f7979ef2bed9941ba":{"kind":"REWORK","_number":4,"created":"2021-04-15 23:09:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/4"}}},"commit":{"parents":[{"commit":"3285b92d6d5c032409734fab817e084ab4b027f1","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3285b92d6d5c032409734fab817e084ab4b027f1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 23:09:07.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bb4b74471c7893217233b38f7979ef2bed9941ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bb4b74471c7893217233b38f7979ef2bed9941ba"}]},"branch":"refs/heads/master"},"246452eef895f0a6d9fe7851d0ffa92fc8898a08":{"kind":"REWORK","_number":5,"created":"2021-04-16 20:45:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/5"}}},"commit":{"parents":[{"commit":"3285b92d6d5c032409734fab817e084ab4b027f1","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3285b92d6d5c032409734fab817e084ab4b027f1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-16 20:45:07.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/246452eef895f0a6d9fe7851d0ffa92fc8898a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/246452eef895f0a6d9fe7851d0ffa92fc8898a08"}]},"branch":"refs/heads/master"},"99828d05ce0f2ad5da2863dda9efbdef0c09e7fe":{"kind":"REWORK","_number":6,"created":"2021-04-16 21:12:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/6"}}},"commit":{"parents":[{"commit":"3285b92d6d5c032409734fab817e084ab4b027f1","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3285b92d6d5c032409734fab817e084ab4b027f1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-16 21:12:35.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99828d05ce0f2ad5da2863dda9efbdef0c09e7fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99828d05ce0f2ad5da2863dda9efbdef0c09e7fe"}]},"branch":"refs/heads/master"},"6dea89bdc7ab47057cb907b0840f8a4777e6dfa9":{"kind":"REWORK","_number":7,"created":"2021-04-16 21:33:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/7"}}},"commit":{"parents":[{"commit":"3285b92d6d5c032409734fab817e084ab4b027f1","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3285b92d6d5c032409734fab817e084ab4b027f1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-16 21:33:48.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6dea89bdc7ab47057cb907b0840f8a4777e6dfa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6dea89bdc7ab47057cb907b0840f8a4777e6dfa9"}]},"branch":"refs/heads/master"},"368e680e5486556965d9d13375cb04c198928623":{"kind":"REWORK","_number":8,"created":"2021-04-16 21:57:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/8"}}},"commit":{"parents":[{"commit":"3285b92d6d5c032409734fab817e084ab4b027f1","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3285b92d6d5c032409734fab817e084ab4b027f1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-16 21:57:38.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/368e680e5486556965d9d13375cb04c198928623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/368e680e5486556965d9d13375cb04c198928623"}]},"branch":"refs/heads/master"},"6b54235b4e8523cf84f611140a65db390a4e5850":{"kind":"REWORK","_number":9,"created":"2021-04-16 22:25:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/9"}}},"commit":{"parents":[{"commit":"3285b92d6d5c032409734fab817e084ab4b027f1","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3285b92d6d5c032409734fab817e084ab4b027f1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-16 22:25:19.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6b54235b4e8523cf84f611140a65db390a4e5850"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6b54235b4e8523cf84f611140a65db390a4e5850"}]},"branch":"refs/heads/master"},"23f383530cfe612533859616c2c037583861c6ea":{"kind":"REWORK","_number":10,"created":"2021-04-16 23:24:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/10"}}},"commit":{"parents":[{"commit":"3285b92d6d5c032409734fab817e084ab4b027f1","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3285b92d6d5c032409734fab817e084ab4b027f1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-16 23:24:26.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23f383530cfe612533859616c2c037583861c6ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23f383530cfe612533859616c2c037583861c6ea"}]},"branch":"refs/heads/master"},"15b6e24d4d61c0579d09dfc5671b4c4ecd3d6b3e":{"kind":"REWORK","_number":11,"created":"2021-11-02 03:50:57.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/11"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-02 03:49:29.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15b6e24d4d61c0579d09dfc5671b4c4ecd3d6b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15b6e24d4d61c0579d09dfc5671b4c4ecd3d6b3e"}]},"branch":"refs/heads/master"},"1838df8ac51c23fb4065f07f76ba78314746731e":{"kind":"REWORK","_number":12,"created":"2021-11-02 04:12:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/12"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-02 04:11:57.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1838df8ac51c23fb4065f07f76ba78314746731e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1838df8ac51c23fb4065f07f76ba78314746731e"}]},"branch":"refs/heads/master"},"7d98078d2a6c6f03901ca46002bbd9fd7f9b03d2":{"kind":"REWORK","_number":13,"created":"2021-11-02 04:42:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/13"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-02 04:42:20.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d98078d2a6c6f03901ca46002bbd9fd7f9b03d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d98078d2a6c6f03901ca46002bbd9fd7f9b03d2"}]},"branch":"refs/heads/master"},"dc48892f449dcbe27803c5d426a4be4863c44255":{"kind":"REWORK","_number":14,"created":"2021-11-02 05:49:53.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/06/786506/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/14"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-02 05:49:41.000000000","tz":60},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc48892f449dcbe27803c5d426a4be4863c44255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc48892f449dcbe27803c5d426a4be4863c44255"}]},"branch":"refs/heads/master"},"86ccfa60ee71d76f5556c474f1c5ef6a67cf352a":{"kind":"REWORK","_number":15,"created":"2021-11-02 07:36:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/06/786506/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/15"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-02 07:36:12.000000000","tz":60},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86ccfa60ee71d76f5556c474f1c5ef6a67cf352a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86ccfa60ee71d76f5556c474f1c5ef6a67cf352a"}]},"branch":"refs/heads/master"},"eeb2d68a01a3af179c8a23995baaaa48044ef19f":{"kind":"REWORK","_number":16,"created":"2021-11-02 16:40:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/16"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-02 16:39:44.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eeb2d68a01a3af179c8a23995baaaa48044ef19f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eeb2d68a01a3af179c8a23995baaaa48044ef19f"}]},"branch":"refs/heads/master"},"f5552f7351bbef17db0cc761b3466b7ec005f9bc":{"kind":"REWORK","_number":17,"created":"2021-11-02 17:07:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/17"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-02 17:07:42.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5552f7351bbef17db0cc761b3466b7ec005f9bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5552f7351bbef17db0cc761b3466b7ec005f9bc"}]},"branch":"refs/heads/master"},"3c26b8a2191f249b6ae2cabb88f846bb75aae4ec":{"kind":"REWORK","_number":18,"created":"2021-11-02 18:33:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/18"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-02 18:33:40.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c26b8a2191f249b6ae2cabb88f846bb75aae4ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c26b8a2191f249b6ae2cabb88f846bb75aae4ec"}]},"branch":"refs/heads/master"},"c147706c02a5aa1f1db885ddf21abf6e1496c65e":{"kind":"REWORK","_number":19,"created":"2021-11-02 21:54:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/19"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-02 21:54:24.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c147706c02a5aa1f1db885ddf21abf6e1496c65e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c147706c02a5aa1f1db885ddf21abf6e1496c65e"}]},"branch":"refs/heads/master"},"969a0271222f08fb032c1543fa94ea2d7ba116db":{"kind":"REWORK","_number":20,"created":"2021-11-02 22:33:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/20"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-02 22:32:56.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/969a0271222f08fb032c1543fa94ea2d7ba116db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/969a0271222f08fb032c1543fa94ea2d7ba116db"}]},"branch":"refs/heads/master"},"ad43f02a1ca81dbf57c0fabe21f49a7fb140f086":{"kind":"REWORK","_number":21,"created":"2021-11-02 23:17:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/21"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-02 23:17:11.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad43f02a1ca81dbf57c0fabe21f49a7fb140f086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad43f02a1ca81dbf57c0fabe21f49a7fb140f086"}]},"branch":"refs/heads/master"},"d146acb0a2f4fe3eabe9bdfaca20308c96d54b16":{"kind":"REWORK","_number":22,"created":"2021-11-03 18:43:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/786506/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/22"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-11-03 18:43:41.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d146acb0a2f4fe3eabe9bdfaca20308c96d54b16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d146acb0a2f4fe3eabe9bdfaca20308c96d54b16"}]},"branch":"refs/heads/master"},"7422731d72f1a4cb4c40e549f257ae66052a1b28":{"kind":"REWORK","_number":23,"created":"2021-11-03 21:50:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/23"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-03 21:49:56.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7422731d72f1a4cb4c40e549f257ae66052a1b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7422731d72f1a4cb4c40e549f257ae66052a1b28"}]},"branch":"refs/heads/master"},"efdf9d9d00e1f519734a7df6549dfbca009b88ad":{"kind":"REWORK","_number":24,"created":"2021-11-03 22:47:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/24"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-03 22:25:57.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/efdf9d9d00e1f519734a7df6549dfbca009b88ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/efdf9d9d00e1f519734a7df6549dfbca009b88ad"}]},"branch":"refs/heads/master"},"e1a65237008b681987b555aa82aec309ad88cfae":{"kind":"REWORK","_number":25,"created":"2021-11-04 20:07:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/786506/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/25"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-11-04 20:07:38.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e1a65237008b681987b555aa82aec309ad88cfae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e1a65237008b681987b555aa82aec309ad88cfae"}]},"branch":"refs/heads/master"},"e27e057389f1dd81d3c809142cfc5c3c63d6efcc":{"kind":"REWORK","_number":26,"created":"2021-11-04 21:03:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/26"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-04 21:03:28.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e27e057389f1dd81d3c809142cfc5c3c63d6efcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e27e057389f1dd81d3c809142cfc5c3c63d6efcc"}]},"branch":"refs/heads/master"},"31faf1149cdf4ed7941df9930568abb85efed4a1":{"kind":"REWORK","_number":27,"created":"2021-11-04 21:45:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/27"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-04 21:44:07.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/31faf1149cdf4ed7941df9930568abb85efed4a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/31faf1149cdf4ed7941df9930568abb85efed4a1"}]},"branch":"refs/heads/master"},"af4625bef78228bae362eec896bf1ad2aa1a6e40":{"kind":"REWORK","_number":28,"created":"2021-11-04 22:09:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/28"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-04 22:09:32.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af4625bef78228bae362eec896bf1ad2aa1a6e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af4625bef78228bae362eec896bf1ad2aa1a6e40"}]},"branch":"refs/heads/master"},"1bebdab2850eab47b9144594883089d3249d1b76":{"kind":"REWORK","_number":29,"created":"2021-11-04 23:17:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/29"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-04 23:16:56.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1bebdab2850eab47b9144594883089d3249d1b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1bebdab2850eab47b9144594883089d3249d1b76"}]},"branch":"refs/heads/master"},"1b9aa5b3bb995f8ff85309f4f4bfe49c70478e40":{"kind":"REWORK","_number":30,"created":"2021-11-04 23:54:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/30"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-04 23:53:59.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b9aa5b3bb995f8ff85309f4f4bfe49c70478e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b9aa5b3bb995f8ff85309f4f4bfe49c70478e40"}]},"branch":"refs/heads/master"},"1a5572299fee3dea5a25b5546b8679042e463bdd":{"kind":"REWORK","_number":31,"created":"2021-11-05 00:33:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/31"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-05 00:33:03.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a5572299fee3dea5a25b5546b8679042e463bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a5572299fee3dea5a25b5546b8679042e463bdd"}]},"branch":"refs/heads/master"},"047de4977777d151a0e5c933add983310b5c8699":{"kind":"REWORK","_number":32,"created":"2021-11-05 16:19:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/32"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-05 16:19:50.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/047de4977777d151a0e5c933add983310b5c8699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/047de4977777d151a0e5c933add983310b5c8699"}]},"branch":"refs/heads/master"},"1f9ed51d3cb00d6de446f44c777c77de83026068":{"kind":"REWORK","_number":33,"created":"2021-11-05 17:00:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/33"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-05 17:00:01.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f9ed51d3cb00d6de446f44c777c77de83026068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f9ed51d3cb00d6de446f44c777c77de83026068"}]},"branch":"refs/heads/master"},"7c9ba267642cc247ea02a35457f0bac3e21da94a":{"kind":"REWORK","_number":34,"created":"2021-11-05 17:45:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/34"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-05 17:45:44.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c9ba267642cc247ea02a35457f0bac3e21da94a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c9ba267642cc247ea02a35457f0bac3e21da94a"}]},"branch":"refs/heads/master"},"feaffb154c32e881c0af562b098a438cd14e0f80":{"kind":"REWORK","_number":35,"created":"2021-11-05 18:19:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/35"}}},"commit":{"parents":[{"commit":"146fd3632d0f107d461525b339b304176193f30e","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146fd3632d0f107d461525b339b304176193f30e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-05 18:19:49.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/feaffb154c32e881c0af562b098a438cd14e0f80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/feaffb154c32e881c0af562b098a438cd14e0f80"}]},"branch":"refs/heads/master"},"c66c8346029c85e51c6ce06d5fe7d9e3b30620da":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2021-11-09 01:47:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/786506/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/36"}}},"commit":{"parents":[{"commit":"4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-11-09 01:47:57.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c66c8346029c85e51c6ce06d5fe7d9e3b30620da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c66c8346029c85e51c6ce06d5fe7d9e3b30620da"}]},"branch":"refs/heads/master","description":"Rebase"},"71841c2a409264932eebf9692c072b153522ebc5":{"kind":"REWORK","_number":37,"created":"2021-11-09 19:06:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/37"}}},"commit":{"parents":[{"commit":"4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-09 19:05:44.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/71841c2a409264932eebf9692c072b153522ebc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/71841c2a409264932eebf9692c072b153522ebc5"}]},"branch":"refs/heads/master"},"4ece8b5cb79d3bb498f7c11ff742921de8f23ca8":{"kind":"REWORK","_number":38,"created":"2021-11-11 23:07:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/38"}}},"commit":{"parents":[{"commit":"4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-11 23:07:15.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ece8b5cb79d3bb498f7c11ff742921de8f23ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ece8b5cb79d3bb498f7c11ff742921de8f23ca8"}]},"branch":"refs/heads/master"},"958621f95964b56b92efae1b0df21e53a8f56219":{"kind":"REWORK","_number":39,"created":"2021-11-12 00:09:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/39"}}},"commit":{"parents":[{"commit":"4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-12 00:09:49.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/958621f95964b56b92efae1b0df21e53a8f56219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/958621f95964b56b92efae1b0df21e53a8f56219"}]},"branch":"refs/heads/master"},"763eaa7efe5a06b3c1325d25d40984f842394362":{"kind":"REWORK","_number":40,"created":"2021-11-12 17:53:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/40"}}},"commit":{"parents":[{"commit":"4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-12 17:52:55.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/763eaa7efe5a06b3c1325d25d40984f842394362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/763eaa7efe5a06b3c1325d25d40984f842394362"}]},"branch":"refs/heads/master"},"956ce104a4cdf25512425d28c28b57ca6795f6ba":{"kind":"REWORK","_number":41,"created":"2021-11-12 21:48:57.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/786506/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/786506/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/786506/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/786506/41"}}},"commit":{"parents":[{"commit":"4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9","subject":"Revert \"Move back to mod_wsgi for devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f270c3aa33c37c8a4aa9de4ac4bdfa91f608de9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-15 21:00:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-12 21:48:45.000000000","tz":0},"subject":"DNM: Debug uWSGI","message":"DNM: Debug uWSGI\n\nChange-Id: I638626562038dd29e9a10510f07c4dcb1bae6224\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/956ce104a4cdf25512425d28c28b57ca6795f6ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/956ce104a4cdf25512425d28c28b57ca6795f6ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
