)]}'
{"id":"openstack%2Fdesignate~786586","triplet_id":"openstack%2Fdesignate~master~I1b68845c27ecb13568394f3a1fc7b310734ae991","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I1b68845c27ecb13568394f3a1fc7b310734ae991","subject":"Drop lower constraints testing","status":"ABANDONED","created":"2021-04-16 07:07:41.000000000","updated":"2021-05-24 06:40:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0c0ce87c4ed06af7f6ad03082a51b7d461db68b","_number":786586,"virtual_id_number":786586,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-16 08:08:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-16 08:08:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e20893498c4312cbbfff5266d805471089b50efe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-04-16 07:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"445a74770d70c15d3a50c21068977f626abed4c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 08:08:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/591a58a7de734a91bd3c2ebbdd9142cf : SUCCESS in 38m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b1454d9c48164bbfa2b8c2cab46aa151 : SUCCESS in 34m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/936d004c5f154fe0a1b9926fa8463741 : SUCCESS in 55m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e3fc668ddb74613aeff41066d686d19 : SUCCESS in 33m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a2fe2b46201546e0b51a556c55306508 : SUCCESS in 35m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eace12d775ea479c86733c4414c4327c : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a627977023c545058edac15e57a50607 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16c0f7440aa64ba297f91dec646c7525 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc546606a0a5418db020a2a5c1be2b4c : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18d9cf0476fb4aa8a06ba19744e69bf5 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed39eff7abdc4306a2182c4eba01b298 : SUCCESS in 8m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1028b2527a174e5bbfa4810fd42ad5a2 : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e0c0ce87c4ed06af7f6ad03082a51b7d461db68b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-05-24 06:40:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"880a2e686372cda852dd3cabdc2fca26dfc1fb84","revisions":{"880a2e686372cda852dd3cabdc2fca26dfc1fb84":{"kind":"REWORK","_number":1,"created":"2021-04-16 07:07:41.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/86/786586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/86/786586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/786586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/786586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/786586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/86/786586/1"}}},"commit":{"parents":[{"commit":"518e8a74c6d3809914ab19241d7062815ae69623","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/518e8a74c6d3809914ab19241d7062815ae69623"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-04-16 07:07:13.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-04-16 07:07:36.000000000","tz":480},"subject":"Drop lower constraints testing","message":"Drop lower constraints testing\n\nProjects like oslo are dropping lower constraints\ntesting. There is also discussion to drop it until\nwe find a proper setup. For details, please follow\nthe ML thread. TC will have updates soon.\n\nhttp: //lists.openstack.org/pipermail/openstack-discuss/2021-January/019679.html\nChange-Id: I1b68845c27ecb13568394f3a1fc7b310734ae991\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/880a2e686372cda852dd3cabdc2fca26dfc1fb84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/880a2e686372cda852dd3cabdc2fca26dfc1fb84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
