)]}'
{"id":"openstack%2Fdesignate~787959","triplet_id":"openstack%2Fdesignate~master~I2635a3532764bf68ac6a14eac924ab2a7d4b410a","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"28609":{"account":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"last_update":"2022-07-14 20:11:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2635a3532764bf68ac6a14eac924ab2a7d4b410a","subject":"Changing minimum TLL filed to \"0\" instead of \"1\" according RFC","status":"ABANDONED","created":"2021-04-26 11:26:34.000000000","updated":"2022-07-14 20:11:20.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"04e92b041891a5f2253e6425d021b96ceec4edde","_number":787959,"virtual_id_number":787959,"owner":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-26 12:53:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"value":-1,"date":"2022-07-13 16:58:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 11:27:36.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},"state":"REVIEWER"},{"updated":"2021-04-26 12:53:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 16:58:05.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"6467acd8dc1830990b1aba927be480820ae3978b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-04-26 11:26:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e0fb70680d8f8f60f0ca5d4af4d4dea67a558ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:53:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/602ca16cdc76407589f3b2351e72a052 : SUCCESS in 32m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eb3988b283f04a59bb551469baa266cc : SUCCESS in 31m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/462937c621c54cab904d26657aedfaeb : FAILURE in 1h 00m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8318f687b88c47618c61e6e4c7702984 : SUCCESS in 40m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3d060dbdf545486bbb353942c637047a : SUCCESS in 33m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2299d19f6df3467f9a8f60aa8f94b527 : FAILURE in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5aa59b82a594c0e881fec6f0ca3be2c : FAILURE in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc742889809a4e8e859f4d0f48a89c10 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50c5e816b2b14e42b6dbe43a204bf7fe : FAILURE in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e47e0e72ec0d495da089da84dd45b5a3 : FAILURE in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f65407baf1e445f4bf408ff7934fbfb2 : FAILURE in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6c5b69ae8f64d68bb76567cf6bc656a : SUCCESS in 9m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f62c82a38f6045349b61e48aa58210cd : SUCCESS in 29m 35s","accounts_in_message":[],"_revision_number":1},{"id":"14e70fe89009510afad93236845cdb30a877d0b6","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-07-13 16:58:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04e92b041891a5f2253e6425d021b96ceec4edde","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-14 20:11:20.000000000","message":"Abandoned\n\nOpps, yep. Sorry Arkady we missed this. The other patch had test coverage though.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6933666ad673c3cf3af436c3e91e81b0a37752c7","revisions":{"6933666ad673c3cf3af436c3e91e81b0a37752c7":{"kind":"REWORK","_number":1,"created":"2021-04-26 11:26:34.000000000","uploader":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"ref":"refs/changes/59/787959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/59/787959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/59/787959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/59/787959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/59/787959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/59/787959/1"}}},"commit":{"parents":[{"commit":"518e8a74c6d3809914ab19241d7062815ae69623","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/518e8a74c6d3809914ab19241d7062815ae69623"}]}],"author":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-04-26 11:21:04.000000000","tz":180},"committer":{"name":"Arkady Shtempler","email":"ashtempl@redhat.com","date":"2021-04-26 11:26:29.000000000","tz":180},"subject":"Changing minimum TLL filed to \"0\" instead of \"1\" according RFC","message":"Changing minimum TLL filed to \"0\" instead of \"1\" according RFC\n\nRFC: https://tools.ietf.org/html/rfc2181#section-8\n8. Time to Live (TTL)\n\n   The definition of values appropriate to the TTL field in STD 13 is\n   not as clear as it could be, with respect to how many significant\n   bits exist, and whether the value is signed or unsigned.  It is\n   hereby specified that a TTL value is an unsigned number, with a\n   minimum value of 0, and a maximum value of 2147483647.\n\nChange-Id: I2635a3532764bf68ac6a14eac924ab2a7d4b410a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6933666ad673c3cf3af436c3e91e81b0a37752c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6933666ad673c3cf3af436c3e91e81b0a37752c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
