)]}'
{"id":"openstack%2Fdesignate~788874","triplet_id":"openstack%2Fdesignate~master~I5400cfe61b2608aa4a1b383f140ef71c2dc342f7","project":"openstack/designate","branch":"master","topic":"user_recordset_doc","hashtags":[],"change_id":"I5400cfe61b2608aa4a1b383f140ef71c2dc342f7","subject":"Add user doc for managing recordsets","status":"MERGED","created":"2021-04-30 03:10:09.000000000","updated":"2021-06-09 21:35:25.000000000","submitted":"2021-06-09 21:34:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":114,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"788874-user_recordset_doc","meta_rev_id":"3f3f164452c4045457323ba569c73417d5b76089","_number":788874,"virtual_id_number":788874,"owner":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 21:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-06-09 21:08:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-09 21:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-09 21:20:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 03:19:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 23:01:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-05-19 12:41:56.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-05-20 15:26:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-05-26 14:16:32.000000000","updated_by":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"reviewer":{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},"state":"REVIEWER"}],"messages":[{"id":"beadd51180ab9154c2b6f1bf331c9cff55e7b290","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-04-30 03:10:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1970fd4b54b0cba15edcb3e78ba38ddeee71c636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 03:19:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343464d88b7249b48af8f846251e67f3 : FAILURE in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8890b2f006b94d6ab2c48682094ef8ae : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3759f9eb7483e85a16be01db6efaa6a1bec22d99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-05-03 03:55:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8db90c2c47032b638913a0c5c2ad5561156fde0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 04:06:46.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/871ddbae0e8c4e7aa379ef381b48485d : FAILURE in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc6c00b20764c43ae0816e9fa0cb311 : FAILURE in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"18e1bf0302d840f9440eb93216f6ada80cef7acf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-05-03 06:44:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2455b6b93992a6689834993cef00dab3623c0b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 06:56:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5cffef785c34ba3bb43c6e0b3e540ce : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0d65957faa46adb94025aceb4fbf93 : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":3},{"id":"961da6250ea74aa21f2e4c941cc0c981b51a5767","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-18 23:01:20.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)\n\nI have a couple of comments here.","accounts_in_message":[],"_revision_number":3},{"id":"144628ad5b1bff3e17a2751d5525c19444d1ae0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-05-19 06:32:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9307533ff4544a0d5f4cc5ebf384a9449c46aad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 06:44:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c147916baa4f3a96d289933f3b1fe6 : FAILURE in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89384e6375294a9b8503e03b01c352f8 : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":4},{"id":"08283975ef503d86fe4740f3f903f29571ad55ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-05-19 07:19:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"351fe5b12a76874f61346f61e8c02670f0a9e0c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 07:32:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d32e767e4c94e82b6d7ed5c16efd9e9 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3221f469134ff48477694beebdd5bb : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":5},{"id":"d28dfa9be3bdcdf346c4ab315b974d67a482bc61","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-05-19 12:41:56.000000000","message":"Patch Set 5: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e89b90e9d3ea32ae3e354b8b7f48c216605c24d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-05-20 02:57:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7140c68853d38e95b39cc5d5eab5c56d3c4eefef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 03:09:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf830055067449fb96757c921d39b8b : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bccffe009c9944708a9c391465fd9fd8 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":6},{"id":"02db7155e52a2bb58cd6cfb7d6b584690d72064b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-05-20 12:14:51.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":6},{"id":"945d06ca02e791ed8d6b58952219383f21b7b93d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-05-20 15:26:59.000000000","message":"Patch Set 6: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a444e835d5b5bb5586965cf39d399d8796517388","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-05-24 05:01:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a6a2518396e01754087ccc152b231899022789ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 05:12:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74589ae166724180bb5baafb1306e842 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c05e0e3c694defb39c820721e87195 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":7},{"id":"9bd9e85b935f2b9a395f6dfeb2809dbf1724b35b","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-05-26 14:16:32.000000000","message":"Patch Set 7:\n\n(27 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b7c684d5f51a211d5f5b644ac528e4f013386a3f","author":{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},"date":"2021-05-27 22:21:28.000000000","message":"Patch Set 7: Code-Review+1\n\n(21 comments)\n\nHi Michael,\n\nGreat job on the doc! You\u0027re making my life easier downstream. :^)\n\nIf you got barraged with a bunch of comments that look like duplicates, it\u0027s because I initially edited the wrong Patchset (Patchset 3)! Sorry about that. Please be patient with me as I\u0027m still learning Gerrit and the open source way. :^)\n\nHope that my comments are helpful.\n\nPlease feel free to reject any that you don\u0027t agree with. \n\nBest,\n--Greg","accounts_in_message":[],"_revision_number":7},{"id":"68b7cd02a1bd9e435434edc050207469bad2dda1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-06-04 01:31:12.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ab0f9248f4aff7a098e02633bd1b682a880a3370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 01:40:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532a4ccfca7f4a0c9cbf84baef14ca43 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa63f69ebd1b49ada2f58fdb0c58a906 : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":8},{"id":"b7e4cd75c9e6a0b84ff21681f73eee47daabf965","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-06-08 02:23:16.000000000","message":"Patch Set 8:\n\n(19 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b7a6a7a698a77c1bcf3d5ce99813ba3624bddab3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-08 21:45:45.000000000","message":"Patch Set 8: Code-Review-1\n\n(10 comments)\n\nI have a few comments.","accounts_in_message":[],"_revision_number":8},{"id":"65bc638f0155e0c6fe9c1d9fc1bc53657d2895a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-06-09 05:55:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"012d4fbf97f725d3fd37fd1e3a887f95d4e17368","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-06-09 05:57:22.000000000","message":"Patch Set 9:\n\n(10 comments)","accounts_in_message":[],"_revision_number":9},{"id":"609d98ff2ccfc18c970817c5e6bd240ca4bc55fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 06:03:26.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb5c6c16a14244889504ebddb1b6b18f : FAILURE in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663743d181214173978253a7aaa7cd7d : FAILURE in 4m 58s","accounts_in_message":[],"_revision_number":9},{"id":"c8673357ce314b217382740fc90bc430b89d2cdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-06-09 06:08:10.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"26059a6255433e16eaa8369d90ff7ff57d77ab92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 06:26:51.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de5f2337ea50499c8862e9135cbbbec0 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68a3e005170842efb19026e36761143b : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":10},{"id":"8aaf486e421b444a56a7c00b52c8d3fb6a56a00f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-06-09 21:08:31.000000000","message":"Patch Set 10: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":10},{"id":"8fbd277b2278ab7811aac44748903f4b145615f8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-09 21:20:29.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":10},{"id":"3d6e1676d731b6aa16511f597bab7db7b88ed8f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:20:45.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"1c6622bc76cd5bbdfb46c6b779ff1c0508515ed5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:33:57.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749c00c130a24d959df0cb9ee3ea0bf8 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee93747aef246288fb00c43aa1a8b96 : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":10},{"id":"12b02dd0ee2306693c20524ce25ab46841a2665c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:34:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"3f3f164452c4045457323ba569c73417d5b76089","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:35:25.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72991ff246b942548bd9de1be22a1794 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"999abb0d925b3c97274b0d0f116e3fff9cdf35df","revisions":{"1958de443721c0a4862a5bb8ffd8d5669f3fb961":{"kind":"REWORK","_number":1,"created":"2021-04-30 03:10:09.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/74/788874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/788874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/788874/1"}}},"commit":{"parents":[{"commit":"3bde7a843f70d9d00ea51126199c480b6061e9db","subject":"Add user doc for managing zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bde7a843f70d9d00ea51126199c480b6061e9db"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"subject":"[WIP] Add user doc for managing recordsets","message":"[WIP] Add user doc for managing recordsets\n\nChange-Id: I5400cfe61b2608aa4a1b383f140ef71c2dc342f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1958de443721c0a4862a5bb8ffd8d5669f3fb961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1958de443721c0a4862a5bb8ffd8d5669f3fb961"}]},"branch":"refs/heads/master"},"d151ee49a5d071b4876dedc36eed4ae37264b762":{"kind":"REWORK","_number":2,"created":"2021-05-03 03:55:53.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/74/788874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/788874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/788874/2"}}},"commit":{"parents":[{"commit":"3bde7a843f70d9d00ea51126199c480b6061e9db","subject":"Add user doc for managing zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bde7a843f70d9d00ea51126199c480b6061e9db"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-03 03:55:10.000000000","tz":600},"subject":"Add user doc for managing recordsets","message":"Add user doc for managing recordsets\n\nNameserver discovery not possible yet on the client, doc should\nbe updated when this is available.\n\nChange-Id: I5400cfe61b2608aa4a1b383f140ef71c2dc342f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d151ee49a5d071b4876dedc36eed4ae37264b762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d151ee49a5d071b4876dedc36eed4ae37264b762"}]},"branch":"refs/heads/master"},"d038c47bf83c3930e19c219efc6dab9d147f393c":{"kind":"REWORK","_number":3,"created":"2021-05-03 06:44:08.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/74/788874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/788874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/788874/3"}}},"commit":{"parents":[{"commit":"3bde7a843f70d9d00ea51126199c480b6061e9db","subject":"Add user doc for managing zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bde7a843f70d9d00ea51126199c480b6061e9db"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-03 06:43:59.000000000","tz":600},"subject":"Add user doc for managing recordsets","message":"Add user doc for managing recordsets\n\nNameserver discovery not possible yet on the client, doc should\nbe updated when this is available.\n\nChange-Id: I5400cfe61b2608aa4a1b383f140ef71c2dc342f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d038c47bf83c3930e19c219efc6dab9d147f393c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d038c47bf83c3930e19c219efc6dab9d147f393c"}]},"branch":"refs/heads/master"},"4bffa83e6075f76097b50b42a29135202c8d8c73":{"kind":"REWORK","_number":4,"created":"2021-05-19 06:32:41.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/74/788874/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/788874/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/788874/4"}}},"commit":{"parents":[{"commit":"3bde7a843f70d9d00ea51126199c480b6061e9db","subject":"Add user doc for managing zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bde7a843f70d9d00ea51126199c480b6061e9db"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-19 06:32:23.000000000","tz":600},"subject":"Add user doc for managing recordsets","message":"Add user doc for managing recordsets\n\nNameserver discovery not possible yet on the client, doc should\nbe updated when this is available.\n\nChange-Id: I5400cfe61b2608aa4a1b383f140ef71c2dc342f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4bffa83e6075f76097b50b42a29135202c8d8c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4bffa83e6075f76097b50b42a29135202c8d8c73"}]},"branch":"refs/heads/master"},"4019bd6c2a7fa9c2a2504d7e2a96487e6cc27f60":{"kind":"REWORK","_number":5,"created":"2021-05-19 07:19:05.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/74/788874/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/788874/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/788874/5"}}},"commit":{"parents":[{"commit":"3bde7a843f70d9d00ea51126199c480b6061e9db","subject":"Add user doc for managing zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bde7a843f70d9d00ea51126199c480b6061e9db"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-19 07:18:54.000000000","tz":600},"subject":"Add user doc for managing recordsets","message":"Add user doc for managing recordsets\n\nNameserver discovery not possible yet on the client, doc should\nbe updated when this is available.\n\nChange-Id: I5400cfe61b2608aa4a1b383f140ef71c2dc342f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4019bd6c2a7fa9c2a2504d7e2a96487e6cc27f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4019bd6c2a7fa9c2a2504d7e2a96487e6cc27f60"}]},"branch":"refs/heads/master"},"235d0196e87b6d2a929e16533032d54599cf257b":{"kind":"REWORK","_number":6,"created":"2021-05-20 02:57:31.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/74/788874/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/788874/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/788874/6"}}},"commit":{"parents":[{"commit":"3bde7a843f70d9d00ea51126199c480b6061e9db","subject":"Add user doc for managing zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bde7a843f70d9d00ea51126199c480b6061e9db"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-20 02:57:23.000000000","tz":600},"subject":"Add user doc for managing recordsets","message":"Add user doc for managing recordsets\n\nNameserver discovery not possible yet on the client, doc should\nbe updated when this is available.\n\nChange-Id: I5400cfe61b2608aa4a1b383f140ef71c2dc342f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/235d0196e87b6d2a929e16533032d54599cf257b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/235d0196e87b6d2a929e16533032d54599cf257b"}]},"branch":"refs/heads/master"},"8bd479ec82b433a1423cb0e97dec409775fd0efd":{"kind":"REWORK","_number":7,"created":"2021-05-24 05:01:22.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/74/788874/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/788874/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/788874/7"}}},"commit":{"parents":[{"commit":"3bde7a843f70d9d00ea51126199c480b6061e9db","subject":"Add user doc for managing zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bde7a843f70d9d00ea51126199c480b6061e9db"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-24 05:00:13.000000000","tz":600},"subject":"Add user doc for managing recordsets","message":"Add user doc for managing recordsets\n\nDocumentation targeted at member personas who wish to manage records.\n\nChange-Id: I5400cfe61b2608aa4a1b383f140ef71c2dc342f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8bd479ec82b433a1423cb0e97dec409775fd0efd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8bd479ec82b433a1423cb0e97dec409775fd0efd"}]},"branch":"refs/heads/master"},"7e428f311242448e0d71f4b4aaf07efc5b0abd91":{"kind":"REWORK","_number":8,"created":"2021-06-04 01:31:12.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/74/788874/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/788874/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/788874/8"}}},"commit":{"parents":[{"commit":"3bde7a843f70d9d00ea51126199c480b6061e9db","subject":"Add user doc for managing zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bde7a843f70d9d00ea51126199c480b6061e9db"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-06-04 01:31:03.000000000","tz":600},"subject":"Add user doc for managing recordsets","message":"Add user doc for managing recordsets\n\nDocumentation targeted at member personas who wish to manage records.\n\nChange-Id: I5400cfe61b2608aa4a1b383f140ef71c2dc342f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7e428f311242448e0d71f4b4aaf07efc5b0abd91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7e428f311242448e0d71f4b4aaf07efc5b0abd91"}]},"branch":"refs/heads/master"},"18a6586220e8ab8eb2aff4c90c6c024e3e34a813":{"kind":"REWORK","_number":9,"created":"2021-06-09 05:55:16.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/74/788874/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/788874/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/788874/9"}}},"commit":{"parents":[{"commit":"3bde7a843f70d9d00ea51126199c480b6061e9db","subject":"Add user doc for managing zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bde7a843f70d9d00ea51126199c480b6061e9db"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-06-09 05:55:04.000000000","tz":600},"subject":"Add user doc for managing recordsets","message":"Add user doc for managing recordsets\n\nDocumentation targeted at member personas who wish to manage records.\n\nChange-Id: I5400cfe61b2608aa4a1b383f140ef71c2dc342f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/18a6586220e8ab8eb2aff4c90c6c024e3e34a813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/18a6586220e8ab8eb2aff4c90c6c024e3e34a813"}]},"branch":"refs/heads/master"},"999abb0d925b3c97274b0d0f116e3fff9cdf35df":{"kind":"REWORK","_number":10,"created":"2021-06-09 06:08:10.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/74/788874/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/788874/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/788874/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/788874/10"}}},"commit":{"parents":[{"commit":"3bde7a843f70d9d00ea51126199c480b6061e9db","subject":"Add user doc for managing zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bde7a843f70d9d00ea51126199c480b6061e9db"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-04-30 03:09:35.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-06-09 06:07:58.000000000","tz":600},"subject":"Add user doc for managing recordsets","message":"Add user doc for managing recordsets\n\nDocumentation targeted at member personas who wish to manage records.\n\nChange-Id: I5400cfe61b2608aa4a1b383f140ef71c2dc342f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/999abb0d925b3c97274b0d0f116e3fff9cdf35df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/999abb0d925b3c97274b0d0f116e3fff9cdf35df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
