)]}'
{"id":"openstack%2Fdesignate~791631","triplet_id":"openstack%2Fdesignate~master~I1a5f7dfee3d1a5a8b1930c16bdf2efb7a31d17f6","project":"openstack/designate","branch":"master","topic":"pools_auth","hashtags":[],"change_id":"I1a5f7dfee3d1a5a8b1930c16bdf2efb7a31d17f6","subject":"Allow project readers to read pools API","status":"ABANDONED","created":"2021-05-17 02:50:07.000000000","updated":"2022-09-24 05:53:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"315ef454ea9fe15635974a2031b05ce614fbbf8a","_number":791631,"virtual_id_number":791631,"owner":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-17 03:43:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 03:43:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d0b51338f306ec0fd4db2b115d33c0a12fb7a04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-05-17 02:50:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31298005d90abf401192cf0d837728a711bf4a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 03:43:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8235e1d4842544d587aa4e8f3a17477d : SUCCESS in 35m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ce713704b94e4275b34e960e2e18a082 : SUCCESS in 29m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e1031a4b543946c885338d57fa9f8f28 : SUCCESS in 52m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a419e03f571643fa824f0f7f46ec737e : SUCCESS in 36m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7ead239c9f7b471e8003000d3df21879 : SUCCESS in 30m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a8f6ea0b35044afa70fc2ab11419521 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dabe31b354624f28950487b9e45cb81d : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90674fd391b743a6bda8ea3866eb1e57 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25e6647a38874e17af32a48dbe5af6e9 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ca160cfdc4d4ce789c69fe9a34d41de : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f0d65355cf54fb3a39d9906c35e95e2 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26aac83cc38d44c48874162d2b308198 : SUCCESS in 7m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/909261bf44654a3fa589b87b15a8701a : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":1},{"id":"315ef454ea9fe15635974a2031b05ce614fbbf8a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:53:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"176c042460fdeb53388d9598ee6a41b51c0b0966","revisions":{"176c042460fdeb53388d9598ee6a41b51c0b0966":{"kind":"REWORK","_number":1,"created":"2021-05-17 02:50:07.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/31/791631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/791631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/791631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/791631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/791631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/791631/1"}}},"commit":{"parents":[{"commit":"290b8c068e1ebc11faf991df27438c91d4e94b38","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/290b8c068e1ebc11faf991df27438c91d4e94b38"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-17 02:35:36.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-17 02:35:36.000000000","tz":600},"subject":"Allow project readers to read pools API","message":"Allow project readers to read pools API\n\nWhen creating a zone, users are able to use the attributes field to\nhint to the attributes scheduler (if available) which pool to\nuse for the zone. By allowing users read access to the pools API,\nthey will be able to discover which attributes and pools are available\nwithout needing to consult either the cloud operator or cloud-specific\ndocumentation.\n\nChange-Id: I1a5f7dfee3d1a5a8b1930c16bdf2efb7a31d17f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/176c042460fdeb53388d9598ee6a41b51c0b0966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/176c042460fdeb53388d9598ee6a41b51c0b0966"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
