)]}'
{"id":"openstack%2Fdesignate~793325","triplet_id":"openstack%2Fdesignate~master~I7f3662c57ee9bccb42381134523e8fdd21e93740","project":"openstack/designate","branch":"master","topic":"blacklist_doc","hashtags":[],"change_id":"I7f3662c57ee9bccb42381134523e8fdd21e93740","subject":"Improvements to zone blacklist doc","status":"MERGED","created":"2021-05-27 08:55:48.000000000","updated":"2021-06-29 23:51:18.000000000","submitted":"2021-06-29 23:50:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"793325-blacklist_doc","meta_rev_id":"1c2696a0cb803d83b452462d2f3dcc0723eacd7a","_number":793325,"virtual_id_number":793325,"owner":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-29 23:50:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-06-28 04:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-29 23:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},{"value":1,"date":"2021-06-29 05:45:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-29 23:30:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 09:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-02 15:00:38.000000000","updated_by":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"reviewer":{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},"state":"REVIEWER"},{"updated":"2021-06-23 23:42:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-28 04:57:14.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-06-29 05:45:22.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"01f17a022fc310472d9803d514991d8bf311756c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-05-27 08:55:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27e7232095ee34d37024050989596feae69273c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 09:05:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0421dfe5e744b40b4a3fafc7856e12c : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5265528b5a334ccba78014380e9281be : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3a6a80c24945405e7fb90b95ccdf9f74dba6d98d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-06-02 03:36:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6851f962514ca76a05e2e548d79d7d4feaa9b0a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 03:45:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb6cb1ab2cfe47a8949f266f156c6a67 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b7de25ebd74ace9714c5de6517e4cb : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":2},{"id":"78fe993b2a5771e014da809392ce1481d763cfb3","author":{"_account_id":31933,"name":"Greg Rakauskas","display_name":"Greg Rakauskas","email":"gregraka@redhat.com","username":"gregraka"},"date":"2021-06-11 19:26:36.000000000","message":"Patch Set 2: Code-Review+1\n\n(13 comments)\n\nHi Michael C,\n\nGood job! I like how you broke up and annotated the REST API example.\n\nMade some edits, which you can take or leave.\n\nBest,\n--Greg","accounts_in_message":[],"_revision_number":2},{"id":"c3611376483c42435371f65b8fa61da739afb89d","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-06-16 02:01:36.000000000","message":"Patch Set 2:\n\n(9 comments)\n\n@Greg, I met you half way on the earlier ones. I wanted to try to get the topic closer to the start of the sentence.","accounts_in_message":[],"_revision_number":2},{"id":"2757519b516af2360c3cf0f385a7cd20eb9fc3d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-06-16 02:02:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b3fc101ffcb88b999858146a51390891bd187ffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 02:11:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd9a761ecb0449e9b24441a20d2f41a4 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133bc52b82b742a7abe000b6c42931aa : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":3},{"id":"a525c128b3c7cc0b08692034de92b78095b71e33","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-23 23:42:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)\n\nJust a few comments. Otherwise looks good, thanks!","accounts_in_message":[],"_revision_number":3},{"id":"1c844421810833f44f35db4e5d0897daf31f23ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-06-28 03:20:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6063dbdba5346a860920a3ccbeada9f7ee316ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 03:32:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e4527610954eda8098960529323cf6 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ed08111e6f44f080fd1657f98e1570 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":4},{"id":"6259289626a8684be0c3d5d7b32551abeb1c32ff","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-06-28 04:57:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"80a0e328769695589e82a0692223db285b7b903e","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2021-06-29 04:01:48.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"818681fee5fe941db7bf79a9f69c6df839f083d2","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-29 05:45:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0469c10f70d09b08c3c6bfcd371b5f527fc6a1a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-29 23:30:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":4},{"id":"d7123fdab83d1e4a6cf29e6d64e597de51a279d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 23:30:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9832b1e095e6aed406c9c375197f051525bd54b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 23:50:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94bf3f44421c403fa539baee8eb20660 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e4f31cedff49df81a494bc397618b0 : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":4},{"id":"e542e642449db2065e519c287a653f32ac35aad9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 23:50:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1c2696a0cb803d83b452462d2f3dcc0723eacd7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 23:51:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3104dd25383b47b79ec00a12b4f32db4 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0dcc1e0921dd171ae2fb6906119309df1134eadf","revisions":{"00aa11ea26e1ac3d45876480f4f44f7fb7d085c0":{"kind":"REWORK","_number":1,"created":"2021-05-27 08:55:48.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/25/793325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/793325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/793325/1"}}},"commit":{"parents":[{"commit":"290b8c068e1ebc11faf991df27438c91d4e94b38","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/290b8c068e1ebc11faf991df27438c91d4e94b38"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-27 07:11:14.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-27 08:55:39.000000000","tz":600},"subject":"[WIP] Improvements to zone blacklist doc","message":"[WIP] Improvements to zone blacklist doc\n\nReplace http api calls with openstack client commands.\nNote blacklist exception policy\n\nChange-Id: I7f3662c57ee9bccb42381134523e8fdd21e93740\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/00aa11ea26e1ac3d45876480f4f44f7fb7d085c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/00aa11ea26e1ac3d45876480f4f44f7fb7d085c0"}]},"branch":"refs/heads/master"},"98ee59b1cfc490050fe8a99fe599f433983bea02":{"kind":"REWORK","_number":2,"created":"2021-06-02 03:36:58.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/25/793325/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/793325/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/793325/2"}}},"commit":{"parents":[{"commit":"290b8c068e1ebc11faf991df27438c91d4e94b38","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/290b8c068e1ebc11faf991df27438c91d4e94b38"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-27 07:11:14.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-06-02 03:35:48.000000000","tz":600},"subject":"Improvements to zone blacklist doc","message":"Improvements to zone blacklist doc\n\nReplaced http api calls with openstack client commands.\nNote blacklist exception policy for admin users.\nChanged voicing to be more passive\nBlacklists only apply to zones, not records.\n\nChange-Id: I7f3662c57ee9bccb42381134523e8fdd21e93740\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98ee59b1cfc490050fe8a99fe599f433983bea02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98ee59b1cfc490050fe8a99fe599f433983bea02"}]},"branch":"refs/heads/master"},"fa5511a74314de7f95978928f887a3eb6d86faaf":{"kind":"REWORK","_number":3,"created":"2021-06-16 02:02:32.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/25/793325/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/793325/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/793325/3"}}},"commit":{"parents":[{"commit":"290b8c068e1ebc11faf991df27438c91d4e94b38","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/290b8c068e1ebc11faf991df27438c91d4e94b38"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-27 07:11:14.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-06-16 02:02:21.000000000","tz":600},"subject":"Improvements to zone blacklist doc","message":"Improvements to zone blacklist doc\n\nReplaced http api calls with openstack client commands.\nNote blacklist exception policy for admin users.\nChanged voicing to be more passive\nBlacklists only apply to zones, not records.\n\nChange-Id: I7f3662c57ee9bccb42381134523e8fdd21e93740\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa5511a74314de7f95978928f887a3eb6d86faaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa5511a74314de7f95978928f887a3eb6d86faaf"}]},"branch":"refs/heads/master"},"0dcc1e0921dd171ae2fb6906119309df1134eadf":{"kind":"REWORK","_number":4,"created":"2021-06-28 03:20:06.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/25/793325/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/793325/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/793325/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/793325/4"}}},"commit":{"parents":[{"commit":"290b8c068e1ebc11faf991df27438c91d4e94b38","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/290b8c068e1ebc11faf991df27438c91d4e94b38"}]}],"author":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-05-27 07:11:14.000000000","tz":600},"committer":{"name":"Michael Chapman","email":"woppin@gmail.com","date":"2021-06-28 03:19:57.000000000","tz":600},"subject":"Improvements to zone blacklist doc","message":"Improvements to zone blacklist doc\n\nReplaced http api calls with openstack client commands.\nNote blacklist exception policy for admin users.\nChanged voicing to be more passive\nBlacklists only apply to zones, not records.\n\nChange-Id: I7f3662c57ee9bccb42381134523e8fdd21e93740\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0dcc1e0921dd171ae2fb6906119309df1134eadf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0dcc1e0921dd171ae2fb6906119309df1134eadf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
