)]}'
{"id":"openstack%2Fdesignate~794131","triplet_id":"openstack%2Fdesignate~master~I5f7c1ef14daf0a01e4d37cc5416d08cc75f2b485","project":"openstack/designate","branch":"master","topic":"fix_testing","hashtags":[],"change_id":"I5f7c1ef14daf0a01e4d37cc5416d08cc75f2b485","subject":"Fixed a potential circular dependency","status":"MERGED","created":"2021-06-02 05:32:20.000000000","updated":"2021-06-16 22:29:34.000000000","submitted":"2021-06-16 22:28:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794131-fix_testing","meta_rev_id":"56a8fed3a2f160b26e72fa8527c19fabc830d599","_number":794131,"virtual_id_number":794131,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 22:28:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-06-16 21:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2021-06-16 21:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-16 21:18:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-02 06:45:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 21:08:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-16 21:17:20.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"6273c4bb1182444a328c0616de92b020ec34086d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-02 05:32:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d0551535657898b39e8731fa63f9edc77d48fa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 06:45:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b475aff019b74841b58d0c70e9fbea39 : SUCCESS in 41m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4cf49f43ea1a484c82f3e0bdc27267d0 : SUCCESS in 48m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cf223d31d144440c8144261185dd19e0 : SUCCESS in 1h 06m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a081a4ae0d234fe78ddd3ea56d176034 : SUCCESS in 42m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dccc381c626e486e84dae2a2726ff14b : SUCCESS in 41m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/596383ca3d33458ea44b6f2a29db2016 : SUCCESS in 6m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d357fd06ef6649e8989b750d72128bda : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f00eb6e7e354d8eb0248dad7e9a805c : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0b33e600eb54497aba8024bfc32f5ab : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b3fd7b1df8848cb88bf3d5d2e395b52 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ce7e1eb7257418e9dc1f6c879039160 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f92e53b8178d4d2383e132f997291361 : SUCCESS in 7m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/98cc39f404234e589e03791e9dd2e36b : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3dd702c10aba74c4c8839e85dbc8bfea98f8acbe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-16 21:08:40.000000000","message":"Patch Set 1: Code-Review+2\n\nYeah, I see the issue with the utils module and the python module loader problems with partial imports.\n\nReally, these seem like settings I would put in the designate.conf, but since they were historically setup this way, I\u0027m ok with this change.\n\nMaybe in the future we can refactor the need for this constant out.","accounts_in_message":[],"_revision_number":1},{"id":"cb4c23a7f293b180c0ea0163de432fe375e1884e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-16 21:13:31.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+2\n\u003e \n\u003e Yeah, I see the issue with the utils module and the python module loader problems with partial imports.\n\u003e \n\u003e Really, these seem like settings I would put in the designate.conf, but since they were historically setup this way, I\u0027m ok with this change.\n\u003e \n\u003e Maybe in the future we can refactor the need for this constant out.\n\nYep. Fully agree. I can probably follow up with a patch later to correct that. I wanted to keep the change minimal for now.","accounts_in_message":[],"_revision_number":1},{"id":"d610cdcde02b4d230963e246ba0faeb26666c213","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-06-16 21:17:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"922537b180f9bb95a74c76afa0594bb811a11b49","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-16 21:18:00.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c711527a1ca295f624877e6b363a6077a7e0d42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 21:18:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ed7df07546a8bfcab7d2479998671e67144c500","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 22:28:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9d99b4d8f16d4755ad88f31496e210a0 : SUCCESS in 37m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d20df7486ba9407db855c016dfb88277 : SUCCESS in 41m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/891bb660be584b07b11296e5968c47c2 : SUCCESS in 1h 04m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8ade22e0e70547d1b847453eae451162 : SUCCESS in 42m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/669771a1fbce430e8baba538e4d1178a : SUCCESS in 40m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b52b86b7c1214e45ae857bd8d781a552 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd624844e604dd892581d03cd371f41 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c21be7550f2047e9a1c2b2f13e80f28f : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90ebea387f274346be6c94297be2b441 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f214884f2f4a4ee98f237ec56e8ba8f0 : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b8a8e0de91776db932e967cda13c59a9a3681344","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 22:28:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"56a8fed3a2f160b26e72fa8527c19fabc830d599","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 22:29:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ab3154d5cd457aa2188dbb907726e8 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4da58a2438092eb0c1b84dd8259807cacd19b2ed","revisions":{"4da58a2438092eb0c1b84dd8259807cacd19b2ed":{"kind":"REWORK","_number":1,"created":"2021-06-02 05:32:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/31/794131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/794131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/794131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/794131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/794131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/794131/1"}}},"commit":{"parents":[{"commit":"5c60084f135621e987743ab51891cf2ae5f60f64","subject":"Moving to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c60084f135621e987743ab51891cf2ae5f60f64"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:30:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:32:12.000000000","tz":-420},"subject":"Fixed a potential circular dependency","message":"Fixed a potential circular dependency\n\nWhen running individual unit tests there was a possibility\nof triggering a circular dependency. This patch fixes\nthis by moving DEFAULT MDNS and AGENT ports into\nthe configuration.\n\nChange-Id: I5f7c1ef14daf0a01e4d37cc5416d08cc75f2b485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4da58a2438092eb0c1b84dd8259807cacd19b2ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4da58a2438092eb0c1b84dd8259807cacd19b2ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
