)]}'
{"id":"openstack%2Fdesignate~794135","triplet_id":"openstack%2Fdesignate~master~Ib55d16b2f05269ae58bbcf38e816fb776dbe4f3c","project":"openstack/designate","branch":"master","topic":"scheduler_tests","hashtags":[],"change_id":"Ib55d16b2f05269ae58bbcf38e816fb776dbe4f3c","subject":"Add simple scheduler permutation tests","status":"MERGED","created":"2021-06-02 05:45:20.000000000","updated":"2021-06-17 20:12:10.000000000","submitted":"2021-06-17 20:10:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"794135-scheduler_tests","meta_rev_id":"2f4b8b2fd1d37f799669e94dc561e408f0a640d4","_number":794135,"virtual_id_number":794135,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 20:10:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-17 18:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-06-17 18:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-02 07:25:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 22:18:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-17 06:01:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-06-17 17:41:59.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"582935842e7837147630e6a2215584bb4abea0b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-02 05:45:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b042f0fccb470196c434414d127f93ddc475c28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-02 05:46:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d523f58fc20f0b8e921e37933e20a08c9e1c9ea5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-02 06:03:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cf08ad346c78899ab1f8be701b1ae8631ddd63a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 07:25:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/87e041f54cc340ac966d18bbad4dbfe9 : SUCCESS in 43m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f14fd65753cc48d7aff02e93aea8ba00 : SUCCESS in 40m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/35aaf7807c2744b6a8c9686010b3fd9d : SUCCESS in 1h 05m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6268754e26d245779b15aae87d803e7a : SUCCESS in 43m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d2615b494784446396d8c5940a3dad7f : SUCCESS in 43m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2562ca376aab4c3180c7fadf4626bf04 : SUCCESS in 7m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c45b72641884b8e870d4e658dc4a67b : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b5798ec13c4bfdad39ff9fa8092056 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19c31c9eaeb049b69da5c1ed149d493f : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d79e416e41840509d5c83edc3cba3b9 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50dcd8fd6f9f4259ab478e5ad481333a : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2815c4eb0f847d385d41d0228834303 : SUCCESS in 10m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ec75c9fcb0f540229ec25be0d83fe7be : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c33897040e10b224af5a4c5062c031fdc4bf2aba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-16 21:18:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2282946cc084fcf13269d7666267c6d5b0264b7c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-16 22:18:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nJust a question about the mock in the default pool filter.","accounts_in_message":[],"_revision_number":4},{"id":"5cf3da5e891fafe69f1e71af8c9a006e0f7b7748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 22:20:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8771e2fc8304a4c828e2ef3dff6fb70 : SUCCESS in 38m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6d987ef517e14573a3bff7fb6622cf86 : SUCCESS in 34m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a3fcfc0d0bec497a813aa661963e9e7d : SUCCESS in 59m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/926ce3c0f6a64317b6c042a43b711d3b : SUCCESS in 35m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a13feed175ec4f4eb8240f6f88bac11a : SUCCESS in 36m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a8afb2d58aa474f84b134366594e537 : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a08a225385ad4a349aefcb7f435be5b0 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/364333e5b9a147f49f788acb9a4ee494 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1c2de68b64f4ae7a48848a7632da442 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd11c88b4b43438b9c25a9f926cebf8f : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b0fb9a204904d6693f35306505572bd : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d10f08093e74fcf9459341864940ccc : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03b8b3b3959d4df98b9fdc33e85723ae : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":4},{"id":"e2fb906bf384b1a71180afffd29a2512581bde4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-17 00:45:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e26162e38b65b161b62f0114f38b5b9ad7c734c5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-17 00:48:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aed5c2258ccac0196b20831e0464ddf41853f5fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-17 00:55:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b92d73e8f6b70d7f2065b590fea1d04316b890f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 02:15:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0642be2289cb4ec4a2606e38a82ca631 : SUCCESS in 45m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ba2babf357ab4140a9b042eb144f1d2a : SUCCESS in 31m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/400944aa5966446d85b6dc15148ce24b : SUCCESS in 1h 12m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4132d27533fb449d9af14bab319ee432 : SUCCESS in 42m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6d9cb268c902468380a61567d8c9684e : SUCCESS in 43m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5370e7290080411e9364de527e3c144c : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7591134c3c154309a6544d5dc2853bfa : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6469f0f53444729931e3bf91f0da230 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e379776b19b448289b0d1a88738beb5e : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ef36ff710f649f59662264c854eb4a2 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b1175724b5c49c19cbee0f13e95bb89 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c0d06a4a5254aff90b89944c91b9a85 : SUCCESS in 7m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/45ab824295af46b78756603dd1abe8d9 : SUCCESS in 32m 41s","accounts_in_message":[],"_revision_number":6},{"id":"b4160ee89d937c83d5fdd81ce34b530b2b601572","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-06-17 06:01:48.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)\n\nI like this, one idea for improvement inline","accounts_in_message":[],"_revision_number":6},{"id":"ca2e5fa252276a8edae34773004b8e8aa5f627f1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-17 17:36:39.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)\n\nIt is workable, but I think we have too many, different, default_pool_ids floating around here.","accounts_in_message":[],"_revision_number":6},{"id":"bf63000274ea5ec648bf7aa337afa08638b47452","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-06-17 17:41:59.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9b4884d366a0f2c65fac77ad40110437677ed135","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-06-17 17:43:39.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nBesides the suggested cleanup of the pool id I am happy with this change.","accounts_in_message":[],"_revision_number":6},{"id":"5c5d03ec9e04c13598d3b6c5c8a2bb63df115ae5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-17 17:57:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"35b8d31dc42f73ee7751e59e86e198413fe0d707","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-17 18:00:40.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"7641f7649e958e5204e74d05b8f6da083616818b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-17 18:39:34.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e4eeca55448a2603cb0d908f5f7a4e85ce9223e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 18:57:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e52989c0e0644268b0c4ed36bcc8c50c : SUCCESS in 35m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4c2310427ad14fa6ab294a2ae603334e : SUCCESS in 35m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5d428e05a8d0465fae94d439ac733a47 : SUCCESS in 58m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/094c2cc6b4674f65ac398998558980c3 : SUCCESS in 36m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0a21b6a3236b46d19db17c0e7b6b6a5b : SUCCESS in 35m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6282efbf3d348a08d6c98ea4647de36 : SUCCESS in 6m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31a70c4e54e9419a9fa7e368ccc2398b : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8e6f0ca1304fe2b8e19ec1739fccbb : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebd024a2cd9d4878bd56967c74841946 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02fdc8a2ca804fc6ac8ac047a34f95aa : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f940987a95c427ab9d6cf0b3ec3cf98 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03100d75a5754f6a9c755744c517649b : SUCCESS in 9m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eedea74ea980485c833fb4c0eb759a7a : SUCCESS in 27m 24s","accounts_in_message":[],"_revision_number":7},{"id":"b59eda08164ab9fb440f4401c5ea9fde8534a07c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 18:57:59.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"8d2fe663203e7887ba96af8bf3da5507dc16189f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 20:10:43.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4205965fd34640c29cc6082fd45e92c8 : SUCCESS in 35m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4518d69e052d47ffa188a29d22bf7d25 : SUCCESS in 41m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/64b69e06f1d44a58bf8210b0bc39fbec : SUCCESS in 1h 08m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d0eb94d4e4b04f248c9ab027896b2ec8 : SUCCESS in 41m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/57c28d1bdff3434092f19e8eb4991cba : SUCCESS in 43m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ed81d437ee548f3ad3e71b210906a53 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c45c33eb754e9a8f731b36dd56ee14 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2577ce864d3546999bf26e2d6503b119 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/508adc5e3a954409b4554c89daaf4d96 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ffb0fd62406458485574e7aa67847a9 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":7},{"id":"1080d3b12a668698f7effe26824bd6d1edbfc147","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 20:10:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"2f4b8b2fd1d37f799669e94dc561e408f0a640d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 20:12:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0a728076b545a5ba81ae555bdab94f : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c0bd7c7ff0fd56a12b41b545daa78e04c2355175","revisions":{"ad829c1565ee6522d42df3a54bc5d6d37a7ea212":{"kind":"REWORK","_number":1,"created":"2021-06-02 05:45:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/35/794135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/794135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/794135/1"}}},"commit":{"parents":[{"commit":"5c60084f135621e987743ab51891cf2ae5f60f64","subject":"Moving to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c60084f135621e987743ab51891cf2ae5f60f64"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:44:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:44:50.000000000","tz":-420},"subject":"Add simple scheduler permutation tests","message":"Add simple scheduler permutation tests\n\nChange-Id: Ib55d16b2f05269ae58bbcf38e816fb776dbe4f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad829c1565ee6522d42df3a54bc5d6d37a7ea212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad829c1565ee6522d42df3a54bc5d6d37a7ea212"}]},"branch":"refs/heads/master"},"e16b78dd6f8856f82bb77a7b78238a3587c5ead7":{"kind":"REWORK","_number":2,"created":"2021-06-02 05:46:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/35/794135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/794135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/794135/2"}}},"commit":{"parents":[{"commit":"5c60084f135621e987743ab51891cf2ae5f60f64","subject":"Moving to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c60084f135621e987743ab51891cf2ae5f60f64"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:44:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:45:56.000000000","tz":-420},"subject":"Add simple scheduler permutation tests","message":"Add simple scheduler permutation tests\n\nChange-Id: Ib55d16b2f05269ae58bbcf38e816fb776dbe4f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e16b78dd6f8856f82bb77a7b78238a3587c5ead7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e16b78dd6f8856f82bb77a7b78238a3587c5ead7"}]},"branch":"refs/heads/master"},"9240650649757b6e8fd34ae6ba72f6107bab3ea9":{"kind":"REWORK","_number":3,"created":"2021-06-02 06:03:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/35/794135/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/794135/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/794135/3"}}},"commit":{"parents":[{"commit":"5c60084f135621e987743ab51891cf2ae5f60f64","subject":"Moving to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c60084f135621e987743ab51891cf2ae5f60f64"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:44:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 06:02:43.000000000","tz":-420},"subject":"Add simple scheduler permutation tests","message":"Add simple scheduler permutation tests\n\nChange-Id: Ib55d16b2f05269ae58bbcf38e816fb776dbe4f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9240650649757b6e8fd34ae6ba72f6107bab3ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9240650649757b6e8fd34ae6ba72f6107bab3ea9"}]},"branch":"refs/heads/master"},"3ae1519d579c42f84dbe3efe534d7b963f1adc73":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-16 21:18:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/35/794135/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/794135/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/794135/4"}}},"commit":{"parents":[{"commit":"057dd2b3c6689be84dac58d6e5a19726d54735a0","subject":"Re-enable the tempest scenario jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/057dd2b3c6689be84dac58d6e5a19726d54735a0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:44:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-16 21:18:31.000000000","tz":0},"subject":"Add simple scheduler permutation tests","message":"Add simple scheduler permutation tests\n\nChange-Id: Ib55d16b2f05269ae58bbcf38e816fb776dbe4f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ae1519d579c42f84dbe3efe534d7b963f1adc73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ae1519d579c42f84dbe3efe534d7b963f1adc73"}]},"branch":"refs/heads/master","description":"Rebase"},"d59b45f162783da1fd3de2d5e80d8e4633f65e6d":{"kind":"REWORK","_number":5,"created":"2021-06-17 00:45:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/35/794135/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/794135/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/794135/5"}}},"commit":{"parents":[{"commit":"057dd2b3c6689be84dac58d6e5a19726d54735a0","subject":"Re-enable the tempest scenario jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/057dd2b3c6689be84dac58d6e5a19726d54735a0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:44:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-17 00:45:24.000000000","tz":-420},"subject":"Add simple scheduler permutation tests","message":"Add simple scheduler permutation tests\n\nChange-Id: Ib55d16b2f05269ae58bbcf38e816fb776dbe4f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d59b45f162783da1fd3de2d5e80d8e4633f65e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d59b45f162783da1fd3de2d5e80d8e4633f65e6d"}]},"branch":"refs/heads/master"},"7ea620e4c0f60b7a7007eecdbb5978af54babb88":{"kind":"REWORK","_number":6,"created":"2021-06-17 00:55:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/35/794135/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/794135/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/794135/6"}}},"commit":{"parents":[{"commit":"057dd2b3c6689be84dac58d6e5a19726d54735a0","subject":"Re-enable the tempest scenario jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/057dd2b3c6689be84dac58d6e5a19726d54735a0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:44:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-17 00:54:58.000000000","tz":-420},"subject":"Add simple scheduler permutation tests","message":"Add simple scheduler permutation tests\n\nChange-Id: Ib55d16b2f05269ae58bbcf38e816fb776dbe4f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ea620e4c0f60b7a7007eecdbb5978af54babb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ea620e4c0f60b7a7007eecdbb5978af54babb88"}]},"branch":"refs/heads/master"},"c0bd7c7ff0fd56a12b41b545daa78e04c2355175":{"kind":"REWORK","_number":7,"created":"2021-06-17 17:57:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/35/794135/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/794135/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/794135/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/794135/7"}}},"commit":{"parents":[{"commit":"057dd2b3c6689be84dac58d6e5a19726d54735a0","subject":"Re-enable the tempest scenario jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/057dd2b3c6689be84dac58d6e5a19726d54735a0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-02 05:44:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-17 17:57:23.000000000","tz":-420},"subject":"Add simple scheduler permutation tests","message":"Add simple scheduler permutation tests\n\nChange-Id: Ib55d16b2f05269ae58bbcf38e816fb776dbe4f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c0bd7c7ff0fd56a12b41b545daa78e04c2355175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c0bd7c7ff0fd56a12b41b545daa78e04c2355175"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
