)]}'
{"id":"openstack%2Fdesignate~794412","triplet_id":"openstack%2Fdesignate~master~Ic8c9c7bbd2ad21e5b7ee111bc0141146b88d0eda","project":"openstack/designate","branch":"master","topic":"ci_stability_test","hashtags":[],"change_id":"Ic8c9c7bbd2ad21e5b7ee111bc0141146b88d0eda","subject":"[DNM] CI stability test","status":"ABANDONED","created":"2021-06-03 07:26:26.000000000","updated":"2021-06-13 23:40:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f35b1fef57778eed3e1831de9707d4bddbee4e3b","_number":794412,"virtual_id_number":794412,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-03 08:22:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 08:22:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"788e2f82d4d5b22509433c083ce96a11ee41e3a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-03 07:26:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7f96499455f5d16a609ed5c9659a2a57a312976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 08:22:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e9c019b96912479784e41e8890c4ebe4 : SUCCESS in 30m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/572d825cdb0445d3951e6b943b2f7a01 : SUCCESS in 37m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c96afe9f24ca44d087a1c32dc04c7490 : SUCCESS in 53m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a18b27f07ada4be3b101d4c944fe43d6 : FAILURE in 37m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/05a620e8a86d45dc9040e16a22912a96 : FAILURE in 34m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b220552abb69464db08918f123068f24 : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf2886c55a9f425e84279b0f5dccb6a3 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e683040b61db4b1f9633829a52fde09a : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/865b2a301a7a4d278bb393ff18ad81cc : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/142f1d23c4274688bcc25f69cda8688d : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/143e154b60e1460a8702781d8557de33 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4151b1071ef4f85b5b0d957d846ba00 : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/269abeb768e342c6832b7ce7d8c8bd33 : SUCCESS in 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3d30142cac66ed40bdd3e6170495f8a3fdae84ae","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-03 17:12:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c383d8dafb4beb69371248fa4a1cffc6e5a94680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 18:38:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6759b9d3f1fb4185b854f3c16d2e009b : FAILURE in 36m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/318e42c978694547ba1b9e37c6893137 : SUCCESS in 37m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f723c9e6d79b486d906fc591baa0589a : SUCCESS in 1h 03m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4f7b700ba08346c8887bf5c82170d6b0 : FAILURE in 38m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d016a25875e44649831fe493e0455171 : FAILURE in 42m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2441468135b4fb6a100c508ba6b7935 : SUCCESS in 6m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/403751889a5f4953afb704ce4cc1bbc4 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c27a415cef4906ac5ce20734744b74 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/525af1f0b7af44bd91b03d37e7c01ead : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/966421f3e82e49aea683acfc488a6eb0 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3762ee941873478faf9556c01161bb0c : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7684a7e39410437895befb99ce894085 : SUCCESS in 10m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c04664ce44694a11872d99cd2ec33b7d : SUCCESS in 27m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f35b1fef57778eed3e1831de9707d4bddbee4e3b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-13 23:40:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82b437e55c2bf170a170af5cd88e67969a47ec18","revisions":{"82b437e55c2bf170a170af5cd88e67969a47ec18":{"kind":"REWORK","_number":1,"created":"2021-06-03 07:26:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/12/794412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/794412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/794412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/794412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/794412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/794412/1"}}},"commit":{"parents":[{"commit":"5c60084f135621e987743ab51891cf2ae5f60f64","subject":"Moving to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c60084f135621e987743ab51891cf2ae5f60f64"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-03 07:26:08.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-03 07:26:08.000000000","tz":-420},"subject":"[DNM] CI stability test","message":"[DNM] CI stability test\n\nChange-Id: Ic8c9c7bbd2ad21e5b7ee111bc0141146b88d0eda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82b437e55c2bf170a170af5cd88e67969a47ec18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82b437e55c2bf170a170af5cd88e67969a47ec18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
