)]}'
{"id":"openstack%2Fdesignate~795906","triplet_id":"openstack%2Fdesignate~master~I4fee32f9be080eea6eef38fcecbdb2dd3940b9e6","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I4fee32f9be080eea6eef38fcecbdb2dd3940b9e6","subject":"Fix migration for sqlalchemy 1.4","status":"MERGED","created":"2021-06-10 22:34:08.000000000","updated":"2021-06-14 23:59:01.000000000","submitted":"2021-06-14 23:57:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795906","meta_rev_id":"dd234dde16bf0bb85ba909443a931aebcde33ba2","_number":795906,"virtual_id_number":795906,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-14 23:57:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-06-14 23:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-14 22:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-14 23:41:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-14 22:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"CC":[{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 23:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-11 01:29:34.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2021-06-11 06:10:18.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2021-06-11 15:52:32.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-06-11 16:39:14.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-06-13 07:19:56.000000000","updated_by":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"CC"}],"messages":[{"id":"6a37274f98697f0e2f8c92dedb9a0d99bda125bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-10 22:34:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e132c03c8b1307fd2ef97cb6f3b88ac5a5ea544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-10 22:53:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"51d526b14bc0e552f247d403e7b7c6f85bf29589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 23:48:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/163c426b42c84c148e2357e88697bf50 : SUCCESS in 35m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d63e5b0ecb244daaa2ff841048ca0364 : FAILURE in 46m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d3157b5046824fcc9ffab645b85d5909 : SUCCESS in 53m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a707abf209ea419cb876ffb802810f52 : SUCCESS in 33m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6d1d8e38376f467f9df2d6f814b375aa : SUCCESS in 35m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2985a89f4fa4a36b4b8eeee02b74a73 : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5053f85d08847f5ae7e9bc0dfbb997d : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42b50a7bec53454fabe84f42bc6a9a59 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e77a97c6f2e4a90a4ff54bb11e94ed1 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65f897dfe56049e6b3bbe846c07b4561 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc31542ae528465a962b62c9b76608fa : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d1bd8c69a54d629afee7691f8c2f9d : SUCCESS in 7m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/71e9ec690d964038a2106cfdd87584c1 : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":2},{"id":"818d05622d3a3522e5b34eb8d41c40b285e35959","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-11 01:29:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0d7042fec6960a7e33e0aa2aa79564d316857de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 03:14:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/35a235d800884ffcbc534dd895239cc1 : SUCCESS in 38m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5503b1fd9b11420f9689b7aa379dfefa : SUCCESS in 35m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4dcf459e75044e4b9c6a8d5f5d31fc6a : SUCCESS in 1h 02m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7daa6a7b18f34ef19e877bf865de2f89 : SUCCESS in 36m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6506dcededcc42d088ca5cded456c48f : FAILURE in 37m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd3d6067ec76427aa8c4d48d3a3c6e8f : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9c60b0231f0445680d78547c1e13184 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02c1102c2d54a30a55c914bf9f9fe2e : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19d2c9a2ee6e4dea9e0f4373b35407f3 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45f53543452049cf93b38d9559f0a40a : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e358a277d1be4d70b57a283712cc61bc : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae48d3c510d34f7a932f86b3d2e564e2 : SUCCESS in 8m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d8e14ceaab24021beb304a092f7b4ac : SUCCESS in 28m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c29bf0027b8df56b22035058469e42cc10e38df2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-11 04:10:47.000000000","message":"Patch Set 2:\n\nrecheck DesignatePtrRecord","accounts_in_message":[],"_revision_number":2},{"id":"ad5e22e61e40ddb74f825ef41ba763e8124caff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 05:20:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/10a1dbed0e6148a9a294ff4fd5eb58eb : SUCCESS in 41m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/babd10cd8fc54828a88d7af9a859446a : FAILURE in 36m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/32cf30a3c5c3481680e9a3b3cbf3def6 : SUCCESS in 1h 05m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dbccff211155446885d85b8e54252b2a : SUCCESS in 38m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bb8fb54a45d74de5869c9e0b6f8b5e83 : SUCCESS in 39m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b96cbed34bfa468a94802abf19873a7a : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa402732792a49cfae223c0b374985ac : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c2239a6e134d1082ec9f68a7163c68 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a4d119849ec4fd6b900534d49f17c89 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a12f50b4505a4ac4bf6b8e4d65f72761 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/135ced0513ad431bb0b708f8e7b09d5c : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a23212097f543c499fa0a2b95df9c1e : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0fb932d9bf5f4966ba25ea7cd629b5fe : SUCCESS in 29m 38s","accounts_in_message":[],"_revision_number":2},{"id":"8933ee59c4e057cefdb4faef04705c1b76861511","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-11 06:10:18.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck DesignatePtrRecord","accounts_in_message":[],"_revision_number":2},{"id":"8e6a6e1e0c9775307dd0e2b0e44462c426c83a4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 07:19:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dded70d2b3a5477bb5d495b3bde7e3c4 : SUCCESS in 36m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/37ecd49f19f6416d9ee20dc3158cd0fe : SUCCESS in 40m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d1b301060dce467f9f0bb91de136e0cd : SUCCESS in 1h 03m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/740775001d1543b88eb76e41ef57ed08 : FAILURE in 37m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4af674b7f5934466b1430d9d7f1d6644 : FAILURE in 43m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc2956d9d9c3436682fc1e1de160b4c9 : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c08fe30f274499da06f355ccef7c9fa : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641cd05045654afc852f03ff0b6a8e77 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4f58581df34414c93287ab70d3b8e36 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b8b649601504a16ba5db130a81916e6 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7ea59a348504ec783f0b2a5e4504238 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e1bea6ef06475ea9529da1ac2eda66 : SUCCESS in 8m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/086e9dc33e8d4f85aa810c91a1ba2fd8 : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":2},{"id":"27f800f336765b653586b7b49867c8c21fc651bf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-11 07:33:20.000000000","message":"Patch Set 2:\n\nrecheck RecordsetOwnershipTest","accounts_in_message":[],"_revision_number":2},{"id":"c8c51d3d8ccabcbc5c981d1b1e9a696dba995aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 08:45:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/582da79c78664c8496907039cf2bc604 : SUCCESS in 34m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6dd814c115f34b7f894fd4a2b6cba85b : SUCCESS in 39m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4ff99ed36b5c4874a22723cea4efb542 : SUCCESS in 1h 08m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6bb83f55f45842fda528f6c3e6777795 : SUCCESS in 37m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ac3b093e477d4b6dbfe7d30ef3f78e68 : SUCCESS in 38m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e15df1ebf2d942fd87a700ea72f3f38b : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/278e3a341ae8477286b5d3e6014cb1a3 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/300cb276305e4bd7b94dca0098ca00b9 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3f167dd4337428daebea6b22224ee3a : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e541525c632048929a3c829ee80d9475 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dda1d5d07264ee68e8e724d522e720f : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21fd4dde2c014a13a1bdf30cf4885ce6 : SUCCESS in 7m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/85d47ee006b24a88b6bf2328a74f2104 : SUCCESS in 32m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9414ca021eb6d56a0971fbf32d40c5515c79a610","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-11 15:52:32.000000000","message":"Patch Set 2: Code-Review+1\n\nverified this works in https://review.opendev.org/c/x/ansible-role-refstack-client/+/795739","accounts_in_message":[],"_revision_number":2},{"id":"822e4985773c9749b499fddcf8bbf6d3c71b688f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-06-11 16:39:14.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for fixing!","accounts_in_message":[],"_revision_number":2},{"id":"cbbbd0b9bf66f72bd281a1c1fc062fcc7c341667","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-11 19:49:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"78879a52d2d032d953a1593228336f38af6696a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 19:49:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e408125a9e1338dafaeb0d136d4e5661dc5720b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 20:46:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7966a3213b174a4ab0b8e8f421a0acfb : SUCCESS in 40m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f4732ccca3b0417b93bcd47842c4aaa5 : SUCCESS in 38m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/32313792f311422c9b35a235bf50f6f6 : FAILURE in 55m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7d53e3aa9400470ba8e22c4e372c9dc3 : SUCCESS in 43m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e9b3551121c9443b9f85a1c6a21bf239 : SUCCESS in 39m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a31ed2c60f24bb192fa7e2335c482e6 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5ec0efa9fe4fedafe6fcd2428fb1e5 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d81b66112c2b40c98c55e7b3987f15c8 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6ae88d8823c42108f205ca678b0bcf3 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2a73e7867ee447ca5b820d0005b7a4f : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":2},{"id":"ab3ef34ca4ab92afc4968b702c4e74c7ebeeef63","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-11 21:31:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"801f7391d151fc2531581a473386effd47c10401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 23:12:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ab5575e28b3c471aa115b7fc3d685461 : FAILURE in 45m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ce4e8b75bdf748439f17226e67ba08ab : SUCCESS in 43m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/751412379af74cd0be65767afe268c0c : FAILURE in 1h 19m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e16e16e3c1484ab1b653f96fb8137058 : SUCCESS in 39m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dc6ecd8f27924a2fbfceae1843df9d53 : SUCCESS in 40m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72ed141cbf774506b168a58b5887ff3b : SUCCESS in 6m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c84299a77a444befa2228d54e6c50f38 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2015309b1f14924a7bf72f67ff199a1 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d963865fd1c04e1fbef1911211b86730 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df554feb77834ffab81f893abc3a88ce : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a836c22da4de4c2ea460a3f905dc3214 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8582b347718141daa766c570bd80aae0 : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc71880d5cc341cda957b45f35278d43 : SUCCESS in 31m 44s","accounts_in_message":[],"_revision_number":2},{"id":"b2e0cb1e3e1c4a869f56b2585ea9f2ecf8158681","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-11 23:51:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"04cb9fb20102d5d4cc1329aa17510e855d3ef1f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 00:57:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c7357a33d8b04d64bc972218e3b5cbca : SUCCESS in 31m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5e5e798e939645e6a7970498e0a69946 : SUCCESS in 56m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/024b044f8e854e878e4db52d1cfe1820 : SUCCESS in 49m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/485538ad28004affb2043cf543dd9334 : POST_FAILURE in 57m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e0c846bd196f4f0a95c4808484e49d80 : SUCCESS in 31m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53acdb126f65429f836594f42791869f : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6998a22bdb0c4a83ad221ab16d562884 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c3874e3d2134c17a7b6e291915a37e2 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2b60e3a1be84f8c89b287d60fd11ad9 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed8310762cc54984a904f8c9f357f22b : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2e01d4b90be4c3b87cce2837786b30f : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/195fe5c013e945fb96f9b3e2cf18846c : SUCCESS in 10m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/da6424effb9f4a26ae362de4aadf06dd : SUCCESS in 24m 43s","accounts_in_message":[],"_revision_number":2},{"id":"648d04783ac609674b7409e33a82a5bdda00a749","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-12 07:13:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"32b488f93710b11f82cf37eac96b7aa2356ec4f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 08:13:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4a1abb4b1cd5419fa6aa74ec57a3b53f : SUCCESS in 30m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/703db348e9a147e6beabfe964d36d488 : FAILURE in 40m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/788bdc50b48045ca957ccf1adc715b5c : FAILURE in 59m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9b2bf55c95314013835a579d33caf312 : SUCCESS in 39m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1edb7f2799b548e3bc5d96a6b10f372b : FAILURE in 38m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acde1ea8a0a8457db0a451896ca4d38a : SUCCESS in 7m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8d8d9661b0f4219b70e4323752e5059 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e6d55cc83b14adabea7af86e1e6fb6c : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33922639f9e049adab19630db2766ad6 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a02a7dea9a84251ad12bfaec5a3e5c3 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfc33736d3d84b149330c78845b0cbe2 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d93ae112aba4788894f6798bf013d38 : SUCCESS in 9m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a47a33c19373483ab65e380a2e82702a : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b39b633d66bd4b177eb57625b5fd56004d223d21","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-12 08:31:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da2212a4e0cd4cf9b5185dcfe25e05e8e20c26cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 09:45:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1e96144242a44f34929f52142599a307 : SUCCESS in 33m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c8ee5809866145b186398e7ce9db6f7e : SUCCESS in 35m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/810116c2f3d346eaa766404218d004e4 : SUCCESS in 1h 02m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8365749c7c6d4bb79ecbf14629c8285f : FAILURE in 29m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/59305cbabe0449c98626758aab927aa1 : SUCCESS in 40m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f34df5d7cefb45f0a748096db7f23a85 : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9a79152ef6a4510a95fd7555d72e862 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5406053d4844bbb11ede86fc80dd35 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d74063d6bd894a4e837908cdaa9475a0 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/412cc1626ec842b1bf8a3e9cf7808e48 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb7d47869dc843f9be15b9ce1f6f1c84 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2767d3013914949aba40ab2d2ffc03f : SUCCESS in 6m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ab1665088cb84be4980a647302c9d032 : SUCCESS in 22m 54s","accounts_in_message":[],"_revision_number":2},{"id":"894a7bf1e53919ade03914137536bde366864860","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-06-13 07:19:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"56596bbaef5b1eab7b82e11f57d24097877a2a40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-13 08:17:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b18989048e3e49dd81126fc73edba1cf : SUCCESS in 32m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/257911070eaf4abcb7bb93d54d6eb491 : SUCCESS in 29m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/af24c77d44214099a7dc285e7e6f6674 : FAILURE in 54m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/46017b612168425cbd3a939810068c3b : FAILURE in 30m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d5aab430e61445499b19dad2f721c1c2 : FAILURE in 37m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5941e6c3f243498a80e30e7519e51271 : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97a1be229dd444bda9b7501d3af3d9d9 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad569d74b1f4db2a70289e1d83a9a52 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db3dc68e4cca42bfb44ed4a82330fde8 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/689913a767db4c998c04b7b0d5b1f628 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fce4aaca44d54a4996c7c8d24b2d7ebf : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2683c7bccf74a479cd209fff84af330 : SUCCESS in 7m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/11560ec5258a44f4b7affd33ef069a1b : SUCCESS in 23m 42s","accounts_in_message":[],"_revision_number":2},{"id":"0bd2adcf15f1678321eb3412ef7cc154527f77f0","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-06-13 08:21:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f070dadc5035e3ceb839fd06b95ac4290d0e85b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-13 09:13:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f181e45f83284c00bfc6157f9ae2a856 : SUCCESS in 32m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e16e812cc5524d01b3cfb0dca9f0fc53 : FAILURE in 34m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/58a90eeaa56a488d9086de6ecb6bd78f : SUCCESS in 51m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fd73163acda049599ee90d293d59e3ad : FAILURE in 30m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a2d4393808944d33a1a9f3e0b641c607 : SUCCESS in 31m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/526cb728ff54477287878d26c47b4872 : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95c0a37c954d443a869f68c113834f94 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a98973434e2b4d4db86d65e34c50d32d : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b732f315206a4b5ca939fe5c123864b9 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b11d968680da47079cd866602e73f6ae : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8197459398fa47a2a865c463876c3876 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3d7e8d5246245a1a009c864f42b402e : SUCCESS in 8m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4cc1c80b0bfa40cc837e4e39ccd7a58c : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":2},{"id":"cd6f34e9f22ef546f707721d1fbac561a2694aa2","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-06-13 09:18:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6a01bbd88a7ae17c09a004644cf062336699b4b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-13 10:17:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/83ae4fd145584f0f80b0b0b48b2e550d : SUCCESS in 40m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/51f2c0fc81584e2bb359df3bbc2666c3 : SUCCESS in 34m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/68848bae49b84443b284e9297e9aa18b : FAILURE in 57m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/00d33bbef98840fd9cc68d519e457648 : SUCCESS in 37m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aaf315d037af4ee9b84534a8953fae94 : FAILURE in 39m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92a2fe27e5b54bdaa8be899fdc46ff33 : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5553a6ed01f444b299798e1348a70201 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408b92374c904d36a3925d80f38fcd52 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f874d03efd2490da759bee571f98f42 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7e81d631dfc4184ab36f1f7ec57a791 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1cb38997d4a4daa92511aa8d40c88a6 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f28f83f11d4a49b7cf266e8bc2bb6f : SUCCESS in 8m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/587e7e39a27b4b10a32539963622c800 : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":2},{"id":"59618e06999975b4f7ff1eec61174c121ee2933d","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-06-13 11:23:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c57baabfbc01e487e3b620557feba6daa8c188fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-13 12:14:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/24ae9c8d9d1f4e43be44eb50af891a39 : SUCCESS in 37m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4af629bbad474e55a5ef0ae7b93a36c8 : SUCCESS in 31m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3fd01ca329944dd39205f709c9e85eca : FAILURE in 50m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1db1a1f6a2a14b8d90b4ef8f46992c79 : SUCCESS in 33m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c359121c6c97402595c44483658e0ed1 : SUCCESS in 32m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88a87cf06b084d0ea90ba49e20e71cf9 : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de9be9fcce594cc1a7c61c7defb55fda : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f8cd86ed5434ec6a18aaa8251495463 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1155dcba76a84d1db90f68ef0286c056 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82732f2961534c51929224d06a2abb16 : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93727384bb1c40c6bf7e9e8c95bedef9 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792d9487f69a412f9cbfa0711cfdf884 : SUCCESS in 6m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f3f0deccbf644e03a833acc151dceaef : SUCCESS in 22m 47s","accounts_in_message":[],"_revision_number":2},{"id":"17a2443ed6ee42c96cc252ca2adf65824a0c2d40","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-13 17:25:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4827f2bc2653394353d8c5e30f6385879ca2b2fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-13 18:39:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7aff9ed56a234c16b244ffab1a690c6f : TIMED_OUT in 1h 12m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d704f00e9ea846fc8a81892a2bc7b0fc : SUCCESS in 34m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6a83356b278743f7bfe198b5a3375af6 : SUCCESS in 58m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8cd2b0080d3e44789eb908607eaa8216 : SUCCESS in 40m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9a8bf136077247e19ff43d1a26c540a5 : SUCCESS in 36m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7aa600a98c8c4f7ca4a8c9a3becae628 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c7ad8c2202549878ba02759c9c1377c : SUCCESS in 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6e008d66654f75a00ccdd3aab6fef5 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb4ca413f0fb4d169296c2c7b7fa7c60 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a3c131328444592a65ee2b03bcb23df : FAILURE in 16m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6658263032d146368fd2445bf2a5d15e : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1588de7262244ff78a93569589f5c5b2 : SUCCESS in 10m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f22e691595934b478ebd92b0d1602417 : POST_FAILURE in 29m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9899931a535c7e105dee350e2e2ad2bc5f10aab5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-13 23:52:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"164f1e6d632af85a020faaab77a063a4cf19e67f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 01:06:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c22a0c66489a4d6cb9e68f1674a5afc1 : SUCCESS in 33m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2d38ed84a6bd4aea8f3547a920c18c3d : SUCCESS in 49m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bcb0482427e84dafb0fd34615753cca8 : SUCCESS in 1h 09m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f6c5a45e767a492f9970d379325f92d3 : SUCCESS in 43m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8b8e4f2f986044278cad112edbe12b80 : SUCCESS in 31m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e36f2392f0404d9ea7f5f884c2334c1f : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27fd52da81d548dd924984e6f1b82fcd : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f92417b10e4f0c99fd044de4f09e4e : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ec6c7e2be53489ea9ebb6bd8bb7ea8e : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/052adfda200c4e48be549bebed230ef6 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/493ed254e8dc45c78cc83c844e1e9863 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1b6730aeba4b7da4ad7c2f2e26e592 : SUCCESS in 7m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7b2c86f96da5464483fc3094b7fcf33e : SUCCESS in 31m 56s","accounts_in_message":[],"_revision_number":2},{"id":"eabc5721fa6cfb6fdb106c019289859e66b8b6fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 01:06:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36529cf947c4445a43b7d221cd86b3be30cb0cea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 02:11:02.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/975d11da01ea4e8e9d877229af51ee53 : SUCCESS in 40m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/635b19d9365347698a2338b081e2b002 : SUCCESS in 44m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/49426af525de4e6386b3b6b3080d7500 : FAILURE in 1h 00m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/53a771e7baaa4b448ddeb9d15be132c7 : SUCCESS in 47m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a55d1a7205a946a98d52aa3556bc8755 : SUCCESS in 40m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21cac360bc81401482a942e61c49e76d : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e3a1e9166b44387842002a8598b500f : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b3b832a6f2e401b83838d6a1039e843 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c7256f6d0e049d2a97005ad2b40f493 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a1791931214585b273d8a3a71c5594 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a081241e6828c61cf4712a09ecf7298ee8fb9520","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-14 02:15:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"30ce85da0467ae75783fe1580d0a2b05dcfbd106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 03:10:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/68ea23f5ec90442484d7ec49e84875c5 : SUCCESS in 29m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/46f875e4a9df4f3fb56ae9a5ad5012a7 : SUCCESS in 30m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d3c27faf73c54bcca55cf332e2130a7f : FAILURE in 54m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8165ff8e54c44083bb771a4c409e1b1a : SUCCESS in 40m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aee554870826413dbabd6753d8511c64 : FAILURE in 40m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b7eff2ce7f74622b1aa46f960c506a5 : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95d2b6bb013c4c39ae28cf508b89e751 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b1d7bf831b041d0bf3e7babff4e8ad4 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6caebee3f323437c81f8e41aabb6ecd6 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a948bc8e2e94f7c845468224f4e6503 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd67a2194eeb4a30b7a9d1af85e5f295 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32bf79e5bd5b4c99a9b06743a603e029 : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e37d7ff01aa54d34919953f469cb0a44 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":2},{"id":"800afb0dbf935affcc8ecde0143d9d6c471ab576","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-14 03:12:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a15ffd5be4e93b7237fd0e40e39aec6e94631f00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 04:12:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/423a0b2f0c22466fb6fbefd659c6773f : SUCCESS in 38m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c6468c742c214cd68418512a9f428259 : SUCCESS in 37m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/23ade911e7b348b78b51b5833dbdc370 : SUCCESS in 58m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/46ceceba1df74c5fb6db20874f9c0389 : SUCCESS in 42m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/59b782dbbffa4103a3d27968c382c270 : SUCCESS in 37m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af1ecd465dc247a8b4b1cfcfb88fd4fc : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86f8e2382e064119a358c39af1d5b6cf : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aeaa5bac8694b589de8712eaf8ff8b7 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e37e785cd8e4aff95bbff97a9a5f1af : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d33a47ef4774b0abadeed019f1c32c0 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e84417248884ccea1e466b9611d8a14 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca7e547bcba401bbe8ca25e405e8c70 : SUCCESS in 8m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad1709cfbc8348ef984c1c1e13bccfab : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a31d60e8eed34bbe09a29f14bb1abb03e26e852c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 04:12:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb03afe564c01c540bbe002a3b26e47a57b086f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 05:10:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bee38282cb5d4b58849675fc4f6579e8 : SUCCESS in 36m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3eb05353201041bba0814e42bdecec37 : SUCCESS in 50m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a28f263fb548405089bddb1417c49c81 : FAILURE in 55m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2bbe452d19e14df3bfc28d81ba62f7d5 : FAILURE in 56m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2bc899faa38a428aab73a53cc86c653c : FAILURE in 36m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4ecce7abc2b453da4adb66ed4b666f2 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a0713f8002494ab543da70a48a07b5 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e8c86767921481d9fe1874b1999993d : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f7f87f76c6c48f1a9521d397731ec9a : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665e01b964df401fa5e0ae2a7abeea59 : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c28dbfa216203c8240fac78b365122a6744bb434","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-14 05:23:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6d731f4b7e3c0cc514bfc7de4003a9184da43e9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 06:32:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/faa9e300c2e141038060a639cbb342ff : SUCCESS in 43m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c18d268406d3472b963a7b1262e26934 : FAILURE in 39m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d2d2d484b8b84356aa0849005c7db4b1 : SUCCESS in 1h 06m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c6acb2d0242f47b5b4e54b56bc36a2bb : SUCCESS in 40m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1705e5f56e0f4947b05b73e687c28000 : SUCCESS in 37m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c39fabf9a9348eb80bac00bbb4ccf52 : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27d2b67623a44d10bc6e3f6fdb43a585 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d6d3850f87c47d3978296e11c5f48a8 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4077e660ebf14c2491e02b6f063d0a28 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0170da90c9834ae09d52db1759730b5e : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93d3bf2f19c540ac96eeb4e1e074c81f : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/886425acbab349d8922694c16a9e2eca : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/45354b4510fa4692a99b8e48be3c30b0 : SUCCESS in 26m 43s","accounts_in_message":[],"_revision_number":2},{"id":"aaedcd6b0c11c4c0afb8dd802217dbff6ff66f74","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-06-14 07:03:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c342d6f3c8e1bcad7e152339d38084901fe92201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 08:14:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3855697b03c64b219db340f557513966 : SUCCESS in 39m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b94a72cd176a4ec4972d05646011805e : FAILURE in 42m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a862164853f942aea6ded530ffef7d2a : FAILURE in 1h 04m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b77f7309110d4d2bbaff8d665c399b7a : SUCCESS in 42m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/67a87295d9184b3eac346e6c08744d55 : FAILURE in 40m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9ff9bd9a8b749bab0468b42764a9096 : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/397e85af1f6c44bd9a637f84d59e1072 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a63524eb8924450af20024be96fb938 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d6edf7ac4b742ed9d0620cff092e9e3 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0ab7ee1dc9148d4b16d907094defd78 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fb13bfe610b483882d4606a82a64216 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f366df8e658e4c5a9e85613cd7454e13 : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/60b1cf7973624514ab39eaae7dc247ff : SUCCESS in 33m 03s","accounts_in_message":[],"_revision_number":2},{"id":"20dd2609b56573bb7d83e4c12f7c258dd696bb34","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-14 08:49:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f959f60a7da361dcbb93df60eb80d769e9e4e6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 09:58:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8fab089c46d44d1ba4eada109f4c84ab : SUCCESS in 41m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8c8f8b261d964fc0af48c4928ea1ac1d : FAILURE in 42m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/10d88f10228b4ed0b3706d51513e832d : SUCCESS in 1h 03m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/953fc166499344aaac7ec36cd9828233 : SUCCESS in 40m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a2a1401204834827bdf673d938ef0425 : SUCCESS in 42m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adb37d81d23e430ebebc7772bab8464b : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a612df9fffe5456cb1e0fbc9f6c4eaec : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56b2545d456e4f3b817f0586419aa23f : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b131f1d345a842c2bdde95fb77366061 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6c82d7f87344a679dd0dbc6c7e46f02 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa13164180234202a1a1853e9ce58105 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb89d70a5d04e5fa298fe7a489921d1 : SUCCESS in 8m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/34d4e2b3c06a420fab40c306adabf758 : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":2},{"id":"12d92e6f9440d8774ae28e78d84f27eb20e588f4","author":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"date":"2021-06-14 11:21:49.000000000","message":"Patch Set 2:\n\nThis patch fails because: https://bugs.launchpad.net/designate/+bug/1931795\nAlso, there is a patch (designate tempest) that is coming to skip the failed test:\nhttps://review.opendev.org/c/openstack/designate-tempest-plugin/+/794708\n\nUnfortunately we have a kind of dependencies loop here, as both patches are tied and each one of them is preventing the other one to be merged.\n\nIt seems to me that we need to get this patch merged first, probably by disabling \"Failed\" gates if possible.","accounts_in_message":[],"_revision_number":2},{"id":"a326cf25094e1d0795095563ceb2c7c5cc81ffec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-14 21:45:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"30785359adef09edf7ec21cfb173d22029291757","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-14 22:07:51.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"8bedba050ba658bf276eb63680c0acb0a9a99fdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 22:26:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22d05a0e20034ea785d1493457fc20cc : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a192067609645e7960f8b0a5be461fb : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750c206b1fe8471ba0ad84723867ef1d : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/026daaec3aa148f0a1e5db0ff3ffdea4 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dca55307a629496ca4a71ce3edd1151a : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f94fbe180334477ca6dd75682de0c31b : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a414dbb18a0c4b308d77b5446cf7939c : SUCCESS in 8m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ddc5f96b48a44f99ac92765272bd06e : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":3},{"id":"7887266646f6f8333e61d7959e93391590f8bd70","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-06-14 23:10:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e6d82e38409050b5f2cd9290fc0b0438702a26c9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-14 23:41:10.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8ec5aa2124a6faf80afea0b43d158a7daef78135","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 23:41:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e0094e6680e675ce6479d836c3b24051a6536b4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 23:57:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3012b0f0a06d4f80aed4d787aa5d07e5 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca15bd7a2260408c8ac654888df4c528 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e34ccc5ae0644b7f8ad5b412228a3ac4 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef857123abb5410fb6e61f8a11cac351 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593197a305bc4b758df602d5172a5511 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":3},{"id":"5a53fe8577cc7d475790bf8765b1f002dc04d557","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 23:57:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dd234dde16bf0bb85ba909443a931aebcde33ba2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 23:59:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b10d0fa79b014d00bc2c039eec5bfd0b : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fba57ab1caae768b7ca5c73a0db9e3e0ba241533","revisions":{"f40bb13c33dab5acb504194d4c7aa66546fd8f68":{"kind":"REWORK","_number":1,"created":"2021-06-10 22:34:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/795906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/795906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/795906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/795906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/795906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/795906/1"}}},"commit":{"parents":[{"commit":"ddbbd430df07a9cfc88f089e41717220e9aadcdb","subject":"Merge \"Add user doc for managing recordsets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddbbd430df07a9cfc88f089e41717220e9aadcdb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-10 22:33:06.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-10 22:34:07.000000000","tz":0},"subject":"Fix migration for sqlalchemy 1.4","message":"Fix migration for sqlalchemy 1.4\n\nThis patch fixes a database migration for sqlalchemy 1.4.\nIt also removes some unused functions from that migration.\n\nChange-Id: I4fee32f9be080eea6eef38fcecbdb2dd3940b9e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f40bb13c33dab5acb504194d4c7aa66546fd8f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f40bb13c33dab5acb504194d4c7aa66546fd8f68"}]},"branch":"refs/heads/master"},"d58cb6b5e2f96f54875065a01634dc14737c0429":{"kind":"REWORK","_number":2,"created":"2021-06-10 22:53:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/795906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/795906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/795906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/795906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/795906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/795906/2"}}},"commit":{"parents":[{"commit":"ddbbd430df07a9cfc88f089e41717220e9aadcdb","subject":"Merge \"Add user doc for managing recordsets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddbbd430df07a9cfc88f089e41717220e9aadcdb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-10 22:33:06.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-10 22:53:01.000000000","tz":0},"subject":"Fix migration for sqlalchemy 1.4","message":"Fix migration for sqlalchemy 1.4\n\nThis patch fixes a database migration for sqlalchemy 1.4.\nIt also removes some unused functions from that migration.\n\nChange-Id: I4fee32f9be080eea6eef38fcecbdb2dd3940b9e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d58cb6b5e2f96f54875065a01634dc14737c0429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d58cb6b5e2f96f54875065a01634dc14737c0429"}]},"branch":"refs/heads/master"},"fba57ab1caae768b7ca5c73a0db9e3e0ba241533":{"kind":"REWORK","_number":3,"created":"2021-06-14 21:45:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/795906/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/795906/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/795906/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/795906/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/795906/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/795906/3"}}},"commit":{"parents":[{"commit":"ddbbd430df07a9cfc88f089e41717220e9aadcdb","subject":"Merge \"Add user doc for managing recordsets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddbbd430df07a9cfc88f089e41717220e9aadcdb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-10 22:33:06.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-14 21:43:55.000000000","tz":0},"subject":"Fix migration for sqlalchemy 1.4","message":"Fix migration for sqlalchemy 1.4\n\nThis patch fixes a database migration for sqlalchemy 1.4.\nIt also removes some unused functions from that migration.\n\nNote: This patch temporarily disables the scenario jobs to allow\nthis gate fix to merge while the tempest test issue is being\nresolved.\nA direct follow up patch will re-enable the tests.\n\nChange-Id: I4fee32f9be080eea6eef38fcecbdb2dd3940b9e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fba57ab1caae768b7ca5c73a0db9e3e0ba241533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fba57ab1caae768b7ca5c73a0db9e3e0ba241533"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
