)]}'
{"id":"openstack%2Fdesignate~796171","triplet_id":"openstack%2Fdesignate~master~Ia5c8a88d91b125542635efd4feca7a2c6f062601","project":"openstack/designate","branch":"master","topic":"update_context","hashtags":[],"change_id":"Ia5c8a88d91b125542635efd4feca7a2c6f062601","subject":"Updated middleware","status":"ABANDONED","created":"2021-06-14 02:44:25.000000000","updated":"2021-08-28 19:18:38.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"0b14ca04fb2bc5d3608f9a5a24d18aa62bbbc1e7","_number":796171,"virtual_id_number":796171,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-01 07:53:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 04:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04ef7179a1f4e9452af858adf750ab2916b4adc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-14 02:44:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2dd6955738666480c21f874edf5510cb75a79c8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-14 02:47:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"762893bc388b847f81e96140ea758f79962d604f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-14 02:48:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff19aba279ff0b727ea3ea0ca9bfec7d5cd2e6ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-14 02:48:46.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"28e8d25a6eb629eae4cd1b5f02042e44900077fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-14 02:55:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89715ab1f33374ae3378c906abe392b16453a326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 04:02:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7b870c3c30474e929c86c7f8d9aff576 : SUCCESS in 43m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cb5086bfc3044280bd5c80947f57c7c1 : FAILURE in 38m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8bd7ee39f4eb4131afd8461e76d6a52e : SUCCESS in 1h 03m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1cc664c6245d44db87e6641f0fea9d1f : SUCCESS in 36m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1e52edd1b5b64221bc051ccce244a35c : FAILURE in 39m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e89ebf8d27c44213b7bad3d9b07bf28a : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3aa512cf339e46228e5fb1a93408bf36 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25eb6e702a244a8b9ed4a776070b292d : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06f10e33dd8649439b8fdb7ee0e49e80 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8a356691c094907aa2120bfc201d198 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf8384c5a95e433da28b54e0fa5e2fbd : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759fcbdc98ab40ab97705f899d0ee4dd : SUCCESS in 7m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/64423ad5a2054b1abe4230fc0aa4891b : SUCCESS in 32m 32s","accounts_in_message":[],"_revision_number":3},{"id":"f6d7289c7724e3c9cbdb6115becd5cfb54c537a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-19 01:31:28.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f37f4558a985b5d7fe9eee120a76e906bb068fbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-19 01:40:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"12e0f2678a252a285fc5c2154776c41f5bb82665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-19 02:49:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/de4498ddf38e40018b95ba634c1a692a : SUCCESS in 30m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/99ec4a27ff2a42269986ff7d1f53175e : SUCCESS in 31m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a6f4c5f878624f91b74600985d91846e : SUCCESS in 1h 03m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/214325142e6a4278b76c97905f00b8a7 : SUCCESS in 33m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/39fdb172723d4559b5d338c145a3a7d5 : SUCCESS in 32m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b91d428fb055483b9398e13012040450 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5579bebc82024762abf48da1af63137c : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/989aa28cb2c448d0929d8ee22e69b3cf : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c581adf63f74f1d972b1fb6740e262f : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/942569cbc5bf42ff813c98c4923f3879 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd3aa35633764b88bb0227cc5d0c7be4 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c0a67572da40fb8bc5c1e3815784f4 : SUCCESS in 7m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9869eca2f48b400eb5b1d4051a41c7e6 : FAILURE in 26m 58s","accounts_in_message":[],"_revision_number":5},{"id":"7cef2509488778ec7de0ac783b5c9eb1f65cf310","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-19 03:25:27.000000000","message":"Patch Set 5:\n\nrecheck HTTPSConnectionPool(host\u003d\u0027opendev.org\u0027, port\u003d443): Read timed out.","accounts_in_message":[],"_revision_number":5},{"id":"6bc3685ef5da2da5bfd84b34937341b4497913d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-19 04:40:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0dedcd97a514443288771f203394e3ab : SUCCESS in 46m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9d5ba05269484b2990ac5cae8c80fc0b : SUCCESS in 44m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/03c55fc34fb844c08cc1cea8f62f7123 : SUCCESS in 1h 13m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/615ccc20e20b4780905b90ed60302ff3 : SUCCESS in 37m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f575ad5ae0964745ba9559384771239f : SUCCESS in 49m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91af7b507fe54587aa706978da638717 : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57b32c20915b444589e6f6bd24a42ec4 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1722d88462ea4702b4ff103e61f2315f : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/214efdbd71ab4f9fa774c9e1c371be4f : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cac426c972d848539c1afbc8603ab036 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be3febb0804647f19f09b608a66e5578 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae63ec42e3d4da7a711e5c5a2dc1f09 : SUCCESS in 8m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/163f55cbd67044abb9b1c7806213b798 : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":5},{"id":"c89371fe32e4eae5cb8bc330c7c27a0611713491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-07-01 06:44:31.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"4a276861fa1a7cc7b62c4874b77e9eaf15fcf0cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 07:53:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1e7b14ec1b974e628d1495a5f494f363 : FAILURE in 14m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1806f289b9164790b65892ca58311215 : SUCCESS in 38m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c606edbbbe1a410ba73caa1bec36aa63 : SUCCESS in 1h 00m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/676a4736a50d483bb093c50083533874 : SUCCESS in 37m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f3c92a0e3f4e4604aa930f272b42eb37 : SUCCESS in 39m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1567e297a7a348f5967191cbabdb7c59 : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e8096d0e92047319f890f4b58910f30 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36638d5073544fd9a58f2236a3d61ca : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2df46de007ba49d78cbcf4aa68e850df : SUCCESS in 8m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77fdda72ea234af0a9f4fedab0a39d94 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/478301a1b0d44b3799d017f90b8fcb4e : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217291efe60948e8bbace0eb9277295b : SUCCESS in 9m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/96ebaeeb55c9440f8747c1d1858ad85d : TIMED_OUT in 1h 03m 48s","accounts_in_message":[],"_revision_number":6},{"id":"0b14ca04fb2bc5d3608f9a5a24d18aa62bbbc1e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-08-28 19:18:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2fe9f3ac0b3bc2e4545573fafd4f83d3ec3de7cb","revisions":{"9e7b774ad904156e444568559e7a67e107db0a82":{"kind":"REWORK","_number":1,"created":"2021-06-14 02:44:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/71/796171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/796171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/796171/1"}}},"commit":{"parents":[{"commit":"ddbbd430df07a9cfc88f089e41717220e9aadcdb","subject":"Merge \"Add user doc for managing recordsets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddbbd430df07a9cfc88f089e41717220e9aadcdb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-14 02:34:17.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-14 02:43:57.000000000","tz":-420},"subject":"Updated middleware","message":"Updated middleware\n\n- Cleaned up Middleware code.\n- Updated Middleware tests.\n\nChange-Id: Ia5c8a88d91b125542635efd4feca7a2c6f062601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e7b774ad904156e444568559e7a67e107db0a82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e7b774ad904156e444568559e7a67e107db0a82"}]},"branch":"refs/heads/master"},"ebb7bd17f8312a9d4f36f6886a367cb2d4112036":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-14 02:48:46.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/71/796171/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/796171/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/796171/2"}}},"commit":{"parents":[{"commit":"ddbbd430df07a9cfc88f089e41717220e9aadcdb","subject":"Merge \"Add user doc for managing recordsets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddbbd430df07a9cfc88f089e41717220e9aadcdb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-14 02:34:17.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-14 02:48:27.000000000","tz":-420},"subject":"Updated middleware","message":"Updated middleware\n\n- Cleaned up Middleware code.\n- Updated Middleware tests.\n\nDepends-On: https://review.opendev.org/c/openstack/designate/+/795906\n\nChange-Id: Ia5c8a88d91b125542635efd4feca7a2c6f062601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ebb7bd17f8312a9d4f36f6886a367cb2d4112036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ebb7bd17f8312a9d4f36f6886a367cb2d4112036"}]},"branch":"refs/heads/master"},"4b415a9186f55bc05c3850146338526e65773e4f":{"kind":"REWORK","_number":3,"created":"2021-06-14 02:55:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/71/796171/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/796171/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/796171/3"}}},"commit":{"parents":[{"commit":"ddbbd430df07a9cfc88f089e41717220e9aadcdb","subject":"Merge \"Add user doc for managing recordsets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddbbd430df07a9cfc88f089e41717220e9aadcdb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-14 02:34:17.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-14 02:54:52.000000000","tz":-420},"subject":"Updated middleware","message":"Updated middleware\n\n- Cleaned up Middleware code.\n- Updated Middleware tests.\n\nDepends-On: https://review.opendev.org/c/openstack/designate/+/795906\n\nChange-Id: Ia5c8a88d91b125542635efd4feca7a2c6f062601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b415a9186f55bc05c3850146338526e65773e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b415a9186f55bc05c3850146338526e65773e4f"}]},"branch":"refs/heads/master"},"48bdf24f23595932ffb7b5c6a8f6ba5d2bc315b6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-19 01:31:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/71/796171/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/796171/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/796171/4"}}},"commit":{"parents":[{"commit":"40b5633dfd65355e53afd298619e672370d2f64b","subject":"Merge \"Add simple scheduler permutation tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/40b5633dfd65355e53afd298619e672370d2f64b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-14 02:34:17.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-19 01:31:28.000000000","tz":0},"subject":"Updated middleware","message":"Updated middleware\n\n- Cleaned up Middleware code.\n- Updated Middleware tests.\n\nDepends-On: https://review.opendev.org/c/openstack/designate/+/795906\n\nChange-Id: Ia5c8a88d91b125542635efd4feca7a2c6f062601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48bdf24f23595932ffb7b5c6a8f6ba5d2bc315b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48bdf24f23595932ffb7b5c6a8f6ba5d2bc315b6"}]},"branch":"refs/heads/master","description":"Rebase"},"5be7ed39bec43258e1a39c2880b2d908373c7f2b":{"kind":"REWORK","_number":5,"created":"2021-06-19 01:40:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/71/796171/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/796171/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/796171/5"}}},"commit":{"parents":[{"commit":"40b5633dfd65355e53afd298619e672370d2f64b","subject":"Merge \"Add simple scheduler permutation tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/40b5633dfd65355e53afd298619e672370d2f64b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-14 02:34:17.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-19 01:40:16.000000000","tz":-420},"subject":"Updated middleware","message":"Updated middleware\n\n- Cleaned up Middleware code.\n- Updated Middleware tests.\n\nDepends-On: https://review.opendev.org/c/openstack/designate/+/795906\n\nChange-Id: Ia5c8a88d91b125542635efd4feca7a2c6f062601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5be7ed39bec43258e1a39c2880b2d908373c7f2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5be7ed39bec43258e1a39c2880b2d908373c7f2b"}]},"branch":"refs/heads/master"},"2fe9f3ac0b3bc2e4545573fafd4f83d3ec3de7cb":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-07-01 06:44:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/71/796171/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/796171/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/796171/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/796171/6"}}},"commit":{"parents":[{"commit":"5d1127d3e06f6590bea5582c58229691bea41c8e","subject":"Merge \"Improvements to zone blacklist doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d1127d3e06f6590bea5582c58229691bea41c8e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-06-14 02:34:17.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-07-01 06:44:31.000000000","tz":0},"subject":"Updated middleware","message":"Updated middleware\n\n- Cleaned up Middleware code.\n- Updated Middleware tests.\n\nDepends-On: https://review.opendev.org/c/openstack/designate/+/795906\n\nChange-Id: Ia5c8a88d91b125542635efd4feca7a2c6f062601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2fe9f3ac0b3bc2e4545573fafd4f83d3ec3de7cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2fe9f3ac0b3bc2e4545573fafd4f83d3ec3de7cb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
