)]}'
{"id":"openstack%2Fdesignate~796633","triplet_id":"openstack%2Fdesignate~master~I5307de429114b20efd9785c3c0cdb33977418423","project":"openstack/designate","branch":"master","topic":"bp/zone-move","attention_set":{},"removed_from_attention_set":{"6994":{"account":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"last_update":"2023-12-16 02:13:49.000000000","reason":"Change was submitted"},"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2023-12-16 02:13:49.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-08-13 07:02:26.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2023-12-16 02:13:49.000000000","reason":"Change was submitted"},"20104":{"account":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"last_update":"2023-12-16 02:13:49.000000000","reason":"Change was submitted"},"36201":{"account":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"last_update":"2023-12-16 02:13:49.000000000","reason":"Change was submitted"},"36073":{"account":{"_account_id":36073,"name":"Ihor Zashchelkin","email":"igorzash@tutanota.com","username":"ihorzash"},"last_update":"2023-12-16 02:13:49.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-12-15 23:56:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-12-16 02:13:49.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-15 22:48:42.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I5307de429114b20efd9785c3c0cdb33977418423","subject":"Add new API for zone move","status":"MERGED","created":"2021-06-16 11:58:20.000000000","updated":"2023-12-16 02:17:36.000000000","submitted":"2023-12-16 02:13:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":87,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"796633-bp/zone-move","meta_rev_id":"58e1778cef1cdc48ac8cfdef288ecb5c98c6494c","_number":796633,"virtual_id_number":796633,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":36073,"name":"Ihor Zashchelkin","email":"igorzash@tutanota.com","username":"ihorzash"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-16 02:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":36073,"name":"Ihor Zashchelkin","email":"igorzash@tutanota.com","username":"ihorzash"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-15 22:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2023-12-15 22:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":36073,"name":"Ihor Zashchelkin","email":"igorzash@tutanota.com","username":"ihorzash"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-12-15 22:49:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":36073,"name":"Ihor Zashchelkin","email":"igorzash@tutanota.com","username":"ihorzash"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"}],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":36073,"name":"Ihor Zashchelkin","email":"igorzash@tutanota.com","username":"ihorzash"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 12:00:03.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"state":"REVIEWER"},{"updated":"2021-06-16 12:00:03.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-16 12:00:03.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2021-06-16 13:01:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-19 08:32:47.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-07-19 08:32:47.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-10-23 12:51:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-06-08 09:38:55.000000000","updated_by":{"_account_id":36073,"name":"Ihor Zashchelkin","email":"igorzash@tutanota.com","username":"ihorzash"},"reviewer":{"_account_id":36073,"name":"Ihor Zashchelkin","email":"igorzash@tutanota.com","username":"ihorzash"},"state":"REVIEWER"},{"updated":"2023-08-09 08:25:25.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-08-09 08:35:19.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"3a12485b497d34bd8ef2b4aeb5f7a3a4fe5b6602","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-06-16 11:58:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c0daaa6f07e942adea552a79e3df60a984cf69d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:01:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/84887bfcdba1460391afe1fd44ed30ae : SUCCESS in 5m 23s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5356e7158b224eb6a3d8dd3bb03c32d3 : SUCCESS in 37m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cfcef030be9145abaa5023abe1417037 : SUCCESS in 34m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d11cdf99f0d74bf6868ba6fa044e4c07 : SUCCESS in 58m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/621759870ab1417aa147f1f38c990632 : SUCCESS in 35m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d86e075f4c92441183d9f04a0a750ae5 : SUCCESS in 39m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2df926120404faa894b5bc20f598232 : FAILURE in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/671b3a434b264684bbf31cb5d7ae2a1d : FAILURE in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28a4263077849d6874ff5c3611fc2ff : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d4b09fbf782442c9cf3f8192881c389 : FAILURE in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c0ceb3849ea4a23ae8b0adb98cdce08 : FAILURE in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/771a16af92fc40c3952c8d23e710dfbc : FAILURE in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdcb2b31a9bb44adb93325a978474ebb : SUCCESS in 9m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a556cf76222646c981662fe15e4b3135 : SUCCESS in 29m 02s","accounts_in_message":[],"_revision_number":1},{"id":"df500c29617dae97ff34e25202e42ea3ba74142c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-16 14:21:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28e507bbc814ded7c1d1d724e5fed5f0b7ff20be","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-16 14:22:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e99fc072fb54305dff767888dcabf3760d5ca24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-06-16 16:00:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ea594daeba0e4c57aec09a292fb1ac2e2c3eddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 17:01:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d1e6f231f124311a07867244b568549 : SUCCESS in 5m 55s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ef322b4f1d4649aaaacb3313907fc8a8 : SUCCESS in 38m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ed5a3b450c68498a837e39a4a16ed184 : SUCCESS in 34m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dd40472c8ba14475abf16366ef2f06a2 : SUCCESS in 59m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/eccee2b2f35741989f9de8d83d466851 : SUCCESS in 35m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4cb6088e188741618779fab9e0e81d8e : FAILURE in 35m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f61ac8a490f14e23b5722addd4151013 : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/848963f322be4d6e8a6e98e9ca4154a9 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc328e160c3e4208bc8400c6ae170632 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8155d8ddbfd14ed1a8077a406d826b5e : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/895c0da0ce494bee97d17a42a846a92f : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2ffc1dd14994c8da4fcdc1ffc351260 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31aa6025c8a64c25909d6456c3b98705 : SUCCESS in 8m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb9afe4580fb42d69b8b68ac5904a7c5 : SUCCESS in 35m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b4df559b376c230d7cc6555b39ad2af225a85315","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-06-17 04:34:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d9e06666a1a65e727eea7b06fc9a88bfac381199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 05:44:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b111a9cfdf294cd99af5e8a36d70dffb : SUCCESS in 4m 15s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7aa700c1e6534447b46a875c66ae4848 : SUCCESS in 42m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a319865b99a04175b23bd300c52b5c5b : SUCCESS in 36m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2935a9a18bc64f9d89735a74f685d13c : SUCCESS in 1h 05m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f81a734b024a40dcb527f4ce5a7c0fd5 : SUCCESS in 38m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fbee77780990400abb573f4fd1088eba : SUCCESS in 41m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe46ffc5f32d48cab84a0e932533f0c2 : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2159c489b64642aeb7027cd2ce58d0c9 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f600d2a63914422fb7ec8508394644a7 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3be85191184d4369aa688b2d7c9c32b3 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e60a5c9e6de48fdb716d6fa030a0dd7 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f64cae3dc2e43a9872a3b7dec3a75f0 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e0e7973c8ac462e8cacc83c52792d99 : SUCCESS in 8m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/67ffde8ff6dd416fb8c1a48ac6b3fc7f : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":3},{"id":"a67f38a06989cc8dc47bbafe9ede96e73a94cd69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-06-17 10:37:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7dca4006a30fc28001f0c2d712ec4ba9f390dcf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 11:35:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/050c00f353a9429fb26eaaea0d736b1a : SUCCESS in 4m 46s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a36b2830a2ac4b39a301466af6f006cb : SUCCESS in 34m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/590a0aea9a5f4373810edb0b47159901 : SUCCESS in 40m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c40369f352af4930a341b2ddcbe3264d : SUCCESS in 54m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d04cafbb488d4d58b793d3489690fa50 : SUCCESS in 34m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a4dc9fa1d1004ec1b8b17af5e498d828 : SUCCESS in 40m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62c6a31a7b164107933cf353202f20e6 : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f49998e372474ba0af0a451b5dc91296 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7afe4f75f4d4e3ca9b0ca5c1bf86aff : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/167563231d51430e83007cd1d1559127 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86f14c5c20b24e7bb576a5c754b34848 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ae8e0278c3a4071b0de3ad7f55b2ad4 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7ddccc81ce463d9445f3eb37074a16 : SUCCESS in 9m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0072316e6dab47bdaefbb51af7163641 : SUCCESS in 29m 45s","accounts_in_message":[],"_revision_number":4},{"id":"a21b898a52b8f2c0619412f50175a5fe2f407778","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-08-01 09:02:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"989eaaca711ef02007ab2980f38aaf5918eaff36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-01 09:51:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9ba3051bb45c4ae3bf280e6694e35e8f : SUCCESS in 4m 38s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/98907e85a0914c8f9dedf5df9b81ed53 : SUCCESS in 37m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8a1d0b740c47442e8abf14d2b5d5ad48 : SUCCESS in 40m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e5e6365471b94101b32654266a4a1b78 : FAILURE in 45m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aba0ca854d2f4f38a82feaf206a47140 : SUCCESS in 40m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d720c4ae058d46beaa70f0e10d5458a0 : SUCCESS in 42m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc2e76e214ef46f4a9cb7c319b6a5965 : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c238e8a700bf4fccb87856645318f163 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0b60d710a1845bead01d23ea7d612ca : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d06aba47b6d417788e82ccab063eabd : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b40982f2ff104c2db7f7831b2d2293fe : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5be771422c6c45b0a7c738e6be8c794e : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad4a393a9a34111a02a7150f51f1344 : FAILURE in 5m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/97f134eb904549fa9cea6e8d67e971b1 : SUCCESS in 23m 40s","accounts_in_message":[],"_revision_number":5},{"id":"f80aa6e9d3a98fab076b282ccf6c4e28f7d6fc63","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2021-08-05 11:45:38.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1be4a3872a54f5001a40ce8600b24cf23118fbce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 12:49:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d01e5f8d06f642ae916bcacf8e4bf005 : SUCCESS in 5m 02s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/66b5a4be95e349608b64778a7d5aafae : SUCCESS in 43m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2e68774683d74f11a2f086faf42d17d1 : SUCCESS in 39m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/09226ee4b94e4b42b4ee278bba4682c7 : SUCCESS in 59m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/15c3a92e25664e20a3f5adc489dd78a0 : SUCCESS in 41m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a3184a9352f5478caf4ba7960cc6b642 : SUCCESS in 47m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b6fb1c9d31a4ea3bdbbca7412158bae : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3734a8106194b2da583909ab821a1df : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84114fcebece43b9bba1cd00f5924198 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfbc7349b2e74d2e927b986fd0ffce96 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b128d0cf28174b7bba067075343dd439 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e66e0fb61b5047e38c8a7d37b685bf7c : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b2b0c7cd4d45bc9d12f8ad4c12074a : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7c2199fab7fc44779f148a7b5cc4e011 : SUCCESS in 34m 47s","accounts_in_message":[],"_revision_number":5},{"id":"a8a15b7572b3334a17054365390fb06e4cb751b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-09-17 15:46:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"315197e28725ff5cfe290f6e49722a3119cb5fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 17:14:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/24de9478ca8c4de5abed9f94d5ed0b4c : SUCCESS in 7m 47s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5c768fbd082c4020b865eaa24313cb90 : SUCCESS in 41m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8cccbade02484fc992776345a89585f0 : SUCCESS in 41m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c3e41eca723c4fb1bb6bc0f83396dcfe : SUCCESS in 1h 08m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3d6c1647acd0493080d52d3cb8a26be4 : FAILURE in 43m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1b107b052d9c4bba9b7cab69d9977939 : SUCCESS in 44m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c21e0a895cb34c328703514c7c80cac5 : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fb2cdc0084c42c2ba9c93c24ab134a4 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bdbfc051b804847a066638c3fea2f50 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d8a35cccdfb4ab4831564e94b956b51 : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02e7943afb834e1ca8a0bf04b3da9320 : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f6fe0874afd4f3c96da7616493c5777 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/758d7fb8f2514dd1b85a3e00493be398 : SUCCESS in 9m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3d89063c25934f249c2084fa12778586 : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":6},{"id":"bdbc805ab1dd707ccb68d7001f8fa318713a02c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-04-30 12:19:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"18528e7e90140461250aedc1ba6c0fde2411c5b0","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-04-30 12:20:19.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ea6433c6602db8c4d42b3e6f3b74b694530d6868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-30 13:05:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/34567a6b1c7f45fcae37f627336b6921 : SUCCESS in 3m 39s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8001773304a74e5cac890a140b687580 : SUCCESS in 34m 33s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/d9919721370648db975923112f0a2ab1 : FAILURE in 13m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3adc248669f94ed4bd3ba0dfa108dcae : SUCCESS in 31m 29s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/fb24331ccd084f6dbe5ae3a31382372e : SUCCESS in 33m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/86095bc8e427423bb3ddbea93cd7dda9 : SUCCESS in 34m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1689b4166083477a936e52b12c3fcf43 : SUCCESS in 44m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9f153b65107b43018ce38be6f33aab6d : SUCCESS in 31m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/61461368b1e24f59910ae49eb4f42743 : SUCCESS in 34m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/363b6f93fac2418284ba3f528b486da5 : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d05fc8e7a5134d47b3c5b1a06d4666cc : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69eeabf09194b32b3f87249a114adad : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4f4f9c597fd418ab135d68598868a3d : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bb304bec2a0435d9495ed100801dddf : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f53f15b50a924cb487ba92686a19ff4f : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe6bd129d8244bd95ca3eaf5d88b068 : SUCCESS in 6m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f749408e73a24a36a020caef4ef0f1fc : SUCCESS in 31m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ec9ee41ec6e048cbb3244a20fb3b10ce : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fd454b0e38ce77b063e9dfb108de5c05d879c1b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-07-20 07:03:52.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"115457be21e350d4edfb280891297e6b1b1ee814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 08:05:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6fd8650bbbb48b58d3452e43fdca8c3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d41731cc5bc14729a1f3a93132fe631f : SUCCESS in 4m 19s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/805d707f7e19452581f3439f5c279dfc : SUCCESS in 41m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ef2da6ad7db945c092ff1b03b7acff53 : SUCCESS in 54m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a2362e0538447b4a8c7322a5fe6ed1c : SUCCESS in 47m 07s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f5631eec1e8d4fd98ba5853f4b2b97ed : SUCCESS in 39m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/427b96baa17b46728896b5a1a818a447 : SUCCESS in 39m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ee0b91fac6244f77b5b5b0bb1cf38dd1 : SUCCESS in 59m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/083f1132385f4f85af1de9789952b7f9 : SUCCESS in 55m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/feff3b56f68544ed9bcfbc92dcb730cf : SUCCESS in 40m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a03eca80ceda4aef8ff5e42283866017 : SUCCESS in 41m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbdc1c22772e4e1c85cac71b26d759dc : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d8f28e747a4d5bb8a7fbfa746e6023 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d85bd9350a44870866de611dca5b3da : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3391bc235484f15bff52965540b02e2 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47d261a5c9c3448da808eb41af7c6031 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d452b89375f5402f8b52cbfe68fdcd3c : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be21a6ad1f0649ccb1d91685e6f73071 : SUCCESS in 41m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/15b4394b3df343fbb2e7393a812ef2a4 : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"814ef25624987f3b3c212e4c7240452a173079fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-22 11:33:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ba8a1f47aef40e0f6e95bcc79e4db4bb747b4e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-22 12:29:59.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6889b8511a594ea79dced6ff80c41659\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c5fc169b30a14e69b6aa1cbbf74e0a07 : FAILURE in 3m 20s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f7bb6a8d5f444d5db54982f00982d5a9 : SUCCESS in 28m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5bc7d9305a4a44bfab2f2791914313af : SUCCESS in 55m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d2b896ff8c144b89035876b2c44a378 : SUCCESS in 39m 17s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/dc55357afb654500b734a2d5e6789c8d : SUCCESS in 31m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5fcc8c5599964420a51e33f215f3a091 : SUCCESS in 41m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2974dfd7afb04747bd05ead0e6d260cc : SUCCESS in 44m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7a7ecc19ab06428a8f9a52a2adf4156a : SUCCESS in 54m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/75f25c1ac58040cd8c4a87e8f0a984b9 : SUCCESS in 31m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8f9655daa00344c2a42c0210d75ce30e : SUCCESS in 38m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d0ba2aebd214b6ba1a83aa36729aa4f : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb380db939e4ba08c2e780f8eb43ce0 : FAILURE in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18cc26ab18324a2580b2f6e450c0764c : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89bbb56761c94af9a15bff666a9a5b5b : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33370a048a41450b82232f84a29a433b : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adecff74755f4c038f2e0540201cbb2b : SUCCESS in 3m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e8b5be70a6f4c9fb0481d80331ba724 : SUCCESS in 30m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3a7a396220aa473c9eb6501f4d8129fc : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2cf33d1db342e5d9e545574a53924cf88151d78f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-23 08:25:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b1768e02d56beef51d6cdd6056e39c7166c1699d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-23 09:47:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74e79489f5534e0193fe52205028ac14\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8c3f4567bedf4a58b3d37b533d81d873 : SUCCESS in 3m 28s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b78d6d4246cf465c9324ee42101a9936 : SUCCESS in 41m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/10615cd83fd24ea3a5a32d31907907c0 : SUCCESS in 1h 03m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbab9679124840cd91056de5cdfd2a41 : SUCCESS in 53m 54s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d09bb14bb274404694d4ca1778cc912d : SUCCESS in 40m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6620acd9bcf54c6fbfd99246517447b4 : SUCCESS in 55m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/37c97222e0854513b44783f783e16e73 : SUCCESS in 1h 14m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ab021a22f8f4607817f059e3298f579 : SUCCESS in 1h 01m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d25ea7ddcea540f988d8954401863574 : SUCCESS in 51m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f406503a3ac54eb9a424fbdda41b609c : SUCCESS in 44m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dac5c1b992d747caa0fc82df0198f0bc : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b34b05e9e8ab41aaa9ad99eeb5e6b891 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7843397bc5d5425b9727e1793797efa1 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59c1b6756b30414e8324b2b6d68b47f3 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3a50b089eb9499f9db7bd3c6cc63c20 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3d88a6af1dc43caa58ce7bab82a5ca3 : SUCCESS in 3m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/adc414a9b26d46dcb88e07a0d73a1e5f : SUCCESS in 32m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8b80d1f0860a47c084b94cbf2361c0c4 : SUCCESS in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"081f35c989a97329ec4f4ab15fded0fa90f11dcc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-23 12:51:36.000000000","message":"Patch Set 10:\n\n(10 comments)","accounts_in_message":[],"_revision_number":10},{"id":"9daf2c007569c8bf68ebe14f350281f38b365b7b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-23 13:18:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"748f7ca0cd56641ed16f65dacc3a042e896847a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-24 09:41:22.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"dd4aa36b1cd677b8b5037297b9832c4204002c7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 10:54:20.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bcb74af18974ccd915bfc84eddf5855\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a8d3f1e7bf1943c8875cd1a3aa59775c : SUCCESS in 3m 27s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2239c2c969854e6a8dc53b864342e842 : SUCCESS in 49m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6354f86176c341e09d0c0d0386ff4ddc : SUCCESS in 1h 06m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/55d512c7d24f490a8ad5b1e24a89fdab : SUCCESS in 53m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/aef52bb13339433faf5f135573223504 : SUCCESS in 40m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e8e1b16573ac4447a07185eed04b471a : SUCCESS in 48m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cfc4aa54d4a94e2e9e2d1a53dd8d40ac : SUCCESS in 1h 05m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/782d33c7223648e69d23ea8e98bce984 : SUCCESS in 1h 05m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/83332481c3e6418cb8353dc385d67b70 : SUCCESS in 49m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f169c299009c4d66bdd26186b18f22fd : SUCCESS in 46m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee508709029e4377b36c74b5979d4cbc : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19afe474a55a402e99de710fe1fa5823 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b212240907245f39c03fed3e8da2fa0 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15a3d1beccea4a06b71148ede1c0ac1d : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560118c4cf0c4e9c9322faa5471a6b46 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0e6a33eba1248778136086433c5c7e1 : SUCCESS in 3m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7fdc1275ddde41e0a3e37fc082fc7f34 : SUCCESS in 29m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c89db83dc6f4450aaa688c693b168e69 : SUCCESS in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"92f4f4c8da5c00dbb74ac9bea0cf201e888b4f76","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-11-03 09:03:29.000000000","message":"Patch Set 11:\n\n(11 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7ab0bd4445b67ea3b6782329e8005560fcc948e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-11-16 14:47:14.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"ce6fab8987e1285c37c51281f6f1dcc61f3f435f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 16:06:42.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a4fcb0a73ee4e42be10f43d4e6601b6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ebdfd368fc0e4c3bbc3a71ad6238a5e9 : SUCCESS in 3m 57s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e415a5eb68ad4184af88fd825d67d40b : SUCCESS in 52m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7b69d178e6ba4acf985d24158e18c017 : SUCCESS in 1h 08m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/730db29ca1014dc79db27da6078dc196 : SUCCESS in 59m 37s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/39ac551b63394837a5f5a517b690d1dd : SUCCESS in 48m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/543daf98823d49229dca21eaa648bde2 : SUCCESS in 43m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0fe536d7a7c146ad8d98f2021154d655 : SUCCESS in 1h 03m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1c763818ee6e4844af25daf6c00952a8 : SUCCESS in 1h 11m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/745213d3579e414580cbb6886b46a2b4 : SUCCESS in 46m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b3a1380b252c4cbabd60c4f00dd924bb : SUCCESS in 48m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f7e310f82a54163ab1cdab23cac54a3 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b0ace2b9d34a478303f6c8219d182b : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c314818e6bd1420d91796ebf26cac322 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b390ab053cb46fa941a690424471982 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6232eec8351a4ce589696bddcf5b600f : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e8ea955726b47988e907880002942f7 : SUCCESS in 4m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e81a555fdfbe41b1b24e278f412c543d : SUCCESS in 30m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b0ca00eaab584b93a7728c472fc9a76e : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"49a09b3c8c37fc5cf50563cbbcc3a6f4d404b12d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-11-23 07:50:50.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"bece03a37afceb810c1dabb0a7253310a28665b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 09:03:52.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eace2c38d46c48eb9ef6af7807b27740\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b9e586caf4f844a0ac28a5a12ff414a5 : SUCCESS in 3m 21s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2d1006b13d1d4ac08e274fd7bb20d997 : SUCCESS in 51m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8f4d44cc35aa4fd0857926d56b2699bd : SUCCESS in 1h 06m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/385f24b726444213bec2a23797df62ec : SUCCESS in 57m 03s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/195383f7358c4add8c7016995f8db632 : SUCCESS in 44m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d2d82757322141689738ea2f6da305f2 : SUCCESS in 44m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0d794650f0b34f719a69a648fe0468f2 : SUCCESS in 1h 04m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f2f86a76d7b74235afc0410356415085 : SUCCESS in 1h 06m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cf0380180baf4b5ea8f1574c3898eb69 : SUCCESS in 48m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/67d46e3cc67f4b169a3d0707918109f4 : SUCCESS in 47m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c590ca0714f845848287796cc805d5ea : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42bbc191875d4cab9a7f62e0188a666e : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10828def0a9f42c68e6705abb52cefeb : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cde819a08ff54048b59df4480cb8a588 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f25c6dbc15b467dbe8e4ec755d777bc : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1eff300e7d248979794af4ab0ae7269 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e8c51d46172f4d9f8dba7fd7b7f2f427 : SUCCESS in 32m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f8c12529c5b541d9b379d7b1b5ce0886 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"bd2ef16fe362c12f7ee3d4e72c95fef1d8283abc","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2023-04-14 07:31:02.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5cd3cb7fb111d1b39739df632e052db2d8cdb04e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-04-17 12:24:29.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"ae2fd2ca7b77839f7da42e023faf9831c664f0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:45:44.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca587a1ec4aa4fdb854a6825622bca1f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a7c2bc2333a14db5854c0a0e056ff996 : SUCCESS in 3m 36s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f723cbb46b064a5d8129cdbd8bc9152c : SUCCESS in 55m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ed71af77023e448da7648a521004192b : SUCCESS in 44m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e22dbc65653f438eb9bf6c59c547d780 : SUCCESS in 39m 34s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1d6d87277e5c432f9b662c9f4e4c1f66 : SUCCESS in 55m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/337cbc77c55449619a5fba4e795f6e88 : SUCCESS in 47m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f284c492c317431ba5aec2ef35713ee0 : SUCCESS in 1h 12m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/94f456f6b5ac45d38bb4ef6a7f60ee9a : SUCCESS in 1h 08m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7a95f08ef5004bb58e1435d6ef7aaf9c : SUCCESS in 46m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/61679760de3741329f91c67126c99466 : SUCCESS in 50m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ec3fdf02f9a45668226d622d212e82d : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94515ad7837641e7893e7f66741480be : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22d3f45934304ff9a14db5b49c56c03d : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f84fee92e5a74e62b10738377c96d090 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b4049e99f14ebe865fa46d06fce7f2 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2442429cdd054f5aa62f506cd37b86f3 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d379d25c715847f6939b4d75542d2aee : SUCCESS in 27m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a0b3ca0033a5462795b1c7a213a205b3 : SUCCESS in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"db150215f74919b0e261e22c6b5162f8203d77e7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-17 17:17:33.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"70633b5ae8a09cad981ca5f637de0719611ef319","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-04-17 17:18:43.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c01a765cc7b5bc38ea4e32c5ce5642bb03e56e04","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-04-18 12:31:44.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"2e5810fa287104a57e1957f469c9d50bf47cbc6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-04-18 12:33:32.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"c8cbbd9685339d1cb1b1f9005de1f816f73058d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 13:35:48.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b11a7b93dfb403e8e23e0ef153f5f3e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/50b42a9728684e5d9838d76365a0b17a : FAILURE in 3m 38s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7e29df094ab94a59ad8657227eb16842 : SUCCESS in 44m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6b31efcf5e05408b95b05d7e15e52443 : SUCCESS in 45m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/edc85755b430483784fdd47b3fa4f19b : SUCCESS in 43m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/efa2afc89ad3449eb5ba0f06c0578601 : SUCCESS in 43m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a76f2ac0a3a948e9b3e7a7bda2e6e60c : SUCCESS in 39m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cf49b8866755410e8cfeaf080925d457 : SUCCESS in 1h 00m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b9a2e656c188432297d18257bcf0608d : SUCCESS in 58m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3b86ffb0a4844abd8fe8e67eb60aa102 : SUCCESS in 42m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c1baf7d65f474951a205f609e10e7963 : SUCCESS in 44m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/713a492554c04677937c9006011223a6 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db71ad5e2ab44f83bfc1f8bc7dc37d5f : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c135004905f3479eb6cc10c74d436393 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7122f3c33a24311a7b02ef7ce30811b : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c659ea210be642ea865a742e6a967930 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0084c2450ec44b9a84b90c6f1347277d : SUCCESS in 4m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/843808d68f3840028f42493bf163720c : SUCCESS in 32m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/126294c8afd04b9f9e0e2aacfc77ff92 : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"273f406a2ce1645a7a96c43c35d5eeb5aa2818b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-04-18 13:48:38.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"1a905941497140ca3575a9ac59fb0fc85cb3992c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 14:57:11.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c58d31b11224a07b5a6d10f4b9c5870\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/18dafd6db5214af5b517a285a29e1c92 : SUCCESS in 3m 39s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e25972c3cdf4440fbc30dde8314e8f7a : SUCCESS in 48m 28s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5c76fa7daa8d46678d99991ed358fe08 : SUCCESS in 47m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2058d105f9fa439e9d8df12510fd0068 : SUCCESS in 43m 58s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8a45bbb0a90c46049af1029e90b7be56 : SUCCESS in 44m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/79cd2254c3cd4f3d9a42031cde7201a8 : SUCCESS in 40m 49s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6828dc1e9b8a449491791239fa5927f1 : SUCCESS in 1h 07m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4708ed105efb495fa4e4d00615d98663 : SUCCESS in 1h 05m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9ff81d84650044bb93d9c17113fc5fbd : SUCCESS in 46m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0a6841080ca44dc0948313c55a3a10ee : SUCCESS in 53m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a956ebdb3c643f0ba42d4507b497213 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d11278089c45daa1a320d0a482dbb9 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcf3a7b8103e4490a0f28ea7520f880c : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/207d3bccf54649dbbef578e433555055 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c803a03b213240a5ba4979a66ecaecef : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c209f92c86424fb3a0b6f58eeb668f81 : SUCCESS in 3m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f8741ed92edb41aa98253210dc64c909 : SUCCESS in 32m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5fcc6c41068844f2a10b662d26cfecff : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"dddfc17578429e9665a54063345dc9612bd5ec29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-05-23 07:03:41.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"8f162984294020df0090cd9b79ce24b686144168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 08:29:38.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4600315f06243b0a18eda3028b537c0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7ade350f00db475b8f1a64b5801dfc40 : SUCCESS in 3m 08s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/234cd7dacb104f71ab2ae44da6d29d0b : SUCCESS in 58m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1e858e26396c4591894d14425c7da75d : FAILURE in 25m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d517804a1ba492099a1c46b0e3bfb2c : SUCCESS in 1h 00m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/dc1a5510b3ec489b8d5a998965ba07ec : SUCCESS in 52m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c6cb71daac854e8dbbf7bba7f432beff : SUCCESS in 52m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0c7b8d511ae44c7b8ab271b4c8579ffe : SUCCESS in 54m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9f9baeb73f6440eb958bbf4089549c31 : SUCCESS in 1h 20m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b903d6b93d644240bbfda7a0e35875f1 : SUCCESS in 51m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ad86e2aa50ad4fdcb7b6ccdf7f9747d0 : SUCCESS in 41m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91135bc6c1374672b60bb0dec1eeb45d : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ee7422d4ab444c81abb239d42b2473 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/293759aa0e174df299106d5d676952ac : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53ebd3287f344ff98c555f508919d0cb : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/367983fe8ed94cda9ba91a3dfe7d1477 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8d8b11c4be4dd4b0c4c3026d71521e : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a5a07af8c8844ec9566d694372b8a4e : SUCCESS in 3m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/966d971612394c4f999ef706d7156e98 : SUCCESS in 40m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/860a5be3c4c843d49463e9e41e396f91 : SUCCESS in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f0ecfc7f1e87579ef7f02f4d9730a849daacfe98","author":{"_account_id":36073,"name":"Ihor Zashchelkin","email":"igorzash@tutanota.com","username":"ihorzash"},"date":"2023-06-08 09:38:55.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"d745b7bb91610c518fc7d3349be7d5aa2efde87f","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2023-06-08 09:55:34.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"99908d61311334362e7a21a4666bdf379734539c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-06-08 17:02:52.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"0e3123f229115741ae67c542a62727f490c7e812","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-06-09 09:02:54.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"8793a6f68aba107dac843d4379f9a90ce710917a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-06-09 09:03:29.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"f18fcaf2164e5b495579fcb14c1d6e58953c8d59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 10:20:29.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cd1036cc5aa41c781574243d1e68a7b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2bf0e928ad89496e8b21f840e47bd9f3 : SUCCESS in 3m 36s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/822ebf564e6b4e3395856100b942fb9f : SUCCESS in 48m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fa1f44299ad54922b8e91f216918f63d : SUCCESS in 55m 49s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/67ee2373746e41f7a17f0c2b5cfa1e89 : SUCCESS in 54m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f1b7457655fb40d68b52dd9987c05fd7 : SUCCESS in 57m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dd6c28e21ef74ae4a2712fad4e807226 : SUCCESS in 51m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cff12898aeff45c3917b7e77b6fcbf72 : SUCCESS in 1h 09m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5e33a9143d9c4f438d4e622e08934949 : SUCCESS in 1h 10m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5f11e3676982474b942d970058a288f8 : SUCCESS in 47m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/373c540e2ce7422a952a2c7287334e45 : SUCCESS in 45m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03b011cc11d446499e868c255814ea81 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4164d3fb32a540b9a0a8f84913d9916f : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffdd2af8750447ce80ac69ddaa356a18 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6332a4ba3a1746ef9af2a06b273fb6cf : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f193aeaeb244c818c146fc08c9ee5ca : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3b870ec5ae4c13b71b4d736953510f : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26feac1bd53945d6b40587bbe2af9c7d : SUCCESS in 4m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3f95154343d14266ba92d958a7154990 : SUCCESS in 31m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/eef23b15ffb34c0d8725d7954a045c1f : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"b35668b4865ecbc7ebf2577e4d319ddaa635f7a4","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2023-06-26 14:26:05.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"16cd4a335d9189d8ddc3546fd4244abc44fb4af3","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-07-25 13:15:51.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"c7242904d170cba80eb14ff726846505891637d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-05 15:19:34.000000000","message":"Patch Set 19: Patch Set 18 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"9e312a9b319963dc656da145469e298563e5d4fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-05 16:37:58.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41256f8cbb884be38a1255149a82b75b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/baac24a4c84a48e3a01a3ef987122351 : SUCCESS in 4m 10s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e48a57451b434daebe84645826e0139d : SUCCESS in 47m 21s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0904452adde44da381276e9d2c29be89 : SUCCESS in 45m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c712af8f32b4844baca8b1b216f0227 : SUCCESS in 39m 49s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/92249b51a40546d5896e090937c46113 : SUCCESS in 32m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e21ff86ec3a04cc6b0a9f7f2c61cb75f : SUCCESS in 36m 32s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a451aec9e8e34c89baf32d0bafc5cb93 : SUCCESS in 56m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/927d4cb6ba2043a1a0f5448a5db042b3 : SUCCESS in 1h 10m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c401ec01e1b342af9d80df6a92a344ae : SUCCESS in 45m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/151c93a20b074f8bb175ffca5e8e9337 : SUCCESS in 44m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5652bcbd01504da4bc0d0ae6f4ac1bcf : FAILURE in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1496482748249aca20bb648498ca235 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/624ed9364c174c62a921c5d94eab97f0 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2d8e6a0ac6d4941ab813edf6bba168a : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cd2c9ee00434ed2b4995bbe54b3e932 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7f39834449447abb6dde5784745ef8a : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2da1bb162934600bc02847b31bfde20 : SUCCESS in 4m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/32a760bd5f6e4442a03d86186e8838b5 : SUCCESS in 34m 06s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e3a27fc55cb94b9883a8a605785ddf89 : SUCCESS in 6m 38s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/63035047741c429a846071e61573eb72 : SUCCESS in 36m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/09b52e1acdcc42e2911a90548be92536 : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"f0c91719f6d07ebfb9d71e73e9ca48dccde782a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-07 12:02:36.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"f68f228dbb9ae6db57226488ac5a5fda7aa45138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 13:36:11.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0015c426f6f641d483909d5c5f1d9b9c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1c546b9aa2a44188a5da6bb9bba6f159 : SUCCESS in 4m 23s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/54cad6dc4df2410083e67dbeef1dc228 : SUCCESS in 38m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1784adf288e14b09906d9f41e74e0762 : SUCCESS in 43m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/47ff7be6b198406a88966929135b63ac : SUCCESS in 43m 49s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d3ac41f231a3418e80b93d64fc5366fe : SUCCESS in 39m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/182f2eb53ebc498592e63b5fb82a5029 : SUCCESS in 52m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/980da0b75a144707ada53d2a22cc0d39 : SUCCESS in 1h 09m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aec306b9a07349218238d87c32543b5e : SUCCESS in 1h 12m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/32051d5a661e4ef1bd0055b8d46be5c3 : SUCCESS in 42m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8f682d6da8ee49af931098deb7bd0f51 : SUCCESS in 38m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e0a4f4a87144cf190658b0bead54882 : FAILURE in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1527bb72cb254dbea4c6b13358b700ae : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c9fddd780f54f728a7cc9658327d3d0 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/102ef9de79d34f4db51f6b4009513588 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0673061e0924ea383cb6535a98df88d : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04de5f3a960d44f69e563af38831a6b2 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9b0f98404c14ff188281356f9c70427 : SUCCESS in 4m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5ac264b9fcce4c79862a3577eec49e1e : SUCCESS in 31m 32s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c947e75c2df84951afc01a3f764e5e8f : SUCCESS in 8m 07s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/30f6955463864b7e94dcb3733ba79d6f : SUCCESS in 49m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f6e7a1c056394557b62e37895ad3c8d8 : SUCCESS in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"54f9d17cc6c4594dc5272fa0b7369d860f3782e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-08 14:52:55.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"704a6de5dcb0294dde6d9187da5b78c1cb86afd0","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-08 14:53:52.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d1a1d92b43a14e1d9f802a1407f2f40669566998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 16:21:54.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/790a2da625b443218fe428e75ecf942f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/000beee4d0244e5c8acc370e9f5a5dae : SUCCESS in 3m 58s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/54177b82d18e4de083633d8bd6be0917 : SUCCESS in 41m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/66b88c007ec64f97b1711377e71466a0 : SUCCESS in 44m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cb3463f2b8c4d839df913f2aa140104 : SUCCESS in 48m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7931aae63f90487dbbaf69a3d182e619 : SUCCESS in 41m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5442c1d485c54db48e20fd2dd6c73b33 : SUCCESS in 45m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c7c421866b834e64939949156534fd5d : SUCCESS in 1h 12m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e522282ad19a4fa9b0d9514e088c7167 : SUCCESS in 1h 14m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ee3433a276f8499a98317238aff937c0 : SUCCESS in 40m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d365b62a6c144823ac89f261dfb7fbaf : SUCCESS in 56m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba5743312d4842e997293fd4ab056718 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8bbb885906449b8ae6a82cee246f276 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86da420e39f24330a6a1de810f5948ed : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfef9bced2764c7c94d1e04794d92a60 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7667b33b38e7462981a703d3e0da7fd2 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590f4b53b9c143679a86dc413a3d4ced : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4aa5e1f5a684d11a3b23e385f2019ba : SUCCESS in 4m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31c5a2386c48469986baa9110deadb6a : SUCCESS in 43m 05s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7d953e4d740a44b9908f9ad09747b2e0 : SUCCESS in 6m 54s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/996c5aaa01a94b8b9b2068ed9b49b17b : SUCCESS in 40m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1ea7141a80c3440698619fdc364defb2 : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"a1d9c0f58ca722c83b8868550daa3a4d9b4f544a","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2023-08-09 07:56:06.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"8d3e1e762502bc856666154262430f91c17e68a6","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-08-09 08:25:25.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"20b2b44e02b77918dc593a9e6605f46028ba5043","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-08-09 08:35:19.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"96f8004889981219fc2719f031f1a39141f02bbc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-10 20:25:36.000000000","message":"Patch Set 21: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":21},{"id":"671138ef6dea000970f49e509aa20d9bbecf326d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-21 14:55:12.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"36dfc98dbd8b3bbc15b14a10596cafac302429da","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-21 14:55:14.000000000","message":"Patch Set 21:\n\n(7 comments)","accounts_in_message":[],"_revision_number":21},{"id":"3717085e4a4a165807701dc3683978d85da48216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 16:08:54.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/895b4ce436d64081b5f6e278574ae030\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e9f55bc299304fe28c476fbb55e9ada5 : SUCCESS in 4m 37s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/808f102f42ce4cc2b469267aade9af9d : SUCCESS in 34m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/22b804d54aa14848af10724b4cb67c6a : FAILURE in 20m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5df5dbfc75b74021a7313d625ae42e2f : FAILURE in 28m 05s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ba2da08421f349f6922ec5c205f2b8b3 : SUCCESS in 36m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a18d97a04a34463cb67b4b00025a28ed : SUCCESS in 49m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f8456e2f4e624951a1b65b7fc7e2970f : SUCCESS in 1h 06m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/163578ecab05434a8032951e599acb7b : SUCCESS in 1h 08m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/459ae59e31fe4126938a362c1112f42f : SUCCESS in 40m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2773fddd64fd4abe8e61529551ca103e : SUCCESS in 48m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57227708e3bf49a19ab3dbecaafb4fde : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5c672b958404c408bb41eba9b087c81 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abc3c8d768124ee38c456394fb946418 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1066361af4c84cb89a79773c29df07d9 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6892a8c4cb364cde94d6c2429dfa3535 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3fa71a33bd7d4e19a7252f0b9c8be872 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba3a16c8f9d34ae7bc24ba445256b29c : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1c5845551584aa8ace9b6da184efec7 : SUCCESS in 4m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5c69b69db6746df94a5c1af77b9eb28 : SUCCESS in 39m 56s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8135c921952244e59a733592c2dfc92a : SUCCESS in 7m 11s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2a2eaa11140f444f98dc45df91014d2d : SUCCESS in 53m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/50165b95d0334b97967940beadda0604 : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"d5a4be4e8f954dac2fc7b3baadb2ea88206b19be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-09-07 10:30:32.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"27899ff8f5f2eee90b0d519058caa50dedf3550b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:50:51.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ddf837ad0eb4c5c800896dfd5a486a1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/388974cb0a8148769db3c8481deaf1bd : SUCCESS in 4m 03s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7a42f962cf274f4f979c3c3135b5a05a : SUCCESS in 46m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/939309210ff340f6aebab85c4873c7d3 : SUCCESS in 1h 02m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9df1791f0e0f430c9ca86d4f4c1111a5 : SUCCESS in 45m 14s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/899e005538274275ace09c5e44294354 : SUCCESS in 53m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3617c5d1aea548d7b532ef9a16a614ad : SUCCESS in 45m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8fc8272179ef427d8e001b64fed11dad : SUCCESS in 1h 13m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/829fdd0cfb604dffbb5215cb41a481f1 : SUCCESS in 49m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0fa91574e8114c2a81dbdba5d8493dd9 : SUCCESS in 37m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6c3374380d4a44bb89f3fd5e427ef978 : SUCCESS in 45m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b5204b0ec894a1e91fc8be1513bf23d : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8dc08b77f04a49b7cb57c7e95e6d8e : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31e0be70fade45018d67548bc399f360 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a9848977de44685b409e1cc8e56d599 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81611b5dc69344d5bdf341833e6a970f : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17be257d0e08445e8f822bbc87203d30 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a5f8afa29b4c7783d2d0e7472749a7 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d3b6f52bbc14d5c85d92fda706817c9 : SUCCESS in 5m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/18a82cdf8eaa49bd8947cc9eebf52bfc : SUCCESS in 32m 30s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cdbe045e4b2940c0b86f574c4289925f : SUCCESS in 6m 33s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0f8f126f36974313bb6e846f7aaee9a6 : SUCCESS in 41m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/784e0661db514d5da776263da3ba204f : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"bdaf4a156e7bf82c9310ee0ab5d6a923b8deb2dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-09-21 12:50:37.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"b563b92cfec3d17bfc649a43a9da2e723f60a5ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 13:57:26.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a10bc3d24ee3481e98e16d56396d1c5e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/93048cee69b94f2c87383c93995f77fd : SUCCESS in 4m 06s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d7a32e3a6c3e4f48b9847aff8b0a842e : SUCCESS in 40m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/13089e11228143b5b117a9212fefff54 : SUCCESS in 50m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9b1655893144470855d39b6346d9e03 : SUCCESS in 47m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5635309e6fc94736b824c41103d353a4 : SUCCESS in 50m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fe2411ec39594b399147641488adaea5 : SUCCESS in 40m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fafec604fcbe4a6a827571527575f344 : SUCCESS in 1h 04m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ace673e5dd81468b9a68a80b175ba29b : SUCCESS in 49m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5f61da7ad01d4f9ba8f51dfd137c90bb : SUCCESS in 35m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eeec0abe5c2e45dbb0da1e40db1c1757 : SUCCESS in 47m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c48988124fd74b388b75eb3d7cf10f7c : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c3d72980ea471cac7ca49728952c7c : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dd7e532065b42a683a59fa6033c70fc : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78154944c9f542958ce527e7eee8a9ae : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25ff067ab78849408888967a1520ccf2 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b229668fde14113a40cf98756b02320 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a2da92dc4945db8da2a07dc7f4714c : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a5f4f0c0c5949168c065ca57289685c : SUCCESS in 6m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/39657465b28c4ce69cd83725f81cc4af : SUCCESS in 33m 35s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3c22af1b749a44ae86bbd9d444a2b3b3 : SUCCESS in 7m 28s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bdb5af61bf2441fa98f32286a1ffb51b : SUCCESS in 52m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5aa1d36634c24298b3d56427ce5637cf : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"c8bc0bb73b41ce1eef39a2371b42acacac43649f","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-09-22 08:49:21.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"b6a87b9e46996d30d7e40b07a85172b0b81ac1fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-25 15:19:21.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"3dc16d238de294c322bc9e71c933568cbca984b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 16:44:38.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0ae1a29da064e5fa15c880494ef64fc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a587ff6097554d6ea4410ed0f610a460 : SUCCESS in 3m 59s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2107da3be81545c488fe72fb536c59b5 : SUCCESS in 32m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4407ff86fceb445b89b02d6a43590b2c : SUCCESS in 54m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/da63d54c465a4913867fd52a108ed8a8 : SUCCESS in 52m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/41b3afd1570240a2b1e33dc0d2c680f9 : SUCCESS in 47m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c56532d2a5e84bd893a6d5d11e3573b4 : FAILURE in 48m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c084c563860c41ae956d506ba6db5d5c : SUCCESS in 1h 15m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0cc5b8e1dcfc4d6191032bd02de330a5 : SUCCESS in 1h 19m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f9366a4f22754793bc5b48f965b80bed : SUCCESS in 49m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9b044a8c22e2425ba634c1f48772deb3 : SUCCESS in 35m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e6d21ca553248388119e95d32a78086 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/877d2193080a4dc2ae4d3e4d8a2142bf : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca6a427ec7f64161aef5137f36c2fea9 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/afb435115f4e4834b396ac0f0b848e8c : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41580ed1464c448b8d8b2ded69529554 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0692d05a64ce4dc7b51998c4b0732a93 : SUCCESS in 4m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/87a0012551544eae965708b6c4194322 : SUCCESS in 38m 28s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7bd6116be55e47ff837f4500802d6352 : SUCCESS in 6m 37s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e165da4ee2714bdeb78b69f4e3850a9d : SUCCESS in 48m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/66bf3c35d5ba4de78e6bff73efe720d1 : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"4e0e5153708be9c7a616aea22964901ec92467ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-10-27 08:08:15.000000000","message":"Uploaded patch set 26: New patch set was added with same tree, parent tree, and commit message as Patch Set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"5d18c41d91b070fc0fc6bca87b12dfcfe210338f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 10:29:32.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73b24104d161430eb01bf9f04bda1121\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/325ea92fbf2346ab951f7c792a9b03e2 : SUCCESS in 4m 30s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7ecc681c429f4181b40637c9b4778cc6 : FAILURE in 34m 46s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/80b3c7a181d442c3aa71e1795aab2cfe : SUCCESS in 41m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/52253cd5d9a64249a72d1492337b8667 : SUCCESS in 50m 34s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/998f328252764734ac865c79a652b2da : SUCCESS in 36m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ff3471d951ff464083532d7bd51875a7 : SUCCESS in 40m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1df6476b62344c89b1bb2a0ac4b2e54b : SUCCESS in 54m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/345a51cc196e41a4ae016fb4476a7776 : SUCCESS in 56m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/37ced184f76043378a73c9d64ec71bae : SUCCESS in 37m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/233601d10544466c9a0bdcac89ed1083 : SUCCESS in 34m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b44d1a0cc1144c7a1c33b1d89861874 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dfbea6d3c48431d896bd9edf021c022 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3355fcccdc24ef1b9b8b9e15426b05d : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95f843e214774042ac539c10b219dbbb : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa7519d0ae241b4844cf59b302a6ba6 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d003152b13024148b35e2d8e7483bada : SUCCESS in 5m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b10005ff3e742c199d070889813a2a4 : SUCCESS in 31m 53s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cdaa852a4bdc4ef3ad2d0df9e26265d4 : SUCCESS in 5m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/aad7e24120bc47dda69f2db9635dd7ae : SUCCESS in 37m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/14cda40ddc2548f89aa1894ab763a81c : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"865deeaea6bcbb9431f44c2a2b9ffc75fc3a89de","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-10-30 13:43:45.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"9955f40200b193f1150793bd7aec2ad5631a0c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 16:43:16.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bef704271bb8446e84249b4e302480a3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/188a7c3f6c5b4885a2bf52f3ddb02a6b : SUCCESS in 4m 01s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f083edfe1eb848e6b8325e60c118a214 : SUCCESS in 35m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a309e4f34abb46cdafd7961cae85c695 : SUCCESS in 43m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a3e316601ca4493b6ac1299510547f3 : SUCCESS in 42m 33s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ef42683cda734e6da167c32d6c82c79c : SUCCESS in 36m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c6891928363e452b8c7d9c96c00b0e23 : SUCCESS in 36m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7b1eb312c11b43b5ab64ce5255221800 : SUCCESS in 1h 00m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9cbad073a689436b8d89c822ff079e23 : SUCCESS in 57m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ffd68fb3efcf425c96b315ec94db0433 : SUCCESS in 35m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e3af9a822ed04ecbba0a7294712337c1 : SUCCESS in 32m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6e8fe37e41246cf9186a4394547d062 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5027bc23bea740d2be3fc8f4730efdbe : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a22bd0893db4cdcaeb5c35d49077e87 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe5def87f65b42c0a2ec1f95eefaf3e7 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654885bc6f8c4f409fea9cf0692bf3e0 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70f03a88b9524fa0b96ad7ee80ac565d : SUCCESS in 5m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/682d3472a8044493899873aafebc375d : SUCCESS in 40m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fc01da1a2a284b40b05d6b23bbc31f2c : SUCCESS in 6m 40s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/699ecf99e6f34ecf81dbf6a5c45935cb : SUCCESS in 38m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/95b63766a6a7423888e526ae9582eb4e : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"a9a81d0aef3701fb85f9094f10ec42d41b101c50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-10-31 13:50:12.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"a981af7188e7d129162e8b10be31cc362a3012ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 15:52:29.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40a6e812a32040f6866929aadbe004e7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c3c65788238a48acb2484c4aa81888b6 : SUCCESS in 4m 25s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b2e0865a8c5248f081fca03e1bffd60a : FAILURE in 33m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3ff86e5dcfbd4b888559d984e9736b25 : SUCCESS in 50m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/25dd8e796cc6457983dedd9effaee6af : SUCCESS in 48m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a4feb248369141cd880d07a335797266 : SUCCESS in 33m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c9186198a87a4efca85dc3c3887c58d1 : SUCCESS in 40m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/065866682fb3410b9619ea3a388e0df9 : SUCCESS in 54m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/be699326418e4187903b234e4b8b8a9b : SUCCESS in 1h 07m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/906a51ba6df04576a8f948202797b924 : SUCCESS in 41m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/13832fc00b8a48b5a7627e9d2f96b761 : SUCCESS in 37m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0745d3b5618402c918ad2c88e1abf8c : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d061def7cade4bf488cac6a73f50f685 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c447b61abbd464c8cc9e7c08263ff76 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3315d52f0947406cae53981e90abf770 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b78d582e804bb09489a012822c7f3f : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f8c8fbef77447dc898cdba1ae64219e : SUCCESS in 5m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d473c4ee83c34fae8ea0d379bc41f22c : SUCCESS in 35m 38s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ba154456183d492487d9bdf3d4c87a1d : SUCCESS in 9m 09s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5baaf6b87a924a2a90c45db2535ce477 : SUCCESS in 34m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b00331f9f2324226b72ff42b7a73035f : SUCCESS in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"b4b66b0db4808c205e8b4e54f3f72654ec3b59fd","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-11-03 15:58:11.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"e9d2a42f6feaddb259e4eccfedfcd05208dd3186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 17:21:12.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80b8a4d6e76a4cc38372d6d67920444c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9e225b8bd63d441295be7992873dffba : SUCCESS in 4m 26s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/945b6086af5a4d9398c8466b4868d34a : SUCCESS in 40m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/20e6af35f3124d1599412bfa6cc6ff1c : SUCCESS in 49m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2115dfc4759497faac3166c587d1164 : SUCCESS in 57m 19s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/19803a69d3d6426f8e260d635bb3eb6a : SUCCESS in 52m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6619a2612fd9456daad4c5f9f807a77e : SUCCESS in 39m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bf8df3c4fe8840a18e125b0e4fb1d467 : SUCCESS in 1h 07m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e712c1059286464d89c2740f9f78e585 : SUCCESS in 1h 17m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c96a4355193f467cbe5b1e3e48932d98 : SUCCESS in 33m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/540d7943586548b6bc0dc215fbd161cd : SUCCESS in 48m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e396b0f928aa4e168067d2232128fa13 : FAILURE in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ca5eddb9e74843af2fed30b85028c4 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b76bb861f2574a34b2ddf9b8afc6349b : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4965f888c74743c58b1e40d852534a87 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b55d64a9e0d4becb697b4372df3f966 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f21513842114166aacb4fc69c35fc24 : SUCCESS in 5m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b4840bbca839432e805210c94a52c2bd : SUCCESS in 39m 28s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9d9a6734fd154274b947fb3d544ef233 : SUCCESS in 6m 54s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f8db9b8140aa4e05b44268f336633446 : SUCCESS in 53m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/46b1472c2a62467486303cee2a474f18 : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"3b11454db6c532190a44431baa9f920d1ccb5639","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-11-06 13:35:07.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"d6f266f370cebec01f75402322e0ec23812d4244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 14:35:14.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b09730435934379b907b3120afbdd35\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4f45760592514357bec0b32125a86e5c : SUCCESS in 3m 51s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3987e236b0c0426c860d836b5349dbf3 : SUCCESS in 33m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f8b2255b8623411c8890fadb9332bffa : SUCCESS in 39m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1f572625cfd42ca8d342728c6c6dd9d : SUCCESS in 38m 40s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/df272589d6be4decb37e0a8537c1d474 : SUCCESS in 34m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cd3619ac141948d0a728770281a56ac0 : SUCCESS in 32m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/07ec53c6712b4b98ba97d7c56f0d5534 : SUCCESS in 51m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bcf554233afc4160b9d0409412461a67 : SUCCESS in 50m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8f842cf67b62408bb2d69c4d1acf7fc8 : SUCCESS in 59m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/23dfea592471490b83bf37bee64ef4e9 : SUCCESS in 31m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6a6abde0657410a9e08451b42cea098 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd576334e65847efa5a4de068c2c1e90 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f94d454776ae4a12865c307b871912f1 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/efd1972cb49a4b2ebb1c02456631b6eb : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/236efd9878104078abc3a05abbcb1355 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f47e911eab374e7e96e7720bb7776e2b : SUCCESS in 4m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9a9976220ed84de2adb657c37bfaf2f5 : SUCCESS in 31m 57s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/557b6632701648a4acbebd6c7db0596a : SUCCESS in 7m 03s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e4db536736714c5b947657656be8694d : SUCCESS in 49m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0c758cd7257c4cd4a5a42c05900c275a : SUCCESS in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"d663d384b68051d36e2480513d427469b97a47e8","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-11-07 11:04:00.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"3b02d487994e818c3daff6b9b5c8f12cddefd6c3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-07 14:57:24.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"081585f03561dd36ad4822c1526f135772d1476f","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-11-07 15:22:41.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"e7ec62f4165632332a978774939ac5f2d20a1103","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-07 16:25:14.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"d5d7ceb4f25d05242cdf4d2cff253bf874a41e02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-11-08 17:09:37.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"d4fa78eff49812efff2c2243372dff647929c2d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 18:33:26.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d83c602a5b47412aa6caa8330373f086\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/28f122f37eec41ef8758ccc7fd52fe3d : SUCCESS in 4m 09s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c19f5d86555744e1bc3f64440b1f4dcf : SUCCESS in 50m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f10f95c4d9e849c0a473243889c62aa9 : SUCCESS in 40m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/59ea4186f3a74211b5578ce661ab1d31 : SUCCESS in 36m 30s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/402e3ac4e9594a70af5ea11418112598 : SUCCESS in 53m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7cf4610818e14b5eb3ad27f4ba19e97b : SUCCESS in 55m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4ceae891c7c14230a9570a151ecf7259 : SUCCESS in 1h 15m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0a9183ca876049a7bc05a447601c2776 : FAILURE in 1h 00m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cfa76909d3d74fc9b477c6ddc1473382 : SUCCESS in 40m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9fe6f2fd0bdf444492da9d3616114fad : SUCCESS in 40m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/098033bb5066432e97e7dbffc83cc7ff : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd8c8d6789754ad99a02877fd746d6ea : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3baf939537284f02b68aee1cc013696d : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c71693b6fc2c4f8580c9fd11ada81e12 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80587bc7167c44d088413c2356a60e2d : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca7f948d221c476bb226f803f05a071a : SUCCESS in 4m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d6ad08c10c4744af99a2a88b4f464c85 : SUCCESS in 37m 44s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e4d894c291b04587aa922bee3c453b94 : FAILURE in 4m 58s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2f2f81e533dd44b4b75547204a33f78e : SUCCESS in 38m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c038ea2ab3c746849c4fed0f3efb9cfb : SUCCESS in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"1bd6032e179e2e9970c6c5f67701534ba8823bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-11-09 11:02:31.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"23179885f659983d9b236ae38bd0f877d9e36c7e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 11:27:52.000000000","message":"Patch Set 29:\n\n(2 comments)","accounts_in_message":[],"_revision_number":29},{"id":"7e8221a6e4f547652f5c4fec772d2b0e9fe21a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 12:13:03.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8cca743a7584977b846ee4b8e91fef0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/62bb066db42c4f408f2f64282b1e9b43 : SUCCESS in 4m 03s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f07e585ecfab4f388d9a442bae30ee52 : SUCCESS in 31m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/820b839e323c43dbab54affe0309762f : SUCCESS in 1h 03m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d0d75af40e64e8698c8778680dd2e7e : SUCCESS in 56m 23s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4efa73c51e0b473a96e2bf3858d31242 : SUCCESS in 38m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3c0d4cf62f1c49a9b974ffe649f7dde6 : SUCCESS in 45m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9bae801c767d48ba91110f64f890d242 : SUCCESS in 1h 04m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cedc7d8dd47b4ce494f3b1e5cc686f95 : SUCCESS in 1h 05m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/07d763ecab2b42da80de519a5a92f806 : SUCCESS in 36m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/00e0e780b6ab45779a9e30f8792ccf49 : SUCCESS in 41m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cab20da4d4624aeaa727d2024c45a220 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3096faa72de8437a838d10ed95405218 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a37e17c6e08848cc8e050c36c58cfff2 : FAILURE in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9666b2ec23f49d0a56d9d97170e34dd : FAILURE in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046c52763efc47e3a5dbf106cf31a4d1 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/555362ce7545459d9583b6d5a49ea4ee : SUCCESS in 4m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c42868d64e34ad1b14580ae2d399b0f : SUCCESS in 33m 45s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/92592937352b40e4af1dc842152cfc29 : FAILURE in 5m 02s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/56e02fb7d64a497ea4ef7c8bc61e5e57 : SUCCESS in 47m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/28ce587cfda44f80951bac8b2918ed72 : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"99cb65ae34dc2a034ba35d3358fa08ad8a01d3fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-11-09 12:27:07.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"4c9895597339a099e65135265ae091fca388652c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-11-09 12:27:10.000000000","message":"Patch Set 29:\n\n(2 comments)","accounts_in_message":[],"_revision_number":29},{"id":"5885ff48bb572baf597e58dbb47873a78ec02ef6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 12:42:54.000000000","message":"Patch Set 30: Code-Review+2","accounts_in_message":[],"_revision_number":30},{"id":"db1f20826d352d901746f0561a571c9758e4b995","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 13:47:42.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2de6ede1e5a4cf1bba8e824ed872886\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9b8732cc66ee41339d569f5cc71e3600 : SUCCESS in 4m 11s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2500ff13ef494ae592b4d9b9d3b3443f : SUCCESS in 37m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8fcc8b4208e9474cbddca0fbef8b9b9b : SUCCESS in 44m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7326e5e7a35042bc944746701e5f7e03 : SUCCESS in 56m 07s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/157ffa59009c4c09bb534d359d2a5437 : SUCCESS in 51m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a2bfc5249d4c41e4b1d6f64df71ad441 : SUCCESS in 48m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/87743ec0c55244cc8b75dd237c0de741 : SUCCESS in 1h 12m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/91901f7cf17148e48056a9f661eade81 : SUCCESS in 1h 13m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d6fcf27b92824bc7aa1f789c1ae49a4b : SUCCESS in 36m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/206db8716ae3454ba241be9449d2819d : SUCCESS in 36m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0c0f7816b434fc29811ef9ce6b376dc : FAILURE in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3928ad05259a48d39dddc5dac058dbcd : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67980453245f453f899b53cb79424bc6 : FAILURE in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6ec6769088b4beaadbf0213d9c5a8c9 : FAILURE in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fee34b4a4a545aca46a495da8e042c6 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f960aaf8bb8e4501b3e4f8a1f3b7103f : SUCCESS in 4m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d6a1b96817634f9ead728efb99e77cdc : SUCCESS in 37m 38s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b2b8e307778b41f7a5ced1313ad0c427 : FAILURE in 5m 05s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e0ad04ba074c447da05cafd9f602756c : SUCCESS in 55m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6c3a060a9cd34a04982776777fd346ba : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"dff6bbcc4da3c203134cbf324a3ffa2721491913","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-11-09 14:12:08.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"7f5403dc3aa1c631f6f648a00b8255d758a6dfdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 15:26:07.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fa7cbd5eda142d6b7eefc3a9672a6eb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7c1a4bed1d1f4ba0a89713a45870c4d1 : SUCCESS in 4m 23s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/57a08638d93f4f8ea67ec85f4fe8da45 : SUCCESS in 47m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1f09c956db294a4bb71922964b9efd51 : SUCCESS in 1h 06m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b1cd1c95fb24e57b599f5b42b3ff3be : SUCCESS in 1h 07m 04s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/98e913ca7031446e8701ba8da1785f6d : SUCCESS in 48m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/47c1bd2b1d3a4a4d9565c6c7488aef16 : SUCCESS in 49m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/31e114ac37224df992de25eaa89bc95e : SUCCESS in 1h 08m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1f28922b1857434e979e352dc6355518 : SUCCESS in 1h 08m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c9f48b8d47f946be987f368ac4b2bb52 : SUCCESS in 50m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b66a999aaf624d44af432a9609d94a10 : SUCCESS in 47m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/091a3c61ee2742a4990b6121d706ea3f : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3496558609544d6c88ef06e2c2d5d9fa : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e441d89efd9745df999306c321bd600a : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/931a5a0826e447ac9957eb5912519d51 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71eeb0eb8a124af790ebb7d2b68de25c : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/009bec8c130841bd8045779d936b56b5 : SUCCESS in 4m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/53209a651efb4ef5a16889de20e371ab : SUCCESS in 43m 18s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/00dc7248ee0e42339b4561f40e4c72ec : FAILURE in 5m 35s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/109ecdcbada445b9a7d38b479e583df7 : SUCCESS in 37m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/90c77db98d344cd1a83f256dda79c3cf : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"6100410153000aecdf7cac853c04f87cebd936b7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 22:42:50.000000000","message":"Patch Set 31: Code-Review+2","accounts_in_message":[],"_revision_number":31},{"id":"7240f5855b96d29cd78110f49377656416ce8c49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-17 12:15:20.000000000","message":"Patch Set 32: Patch Set 31 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"306ffd8fc831c0ee0453b751cd3ba40caf19ec18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 13:13:05.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04ff1cb7fb774b339f3434ffd92a671f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6d419def3e494e5298be6c0574377df6 : SUCCESS in 3m 43s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b76a015fc5d84b7bb574fd6778dc5333 : SUCCESS in 36m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4bf6574b9ca04232979664209affc2fd : SUCCESS in 47m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a33fad2c4814e49b5382c7ce562aa47 : SUCCESS in 42m 19s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6bb31330015543f8b85188284263b7b3 : SUCCESS in 34m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/57ec023337064bd293570364218f739e : SUCCESS in 34m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d0766b100c95488e91eae8981c78a0cb : SUCCESS in 53m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d2fa545bb94d4b67b2bbf4c106c75b29 : SUCCESS in 55m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f0c584e830c24cad8be947a0aa6919f4 : SUCCESS in 36m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/14365aa91e994c28bd41da8fa7d26520 : SUCCESS in 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b14f4faa08440649acdbfa6367d6c53 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5a8ece70938446b85ad7904b068a54b : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dadc73023f645ecaeeac146392531c9 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bfa54b9935e495babee84f4d2dabd9e : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d8725f567646c6826bee392a7c9406 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da57f881d58e40e7b6e392bae56ac7b3 : SUCCESS in 4m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cedc63409e1044acb10fcda061fdb864 : SUCCESS in 32m 09s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d8072d6fa4f34b0d8e221fb58ea60b0c : SUCCESS in 7m 24s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/54bb41612a8e471fadd65370fc2138fa : SUCCESS in 49m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/05bd86010fd741cc86248fe9d73738b7 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"27b24bc912d41f7253a4ee26a1ce96912eff4ccc","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-12-05 11:05:31.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"f820db075e9a9e77c75d8b67800435736f38c23f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-12-13 16:32:48.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"cbc1e97958cc5a1ed957b202c473a0dfe0b716fa","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-13 17:27:23.000000000","message":"Patch Set 33: Code-Review+2","accounts_in_message":[],"_revision_number":33},{"id":"d125006fd269219e9a540aed157f7073f0b06643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 17:51:57.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2aa7b5544f146809772a53b015b3a6c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/739bfaa73ac742288d56a0a4881a8ca5 : SUCCESS in 3m 52s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b51fabca87834758a00ad09b8e17dc37 : SUCCESS in 38m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/34b3888dfdc24b74bf65cef801fb09e9 : SUCCESS in 58m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d9a2c0acbb74d859e728e2a346bdc0d : SUCCESS in 1h 00m 26s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d684c47446c848089acf1734057a9cbd : SUCCESS in 50m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9940e312f2664c8cb95cba1d856ee380 : SUCCESS in 38m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9638a2882be94106847d81feb663b120 : SUCCESS in 1h 00m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8139f4d336d7460db43347f5666e677e : SUCCESS in 1h 11m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/79e999ce5aae4cf0afe91982dcdcae06 : SUCCESS in 51m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/48cb75451f4a4dd3b5823827d12fbc0a : SUCCESS in 34m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5514dfcdb5a34c61a9a9cd04613498e8 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5308a29a8d6144618b66be6fd05b6429 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/671700bc37a9486c9efc0155d39a508d : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/660437d6ae1e4f619058dff5ef8cf19f : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af9196698170458aa00aa1a1f1b5ead5 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/427f0242ca134d8ab5e2988f0037a03d : SUCCESS in 5m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/861d66760a6747d09a8a1dedcff7cdfd : SUCCESS in 37m 06s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5d49be40b138424390c9daa15bf9945e : SUCCESS in 6m 59s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9c0d21685ef8484bbaab5a57feed597a : SUCCESS in 38m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3de3e677397b4ec2b7512c79d08a27cf : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"a1468281edf69fcf6ba0f7b8a2d1314c44e7c2d3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-14 19:11:32.000000000","message":"Patch Set 33: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":33},{"id":"c2542fd3af5eeef2f0bbbaed7e6383dbb55b24a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-12-15 15:51:04.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"176558e6feed9dd1d551483457962757e04ff1f9","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-12-15 15:51:16.000000000","message":"Patch Set 33:\n\n(9 comments)","accounts_in_message":[],"_revision_number":33},{"id":"06998003c96ac9c3cb92e1ce7f7a9fdf51b35f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 17:05:52.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ae1ee44fe79479b8b923391bd57d2f3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/038d476b8f944b2fb102a110841dc870 : SUCCESS in 3m 44s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f4cd48b6c2134c25ade1e679a7ffc5af : SUCCESS in 46m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f2791530ad4c4e2c8b76c0910b74e7c0 : SUCCESS in 40m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/45523442018442738e7399b61f54b23f : SUCCESS in 58m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6d0a14cdf97c4088941a51fe21559219 : SUCCESS in 48m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b6350f4a81cd436b946e6527f5f935b1 : SUCCESS in 41m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/398c41c5aff34e5190af7231820f73ef : SUCCESS in 1h 10m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/240ca1479df74a6481d408ead6d6ec82 : SUCCESS in 1h 12m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/23b0bdfbd4ab4df9825ca73ab3a80d6e : SUCCESS in 39m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fb8bfcee73294badbeb38da022563d52 : SUCCESS in 41m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83de3d8a6cd541c287d532c20319bf18 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d4c73eddf0a4fb3916c75350be053cf : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a2200cc7fb941069bfff4c6864674b7 : FAILURE in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0570a8ade3644045a7e05dd4a012f5f3 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b78aa0eaf0444382b1abd016d9e828 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/550539a583bf40c2b53c357f803125c9 : SUCCESS in 4m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fd359616169242f39586fa450bf3d588 : SUCCESS in 42m 31s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dc2d1230685741bfa495db91a8ea5b5c : FAILURE in 7m 52s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/839f49a0e4ac454fbe3bfa94fb6209a9 : SUCCESS in 38m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/47d948106e2346d8b870a7f6c4bb5621 : FAILURE in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"960340bb853d1f5c854673026a52c04fc404de4e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-15 22:26:42.000000000","message":"Patch Set 34: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"712d1d093d8521bb2dbf3d10560116d3044f3d56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-15 22:31:27.000000000","message":"Patch Set 35: Published edit on patch set 34.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"75fee28a2b6b0bcbc1e32aa23ea0068028006b32","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-15 22:31:33.000000000","message":"Patch Set 35: Code-Review+2","accounts_in_message":[],"_revision_number":35},{"id":"3004fa7137935e07997881ccf0995c03fdec0b4b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-15 22:48:42.000000000","message":"Patch Set 35: Code-Review+2","accounts_in_message":[],"_revision_number":35},{"id":"732533feb64ab8a148eba5a2ccff2bab23c98c46","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-15 22:49:22.000000000","message":"Patch Set 35: Workflow+1","accounts_in_message":[],"_revision_number":35},{"id":"115f464b96d6a8054eeca51a5550cb32938972fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 23:55:09.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fea5799fac3c46a384e7553368471649\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d07981ff726943feaafef90dabc7ea3b : SUCCESS in 4m 59s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e668c896ee214098a1dba2676fd8123c : FAILURE in 36m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/45c1dd8363e244b1a92b4c7740c5212e : SUCCESS in 59m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/46172cd0e0ef4d4c95d0f3717dfece8c : SUCCESS in 51m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/147a12438e6345e2954833e1c678ae9c : SUCCESS in 37m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0efe8e74f50f4b44a5fc6c7a69ed11f7 : SUCCESS in 49m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7044933c0c28462895785b89d6c413fd : SUCCESS in 1h 15m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f8927b011f5047138250b802f16c21fd : SUCCESS in 56m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/727f492dc323483c91b39c4bf170ec02 : SUCCESS in 48m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/28e9a8f5314f4a0dbc73271fa6ad8ad9 : SUCCESS in 46m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07ddf56c0c694946961d3f4c4a5f1334 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35557788dc59495db00fca2fa058538f : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/422a7063e5034e1fa1847508c0400c10 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf08d8f554a54f53922d01fbfbd6bf66 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3eb0f406e554e139dbba135ff539f94 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cfbb4372d8d4a889354309101dc03c5 : SUCCESS in 4m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c18524a9f13482d8b32a9c89f77bb41 : SUCCESS in 32m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ae163184c036492d9f6669c42d1a0eed : SUCCESS in 5m 23s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3338ff9a449e4c408b658e8d4e7a946a : SUCCESS in 47m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7afecdf2f5974f178f0e422fbe60042f : SUCCESS in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"f8c0985d5a440ec74c990782f1df5b33581a605e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-15 23:56:21.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"ac0db3c804c68172353426f2bfcef96c445cd446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 00:59:59.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe083b3824243dd957ad335ec4222a0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/96366d4dc48647cfba6baac8c418bb1e : SUCCESS in 3m 44s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f1f7ae91130043fcb15060ac8cefe287 : SUCCESS in 33m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/dca5e63239f34da3bed756148660b95b : SUCCESS in 38m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f52da78374b487a967ebf83ca94472c : SUCCESS in 47m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/69c7ce17085b448fab94d33c89f761c0 : SUCCESS in 41m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/41f82871bcad4998ba015a53bc3b0cc4 : SUCCESS in 39m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e0fb62fd9808461ba61e4ceff58f0fee : SUCCESS in 1h 01m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7a519f1fa15a423b9f1f629b80baf5d2 : SUCCESS in 59m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/007429cb04ac4e38a4743752d5941067 : SUCCESS in 46m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4b29ccd47d6a4145bad636960dade646 : SUCCESS in 44m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2450b88e2dae4b308829f27fe21430fc : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff2111d32e1d451f83264893656566e6 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a78e2893abe34510b7e6598f3cd3b97d : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54111a96cb1b44828e2d99bc687449c1 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d67ce204714445a668d5e84139b77a : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71ce0444762e47ccbdb32bbc74b9befd : SUCCESS in 3m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/97dad10289bd4263a4df35507f2362cb : SUCCESS in 34m 12s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f060c71911ea42a9b1fb7e6b4e06df9d : SUCCESS in 6m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/66f9fb40cd9b44d683d47a9956271a90 : SUCCESS in 46m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4619886634fa4340914c7cc0717f001a : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"6b1bf3b080b91c25c39400814811bbe105ff2dc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 01:00:34.000000000","message":"Patch Set 35: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":35},{"id":"0ccbee4bd29848bca487f24d5fdb5b8e6dfc1864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 02:13:49.000000000","message":"Patch Set 35: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60ca8ecfb2504353ab660a4f5914e425\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0f03a09e8de14a229d724b6b33c542ee : SUCCESS in 4m 33s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6c29750a02ca45178b4df474d3173931 : SUCCESS in 46m 42s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d66cfe01a6554a548b77df5d3ac7f8c6 : SUCCESS in 51m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4495891911c741a389b920c538053a00 : SUCCESS in 49m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4a9f0d183a6142e2ac39b7bdb39be0ee : SUCCESS in 1h 08m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fa85f7c2025b422eb2db56098fc44105 : SUCCESS in 35m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5c4f4751bf554dd49cb9cc2cede9eac9 : SUCCESS in 35m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637cdf75ec08454780428b7c33ab6198 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b7fb1c0cd174beaacd52b31fbdf7080 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee0ba4b260ca4622980e4302948b5914 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba4fe8c714d74cde89a08d916fe82ec0 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bec86a2081904660aee2d798690429ea : SUCCESS in 4m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e537698244d64f578d69fe657db5c036 : SUCCESS in 37m 44s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fe11583893d44d628f0f98bb2fd2cc56 : SUCCESS in 6m 56s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2bda5c7e95c74f6394ee120d5f285ad0 : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":35},{"id":"973a51c5d5e05902932e96b14f572d58acb97274","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 02:13:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":35},{"id":"58e1778cef1cdc48ac8cfdef288ecb5c98c6494c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 02:17:36.000000000","message":"Patch Set 35:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc6ed4386caa4d27b88566b010ba7046\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b842117abe01464d8c59ea4e0005ac51 : SUCCESS in 1m 34s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/207a900cbffb4154bd522eb878d1b8bb : SUCCESS in 1m 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9b45a69c1fb4192be551ef8a1f87aa8 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":35}],"current_revision_number":35,"current_revision":"8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78","revisions":{"8f74260a64404b857bf42c08c2330ee25cce27f2":{"kind":"REWORK","_number":1,"created":"2021-06-16 11:58:20.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/1"}}},"commit":{"parents":[{"commit":"5c60084f135621e987743ab51891cf2ae5f60f64","subject":"Moving to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c60084f135621e987743ab51891cf2ae5f60f64"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"subject":"[WIP]Add new API for zone move","message":"[WIP]Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8f74260a64404b857bf42c08c2330ee25cce27f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8f74260a64404b857bf42c08c2330ee25cce27f2"}]},"branch":"refs/heads/master"},"579cf4bd747c6cb41864d7ba641f6867ad44e36f":{"kind":"REWORK","_number":2,"created":"2021-06-16 16:00:46.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/2"}}},"commit":{"parents":[{"commit":"5c60084f135621e987743ab51891cf2ae5f60f64","subject":"Moving to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c60084f135621e987743ab51891cf2ae5f60f64"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 15:11:23.000000000","tz":120},"subject":"[WIP]Add new API for zone move","message":"[WIP]Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/579cf4bd747c6cb41864d7ba641f6867ad44e36f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/579cf4bd747c6cb41864d7ba641f6867ad44e36f"}]},"branch":"refs/heads/master"},"29ffbf4557b06d781f3cc6dddf6be470542dc9b0":{"kind":"REWORK","_number":3,"created":"2021-06-17 04:34:29.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/3"}}},"commit":{"parents":[{"commit":"5c60084f135621e987743ab51891cf2ae5f60f64","subject":"Moving to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c60084f135621e987743ab51891cf2ae5f60f64"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-17 04:33:12.000000000","tz":120},"subject":"[WIP]Add new API for zone move","message":"[WIP]Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29ffbf4557b06d781f3cc6dddf6be470542dc9b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29ffbf4557b06d781f3cc6dddf6be470542dc9b0"}]},"branch":"refs/heads/master"},"f9e533d0a54f31b8dcd160f4bf56d2b0f6141b3c":{"kind":"REWORK","_number":4,"created":"2021-06-17 10:37:28.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/4"}}},"commit":{"parents":[{"commit":"5c60084f135621e987743ab51891cf2ae5f60f64","subject":"Moving to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c60084f135621e987743ab51891cf2ae5f60f64"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-17 10:36:28.000000000","tz":120},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9e533d0a54f31b8dcd160f4bf56d2b0f6141b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9e533d0a54f31b8dcd160f4bf56d2b0f6141b3c"}]},"branch":"refs/heads/master"},"37d011c77213ff1ad6027ee123395b276f765a73":{"kind":"REWORK","_number":5,"created":"2021-08-01 09:02:20.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/5"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-08-01 09:02:05.000000000","tz":120},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/37d011c77213ff1ad6027ee123395b276f765a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/37d011c77213ff1ad6027ee123395b276f765a73"}]},"branch":"refs/heads/master"},"090253cf0ef313d79051c9edc95bedf995e4fa39":{"kind":"REWORK","_number":6,"created":"2021-09-17 15:46:11.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/6"}}},"commit":{"parents":[{"commit":"2618eb3afe975899080a6b4a6bd4b37ffb0b2bf5","subject":"Merge \"Fix typos in zone policy deprecations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2618eb3afe975899080a6b4a6bd4b37ffb0b2bf5"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2021-09-17 15:44:54.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/090253cf0ef313d79051c9edc95bedf995e4fa39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/090253cf0ef313d79051c9edc95bedf995e4fa39"}]},"branch":"refs/heads/master"},"23c707b012e83a96151f1ee4c4d2a566d9906e90":{"kind":"REWORK","_number":7,"created":"2022-04-30 12:19:35.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/7"}}},"commit":{"parents":[{"commit":"fc905bf76031a88d1164fc613295c9b6acdfe8fe","subject":"Removing more unused rpc calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc905bf76031a88d1164fc613295c9b6acdfe8fe"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2022-04-30 12:06:42.000000000","tz":120},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23c707b012e83a96151f1ee4c4d2a566d9906e90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23c707b012e83a96151f1ee4c4d2a566d9906e90"}]},"branch":"refs/heads/master"},"bb06ed889edece1b949ddfdf063118dcec602a21":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-07-20 07:03:52.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/8"}}},"commit":{"parents":[{"commit":"ad3bdad7b72e929946d65d3d32bec92ac96b9879","subject":"Merge \"Remove logic for Python \u003c 3.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad3bdad7b72e929946d65d3d32bec92ac96b9879"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-07-20 07:03:51.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bb06ed889edece1b949ddfdf063118dcec602a21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bb06ed889edece1b949ddfdf063118dcec602a21"}]},"branch":"refs/heads/master"},"d0f126055d562da318b92658a21c43a5a059361c":{"kind":"REWORK","_number":9,"created":"2022-10-22 11:33:12.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/9"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-22 11:20:38.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d0f126055d562da318b92658a21c43a5a059361c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d0f126055d562da318b92658a21c43a5a059361c"}]},"branch":"refs/heads/master"},"64db5510547336e6c2673d319bd72e49998ebfc2":{"kind":"REWORK","_number":10,"created":"2022-10-23 08:25:58.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/10"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-23 08:25:51.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64db5510547336e6c2673d319bd72e49998ebfc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64db5510547336e6c2673d319bd72e49998ebfc2"}]},"branch":"refs/heads/master"},"a1e3b7f6e38af1a58ab2cdbc752daabab855e503":{"kind":"REWORK","_number":11,"created":"2022-10-24 09:41:22.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/11"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-24 09:39:35.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a1e3b7f6e38af1a58ab2cdbc752daabab855e503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a1e3b7f6e38af1a58ab2cdbc752daabab855e503"}]},"branch":"refs/heads/master"},"3fe78b508319afdf0a5a819a09622b11c68616de":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-11-16 14:47:14.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/12"}}},"commit":{"parents":[{"commit":"137cec0c9960a8648b5cccb750b98e836f61965a","subject":"Merge \"Update metadata in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/137cec0c9960a8648b5cccb750b98e836f61965a"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-11-16 14:46:54.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3fe78b508319afdf0a5a819a09622b11c68616de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3fe78b508319afdf0a5a819a09622b11c68616de"}]},"branch":"refs/heads/master"},"66916be2ebfa15919203dda9b10d83fa0e102a72":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2022-11-23 07:50:50.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/13"}}},"commit":{"parents":[{"commit":"0483a533216c85314a0f3102d6600d12647f0c11","subject":"Merge \"Fix configuration leak in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0483a533216c85314a0f3102d6600d12647f0c11"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-11-23 07:50:37.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/66916be2ebfa15919203dda9b10d83fa0e102a72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/66916be2ebfa15919203dda9b10d83fa0e102a72"}]},"branch":"refs/heads/master"},"0a6e2dbf245260e4750c0b235b82046d8874098f":{"kind":"REWORK","_number":14,"created":"2023-04-17 12:24:29.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/14"}}},"commit":{"parents":[{"commit":"ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0","subject":"Merge \"Use ids when removing sink managed records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-04-17 12:20:23.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a6e2dbf245260e4750c0b235b82046d8874098f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a6e2dbf245260e4750c0b235b82046d8874098f"}]},"branch":"refs/heads/master"},"6868d142db7dfa6a3407c8ab946be49eeff50b4f":{"kind":"REWORK","_number":15,"created":"2023-04-18 12:33:32.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/15"}}},"commit":{"parents":[{"commit":"ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0","subject":"Merge \"Use ids when removing sink managed records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-04-18 12:33:24.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6868d142db7dfa6a3407c8ab946be49eeff50b4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6868d142db7dfa6a3407c8ab946be49eeff50b4f"}]},"branch":"refs/heads/master"},"f8fb0b566b18d2f6b0cf182111f002ddfd8b3d72":{"kind":"REWORK","_number":16,"created":"2023-04-18 13:48:38.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/16"}}},"commit":{"parents":[{"commit":"ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0","subject":"Merge \"Use ids when removing sink managed records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae3bb0ddf359cd0a0d7b5b4705e331d104c6fce0"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-04-18 13:48:23.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f8fb0b566b18d2f6b0cf182111f002ddfd8b3d72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f8fb0b566b18d2f6b0cf182111f002ddfd8b3d72"}]},"branch":"refs/heads/master"},"a2670d822f29a74fc873f7ef08e9161d194394a0":{"kind":"REWORK","_number":17,"created":"2023-05-23 07:03:41.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/17"}}},"commit":{"parents":[{"commit":"1b4281b3ab08d507196ab55b3b3bdac8e7bbfa53","subject":"Refactor managed record logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b4281b3ab08d507196ab55b3b3bdac8e7bbfa53"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-05-23 06:48:36.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2670d822f29a74fc873f7ef08e9161d194394a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2670d822f29a74fc873f7ef08e9161d194394a0"}]},"branch":"refs/heads/master"},"3a0f440b036686573164a777f452cfdec2d0a60d":{"kind":"REWORK","_number":18,"created":"2023-06-09 09:02:54.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/18"}}},"commit":{"parents":[{"commit":"4fc70c5890044b0f02744a023f071578388374d5","subject":"Merge \"Add docs describing classless PTR delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fc70c5890044b0f02744a023f071578388374d5"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-06-09 09:02:42.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3a0f440b036686573164a777f452cfdec2d0a60d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3a0f440b036686573164a777f452cfdec2d0a60d"}]},"branch":"refs/heads/master"},"2bc1eba1277c8c93ed27b33e3a3083719292d444":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2023-08-05 15:19:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/33/796633/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/19"}}},"commit":{"parents":[{"commit":"2897c21d8973ace33e29c38296362b4a3f7635e3","subject":"Merge \"Simplify test_purge_zones to fix race condition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2897c21d8973ace33e29c38296362b4a3f7635e3"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-05 15:19:34.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2bc1eba1277c8c93ed27b33e3a3083719292d444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2bc1eba1277c8c93ed27b33e3a3083719292d444"}]},"branch":"refs/heads/master","description":"Rebase"},"ce45d7aace9e75ce9823b4ad6857f986cdb51d8e":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2023-08-07 12:02:36.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/20"}}},"commit":{"parents":[{"commit":"d3b3669996a7540042a06f5e4366a7a6644b2ece","subject":"Merge \"Removed unused object fields code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d3b3669996a7540042a06f5e4366a7a6644b2ece"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-07 11:49:44.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ce45d7aace9e75ce9823b4ad6857f986cdb51d8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ce45d7aace9e75ce9823b4ad6857f986cdb51d8e"}]},"branch":"refs/heads/master"},"693eb3b8d83cf6adff22ada149ba03a3d8c3c168":{"kind":"REWORK","_number":21,"created":"2023-08-08 14:52:55.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/21"}}},"commit":{"parents":[{"commit":"d3b3669996a7540042a06f5e4366a7a6644b2ece","subject":"Merge \"Removed unused object fields code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d3b3669996a7540042a06f5e4366a7a6644b2ece"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-08 14:52:41.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/693eb3b8d83cf6adff22ada149ba03a3d8c3c168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/693eb3b8d83cf6adff22ada149ba03a3d8c3c168"}]},"branch":"refs/heads/master"},"396fd00106163c979568583b495c5e7eb427a4e9":{"kind":"REWORK","_number":22,"created":"2023-08-21 14:55:12.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/22"}}},"commit":{"parents":[{"commit":"d3b3669996a7540042a06f5e4366a7a6644b2ece","subject":"Merge \"Removed unused object fields code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d3b3669996a7540042a06f5e4366a7a6644b2ece"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-21 14:55:05.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/396fd00106163c979568583b495c5e7eb427a4e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/396fd00106163c979568583b495c5e7eb427a4e9"}]},"branch":"refs/heads/master"},"f2ffe9298da78d13149247b84a44ca67932fb60e":{"kind":"REWORK","_number":23,"created":"2023-09-07 10:30:32.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/23"}}},"commit":{"parents":[{"commit":"d3b3669996a7540042a06f5e4366a7a6644b2ece","subject":"Merge \"Removed unused object fields code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d3b3669996a7540042a06f5e4366a7a6644b2ece"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-09-07 10:20:52.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n  - Update SOA with one of the new NS record servers\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2ffe9298da78d13149247b84a44ca67932fb60e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2ffe9298da78d13149247b84a44ca67932fb60e"}]},"branch":"refs/heads/master"},"56bfe49573f737203daea837dc2737ea1f22cfb0":{"kind":"REWORK","_number":24,"created":"2023-09-21 12:50:37.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/24"}}},"commit":{"parents":[{"commit":"d6de3cf4f0560fb84677442ecb824899160ca9ee","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6de3cf4f0560fb84677442ecb824899160ca9ee"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-09-21 12:50:21.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/56bfe49573f737203daea837dc2737ea1f22cfb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/56bfe49573f737203daea837dc2737ea1f22cfb0"}]},"branch":"refs/heads/master"},"822c00455a7d3c4fcc599d49b1629da1d6e783a1":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2023-10-25 15:19:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/33/796633/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/25"}}},"commit":{"parents":[{"commit":"6354b5f79c4de234315b94deaba63907e889a80d","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6354b5f79c4de234315b94deaba63907e889a80d"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-25 15:19:21.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/822c00455a7d3c4fcc599d49b1629da1d6e783a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/822c00455a7d3c4fcc599d49b1629da1d6e783a1"}]},"branch":"refs/heads/master","description":"Rebase"},"8bc11efae9c498fc52a880d2f2d0680119ae7539":{"kind":"NO_CHANGE","_number":26,"created":"2023-10-27 08:08:15.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/26"}}},"commit":{"parents":[{"commit":"6354b5f79c4de234315b94deaba63907e889a80d","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6354b5f79c4de234315b94deaba63907e889a80d"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-10-27 08:08:07.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8bc11efae9c498fc52a880d2f2d0680119ae7539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8bc11efae9c498fc52a880d2f2d0680119ae7539"}]},"branch":"refs/heads/master"},"4a761b5b10d0f218f7522d55f6dd8855e710dfa4":{"kind":"REWORK","_number":27,"created":"2023-10-31 13:50:12.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/27"}}},"commit":{"parents":[{"commit":"b1c305c175fbded1ddc11d4674c390c9323735a1","subject":"Merge \"Removed deprecated mdns config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b1c305c175fbded1ddc11d4674c390c9323735a1"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-10-31 13:42:29.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a761b5b10d0f218f7522d55f6dd8855e710dfa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a761b5b10d0f218f7522d55f6dd8855e710dfa4"}]},"branch":"refs/heads/master"},"8b7d2deed9293dfc1642ed17528c4b62d346cf74":{"kind":"REWORK","_number":28,"created":"2023-11-08 17:09:37.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/28"}}},"commit":{"parents":[{"commit":"6067d6d1996edfc061d9fe9d5cc3e3a4e878b2a6","subject":"Merge \"Improve object fields coverage and fixed error handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6067d6d1996edfc061d9fe9d5cc3e3a4e878b2a6"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-11-08 17:09:29.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b7d2deed9293dfc1642ed17528c4b62d346cf74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b7d2deed9293dfc1642ed17528c4b62d346cf74"}]},"branch":"refs/heads/master"},"46f8cc33461f75af5dd96b2cbf0be4e8ea5f099b":{"kind":"REWORK","_number":29,"created":"2023-11-09 11:02:31.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/29"}}},"commit":{"parents":[{"commit":"6067d6d1996edfc061d9fe9d5cc3e3a4e878b2a6","subject":"Merge \"Improve object fields coverage and fixed error handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6067d6d1996edfc061d9fe9d5cc3e3a4e878b2a6"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-11-09 11:02:24.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/46f8cc33461f75af5dd96b2cbf0be4e8ea5f099b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/46f8cc33461f75af5dd96b2cbf0be4e8ea5f099b"}]},"branch":"refs/heads/master"},"cba083875429a90fcf5e0333f571467966a2dabc":{"kind":"REWORK","_number":30,"created":"2023-11-09 12:27:07.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/30"}}},"commit":{"parents":[{"commit":"6067d6d1996edfc061d9fe9d5cc3e3a4e878b2a6","subject":"Merge \"Improve object fields coverage and fixed error handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6067d6d1996edfc061d9fe9d5cc3e3a4e878b2a6"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-11-09 12:26:58.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cba083875429a90fcf5e0333f571467966a2dabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cba083875429a90fcf5e0333f571467966a2dabc"}]},"branch":"refs/heads/master"},"59ca7b17964929c1f293f6077215caeb36021c2f":{"kind":"REWORK","_number":31,"created":"2023-11-09 14:12:08.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/31"}}},"commit":{"parents":[{"commit":"6067d6d1996edfc061d9fe9d5cc3e3a4e878b2a6","subject":"Merge \"Improve object fields coverage and fixed error handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6067d6d1996edfc061d9fe9d5cc3e3a4e878b2a6"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-11-09 14:12:01.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59ca7b17964929c1f293f6077215caeb36021c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59ca7b17964929c1f293f6077215caeb36021c2f"}]},"branch":"refs/heads/master"},"de7a1292acbf078b95e0285a06157cf3817ff9f2":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2023-11-17 12:15:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/33/796633/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/32"}}},"commit":{"parents":[{"commit":"5f6d25c0b79700cffa06152a1fc5c8bbc7966d3b","subject":"Merge \"Add clean_zonefile to BIND9 pool.yaml doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f6d25c0b79700cffa06152a1fc5c8bbc7966d3b"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-17 12:15:20.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de7a1292acbf078b95e0285a06157cf3817ff9f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de7a1292acbf078b95e0285a06157cf3817ff9f2"}]},"branch":"refs/heads/master","description":"Rebase"},"8560c9641178b17cf320f2b3287f15097a5643c9":{"kind":"REWORK","_number":33,"created":"2023-12-13 16:32:48.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/33"}}},"commit":{"parents":[{"commit":"d90691f3a29416dbb209b52aae288a5916bca632","subject":"Improved service coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-12-13 16:29:04.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8560c9641178b17cf320f2b3287f15097a5643c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8560c9641178b17cf320f2b3287f15097a5643c9"}]},"branch":"refs/heads/master"},"47e5b069d20f58e546039c34b0d53a27815ebdae":{"kind":"REWORK","_number":34,"created":"2023-12-15 15:51:04.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/33/796633/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/34"}}},"commit":{"parents":[{"commit":"f5a034272ddeddf80dc858396cf31ab581f4ae5e","subject":"Add upgrade test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5a034272ddeddf80dc858396cf31ab581f4ae5e"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-12-15 15:50:56.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47e5b069d20f58e546039c34b0d53a27815ebdae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47e5b069d20f58e546039c34b0d53a27815ebdae"}]},"branch":"refs/heads/master"},"8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78":{"kind":"REWORK","_number":35,"created":"2023-12-15 22:31:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/33/796633/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/33/796633/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/33/796633/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/33/796633/35"}}},"commit":{"parents":[{"commit":"f5a034272ddeddf80dc858396cf31ab581f4ae5e","subject":"Add upgrade test coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5a034272ddeddf80dc858396cf31ab581f4ae5e"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-06-16 11:43:03.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-15 22:31:27.000000000","tz":0},"subject":"Add new API for zone move","message":"Add new API for zone move\n\nThe new API would be v2/zones/\u003czone_id\u003e/tasks/move\nOnly POST would be allowed on this API.\nThis move zone from existing pool and add it in new pool. After zone\npool_id field will be updated in DB, clone-zone will be created on\ntarget pool backend servers. The zone transfer(AXFR/IXFR) will happen\nand the zone on target pool gets synced with the Designate DB.\nThis command serve as replacement to \"zone export + zone delete + zone\nimport\" procedure.\n\nAdded following things in pool move operation:\n  - Add/Update NS servers of new pool in the zone\n\nImplements: blueprint zone-move\nChange-Id: I5307de429114b20efd9785c3c0cdb33977418423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8733f8f85bf0b8c3cedc35d9a41a511a3cb8dd78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
