)]}'
{"id":"openstack%2Fdesignate~798157","triplet_id":"openstack%2Fdesignate~master~Ic2571caa71dc99c417ea0933d5d4947287cbe312","project":"openstack/designate","branch":"master","topic":"fix_md5","hashtags":[],"change_id":"Ic2571caa71dc99c417ea0933d5d4947287cbe312","subject":"Replace md5 for fips","status":"MERGED","created":"2021-06-25 20:15:07.000000000","updated":"2021-07-12 23:10:59.000000000","submitted":"2021-07-12 23:09:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798157-fix_md5","meta_rev_id":"724b313fb764e60a4eedfbbc41fedd78f077d586","_number":798157,"virtual_id_number":798157,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-12 23:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-06-29 04:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-06-29 05:44:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2021-06-28 21:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-12 21:59:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-12 21:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"CC":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-25 21:20:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-25 23:06:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-28 06:36:56.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2021-06-28 09:21:12.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"CC"},{"updated":"2021-06-28 19:58:17.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"CC"},{"updated":"2021-06-28 21:58:32.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2021-06-29 04:33:14.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-06-29 05:44:28.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"eeedd0a39b6832cf2d05654ec41b3b33c07d644a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-25 20:15:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a1ac28c4b7d4ab70844e6b433c4e8eb82ac5273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 21:20:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d70145e510fb42909ee13ac4540101dc : SUCCESS in 38m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/242ef662ff7446ef840e2f7a49ba2972 : SUCCESS in 39m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4cbe11ccd31543b69f6244fae53ce58f : SUCCESS in 58m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6828d89c1ebf47f2af1d6fbf63a04176 : SUCCESS in 34m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/697f756c0b00407e91cbd08ca5d36f0e : SUCCESS in 35m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60eabbd75d804b90aef0b4c5bdebc427 : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6de73107d3bf4265b6099363f2bbf640 : FAILURE in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b6b3bf7027343bb8652afe77a902c52 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41b80608bd564531b8240d0034555496 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bcfa80dc9c746ad81e006e9b28c6d65 : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbac48f7e6dc46b39069e128c7f569ba : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ff16c20f0f42f7960db844422c5006 : SUCCESS in 9m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5b74fca607504f608c04d7a5b93e1555 : SUCCESS in 27m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d417fe724d4a3dbdcdc1aa1d47bd74638d5188a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-25 23:06:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5455d3c1f5c179f1fa411cc375c4aac7a6c1a66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 00:08:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aa1257b7d05943768d721864cfd26659 : SUCCESS in 41m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6d4c417e2bd145d6be911c5ebcdcdc24 : SUCCESS in 37m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/357f3eb58bfc490c8b1790b4b3310ab3 : SUCCESS in 58m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/daf3d278cec0484186bb6bc4d7d0220f : SUCCESS in 38m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/23f7313bea19498baad409d7355189ea : SUCCESS in 38m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f3cf5c5446041d7918f37fdacf33cc4 : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21063e5d15e344d281463dccdb059d80 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7da017b087a640198c9a643f974cd411 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ecba0c7b2b7492ebe93b9c55d0cb0c5 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca712baf09214017b8cc40446cf50773 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3935ca7d3c5e4077ab5e8bf0aa5e2a06 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7434c6aaad64f9283ddcaaeea32de14 : SUCCESS in 9m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4b613f922da4da3af8e28720d8f60d1 : FAILURE in 4m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6d7a96410a80453fb38298f773b50241 : SUCCESS in 27m 39s","accounts_in_message":[],"_revision_number":2},{"id":"64d2ec626411815d244e876fdf266e942f929684","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-28 06:36:56.000000000","message":"Patch Set 2:\n\nrecheck zuul results no longer available","accounts_in_message":[],"_revision_number":2},{"id":"73a6980f0fadb3b5834c3956b22b5266cfaf18ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 07:33:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6dc28f6d54b43f98ef3a6b9d78d2905 : SUCCESS in 40m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0f196749bbb7466fb0d204a9034ba588 : SUCCESS in 41m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d2a53ee0ae294c4d862f1cf54736b7d2 : SUCCESS in 55m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5919e008dd394eb5a305ca817a38300c : SUCCESS in 41m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0d00c317fb284a249bc63a82c0eafc41 : SUCCESS in 44m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/589d418995ad463ab1c69dfa2ca775e7 : SUCCESS in 11m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44c72248fab84f558b30818b44cf2f1e : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939bf6e063864551a1c967723cfe2a73 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99e50b693a614815a833fd849aa3ec55 : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7dc34dd4750499ba62888f6cdf392b4 : SUCCESS in 9m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42ae2893a9ac48f3bc42acab172c045c : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab8e2c69fb6e47da9629a12e97460811 : SUCCESS in 17m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/42a7c4ba5817447e84aea02cbe1f89d9 : FAILURE in 5m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/942d2cf971cf4fc3b3b2761a3336d294 : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3d60b2cca542aef8f1475f24ee54742c3b9ae778","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-06-28 09:21:12.000000000","message":"Patch Set 2:\n\nYou need to also adjust requirements.txt\n\nERROR: Package \u0027oslo.utils\u0027 is constrained to 4.7.0 which does not match the minimum version specifier 4.5.0 in requirements.txt","accounts_in_message":[],"_revision_number":2},{"id":"61d8ca7feeac233fe1a9b99722c6972235ae5b4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-28 18:14:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5fee74ef3ec9318fd7c2a644176e46b8d708202e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 19:17:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aed78a5feb224d9a9f65754a364c44fc : FAILURE in 41m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7f5a1cd1610a4b4c9a6ad0163be6da2c : SUCCESS in 43m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a63908673f45460b9fefceb7385f4a6d : SUCCESS in 1h 02m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5b8986d6ca784ac1b00204a92d97d3a9 : FAILURE in 43m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d89966011c654bba990f41a2c7bfb260 : SUCCESS in 41m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68308a752e084db692394bd232c88853 : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfd96872312343118992f2f205db09ee : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6e2985cf2f47599810ec790af0a25e : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b02eb593e1dd462da19443153a47a2aa : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5909d32ad67747ada1b39bbd3980aecf : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d2bc5a17f5f4228af33ebb1cab5a033 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164804356e5f445fabcacc5b3d68c482 : SUCCESS in 10m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8aec4a006dd441ecb2e9c06bdd41a508 : SUCCESS in 4m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/32c05a31d9554b6f9df52cbdf0c6806f : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":3},{"id":"47dd3535c4feeb5029c3191fee0321c09e7d783a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-28 19:40:54.000000000","message":"Patch Set 3:\n\nrecheck DesignatePtrRecord","accounts_in_message":[],"_revision_number":3},{"id":"dc36970fadf6c8487227b2416a92e1fe3b9fd6c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 21:07:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1b0475aa42794fb2b3f2aa23aa6e3a72 : SUCCESS in 37m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e428b2a84d1e4a2b96193737bba2acf9 : SUCCESS in 41m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/57d373f98e644dd8bd3fbde57f19c87b : SUCCESS in 1h 22m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d639c72e97d840b8a77021b2d859fc7f : SUCCESS in 35m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/38d2359f345743fc8f05b1ecf9d51755 : SUCCESS in 51m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59603c463eb940aaae9a4f8a07557ebe : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bce6a2406571449ebaea45369424589a : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adfd666c25d04773b1652fd52ea80c90 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee16ec4273a04c4aa4ef4c4d455a0586 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab51b63bd2b44ea89847bc105aa306a0 : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce7b6eba1b9d4ed5a38dab3ec83837fb : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750586dae7d34b56bce1e47c694b98ae : SUCCESS in 11m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/52c0e75904714afbaf255459d41f9382 : SUCCESS in 4m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/62d0af6d443f4a8db462236fd2a3d38e : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":3},{"id":"caae303eba0f870da16faeab19228336dd86f812","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-06-28 21:58:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ced3a9475c4c6107edc980ea6f81c265c97baaf4","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-06-29 04:33:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"41bfaffb94209964913b187f7dc4a4794c889340","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-29 05:44:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9290338e3eac9a865a4fcb67d936cc6c1c58bd85","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-12 21:59:08.000000000","message":"Patch Set 3: Workflow+1 Review-Priority+2\n\nLGTM, thanks Ade!","accounts_in_message":[],"_revision_number":3},{"id":"49fe94595f27d679108ff45216035c6c58e4bf7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 21:59:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cdc6d6e62e1971bc16e2ee95b4639b879a7005c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 23:09:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2753b74f26e745c89cf741379ca49d21 : SUCCESS in 44m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3f4f92be677c4624a75280ab5f5edb36 : SUCCESS in 38m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/167dea5ae81c4e9688558bab173ea45b : SUCCESS in 1h 02m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1e786bc94c02419abbe3834cddb6757e : SUCCESS in 1h 08m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/05b70969695f4b649f9f629a1c201962 : SUCCESS in 37m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/926868009aff4b40976e3c3ceeb34846 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/725f8da5c4e14b2ba9a63d864bbd0370 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/682b95acb9e24e0bb7e9c0815fafdfe4 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5df55481e85942c4869e749d204a22e7 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a53687279e3431483bf714a82775af7 : SUCCESS in 8m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a7824274162143fdba376ef3df025804 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":3},{"id":"78ce5eeb97241131db11571b9bc36642fc70b752","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 23:09:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"724b313fb764e60a4eedfbbc41fedd78f077d586","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 23:10:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4672d5b188564006a380b006076fd9a4 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7ea564329097c078ac4db9ed02dac7d54ef467db","revisions":{"edb388fa830384bd28b156cd5d6716f3fa1a476f":{"kind":"REWORK","_number":1,"created":"2021-06-25 20:15:07.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/798157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/798157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/798157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/798157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/798157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/798157/1"}}},"commit":{"parents":[{"commit":"5a4f0982d6ff599b8db506367251e24d32340071","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a4f0982d6ff599b8db506367251e24d32340071"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-25 20:10:59.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-25 20:15:06.000000000","tz":-240},"subject":"Replace md5 for fips","message":"Replace md5 for fips\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nIn this case, md5 is used to calculate the hash of a database record\nto ensure record uniqueness.\n\nChange-Id: Ic2571caa71dc99c417ea0933d5d4947287cbe312\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/edb388fa830384bd28b156cd5d6716f3fa1a476f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/edb388fa830384bd28b156cd5d6716f3fa1a476f"}]},"branch":"refs/heads/master"},"0c11a0df83f7165e50c4f968e37b6ca4c26c559b":{"kind":"REWORK","_number":2,"created":"2021-06-25 23:06:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/798157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/798157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/798157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/798157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/798157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/798157/2"}}},"commit":{"parents":[{"commit":"5a4f0982d6ff599b8db506367251e24d32340071","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a4f0982d6ff599b8db506367251e24d32340071"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-25 20:10:59.000000000","tz":-240},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-06-25 23:06:07.000000000","tz":0},"subject":"Replace md5 for fips","message":"Replace md5 for fips\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nIn this case, md5 is used to calculate the hash of a database record\nto ensure record uniqueness.\n\nChange-Id: Ic2571caa71dc99c417ea0933d5d4947287cbe312\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c11a0df83f7165e50c4f968e37b6ca4c26c559b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c11a0df83f7165e50c4f968e37b6ca4c26c559b"}]},"branch":"refs/heads/master"},"7ea564329097c078ac4db9ed02dac7d54ef467db":{"kind":"REWORK","_number":3,"created":"2021-06-28 18:14:08.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/57/798157/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/798157/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/798157/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/798157/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/798157/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/798157/3"}}},"commit":{"parents":[{"commit":"5a4f0982d6ff599b8db506367251e24d32340071","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a4f0982d6ff599b8db506367251e24d32340071"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-25 20:10:59.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-28 18:13:58.000000000","tz":-240},"subject":"Replace md5 for fips","message":"Replace md5 for fips\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nIn this case, md5 is used to calculate the hash of a database record\nto ensure record uniqueness.\n\nChange-Id: Ic2571caa71dc99c417ea0933d5d4947287cbe312\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ea564329097c078ac4db9ed02dac7d54ef467db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ea564329097c078ac4db9ed02dac7d54ef467db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
