)]}'
{"id":"openstack%2Fdesignate~800727","triplet_id":"openstack%2Fdesignate~master~I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a","project":"openstack/designate","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-11-28 07:27:43.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2021-12-21 23:29:08.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2021-12-14 01:59:59.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a","subject":"Fix support for scoped tokens and default roles","status":"MERGED","created":"2021-07-13 21:25:47.000000000","updated":"2021-12-21 23:34:03.000000000","submitted":"2021-12-21 23:29:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"800727-secure-rbac","meta_rev_id":"c297f7c84cdeb90c94713a742d8896301d6fc26c","_number":800727,"virtual_id_number":800727,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-21 23:29:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2021-12-21 16:39:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2021-12-16 05:49:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-12-21 22:39:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-21 23:34:03.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-12-21 23:34:03.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-13 22:26:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-28 18:03:44.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2021-09-10 22:45:53.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2021-09-23 18:32:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-10-21 15:51:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-12-20 14:53:11.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"CC"},{"updated":"2021-12-21 16:39:44.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"832b1c1caae43d7a7c2bb27b639a204a76ba45cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-13 21:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcacef01babbb3e8d3c1f9455ad0805336adc4d0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-13 21:26:37.000000000","message":"Patch Set 1: Workflow-1\n\nWIP - This will not pass yet. We have work to do on the tempest plugin side.","accounts_in_message":[],"_revision_number":1},{"id":"82e094072e9cf12d3ee44310ba74a683b0062bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 22:26:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ca87d60992de49ad8a933741c903878e : SUCCESS in 42m 00s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c4486669f3414afe9e7e0132793bc463 : FAILURE in 20m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e52ca35b974e4b53a1ff371924e596dd : FAILURE in 52m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/00c26cd07de243f98cd2c64a087236bc : SUCCESS in 58m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/28a5e21cbe0d4624b487ebd8313aa812 : SUCCESS in 30m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a73d7311282f4d9790b29f831a262ba9 : SUCCESS in 44m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3a999750a8d4553ab030edab5a7b14b : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e1a15a5245e4980970bfb294c038763 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73231d90e19b45a49a20c169a1e2d4ac : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72ddd0db760441c09c5cb56ce91c4318 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49e00b82aaa44f40817578502838d362 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26d959694a1142efbf7fcd513214c9b0 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838ed0e4f83443e18702f7c85c4d8b3f : SUCCESS in 7m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ef64822d0e564952ab22b1da8b643b12 : SUCCESS in 24m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ddc1683024797118e796d760198fff1ea45cffb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-13 22:40:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44d789f08f3fe5fa33609ca0b9396132d8a85990","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-13 22:40:38.000000000","message":"Patch Set 2: Workflow-1\n\nWIP, will not pass yet.","accounts_in_message":[],"_revision_number":2},{"id":"38fa7574425abf2e7a2ac7ed796d824f3ac0fa40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 23:07:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ecaae6d6190f4debbe9783bb9924bb62 : FAILURE in 24m 22s","accounts_in_message":[],"_revision_number":2},{"id":"3ef57a93e9a51215b5798c0fe2840296ad16e435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-15 22:09:35.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dca053d1706299fcde1002b8d8a336df4be84c2c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-15 22:13:08.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"488491bd1fa955045eebc45d00dbdc9583dedf4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 22:36:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c16b471ec63e428686a9c9902aa8dfed : FAILURE in 24m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a0640bacd3a0f2e95592ff10a83e4a437d1c5fd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-15 23:10:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"002b966764eaf99246df3e08dc6fc116370071de","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-15 23:10:28.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"5dad4f57d09861bfbc7ff567877ac71c7e8fa71e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 23:48:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fde93d13ad8b45acb60d7e5c2553ecf3 : SUCCESS in 37m 05s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/dc7b51cbf2bd4338aa8c4f8f7cef9a71 : FAILURE in 26m 01s","accounts_in_message":[],"_revision_number":4},{"id":"1eba86997a763a8d218adbe7957ec4a8bb3a0ec8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-16 23:25:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c29a2af4fbf4da7f90ce9bb8c8677f2a291fb7d6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-16 23:25:43.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"afc9ec90aeff8ae6f76c3817ea8a210c7df3c7cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-17 00:18:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fbdac36f7a5f41b0a2651ce2c116056e : SUCCESS in 39m 39s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7c58e2db117e45388e6719afd4481510 : FAILURE in 51m 21s","accounts_in_message":[],"_revision_number":5},{"id":"0e24c62e8599ea5b298d149f1debbb6251b5a9c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-19 23:46:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1a7ee7e097e3fbedff40e6cb316d03e0c4e12616","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-19 23:46:57.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"c3ee06c22c852112a6f6690b037d9f5da2ae9d8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:24:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/503e2466da7a405797d73666951d1d3b : SUCCESS in 36m 32s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/033bf3d4067c458a8d194bf442f26a66 : FAILURE in 28m 03s","accounts_in_message":[],"_revision_number":6},{"id":"aa797b6e4f3394d1a49526de645c174e0b79f1ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-21 20:57:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b1367fe83d3481c1f17ed7cb49132496e4d08d95","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-21 20:58:51.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"b0d9c6e745c8005726aacd0db56c3a885dd476ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 21:36:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/76ff895797ed41a19949c00243be990d : SUCCESS in 36m 40s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/dcc3877568ef4c91bbc9c11a1f4bafb8 : FAILURE in 34m 52s","accounts_in_message":[],"_revision_number":7},{"id":"8067cc9e6c9e8f63317c03f409580af7e49fe877","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-21 21:58:31.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"761aafe859dfe6a472c00269fcc8cc11b40946b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 22:11:43.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"0f422018cb65b4d9f35396d1e65c868d32322d2a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-21 22:12:05.000000000","message":"Patch Set 8: Workflow-1\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"21173bc049fdc215d7266fc1947c0bbf7f44441f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-21 22:17:48.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"563f88ac7818b07f4b4f31e9140f7e02db5c192a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-21 22:17:57.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"c394a3ea5f5673ed9c03ec780a16da675fd0bb99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 23:53:25.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c7dd6b73611d4e369c39d2454a6a1018 : FAILURE in 38m 55s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4d1daedb03bd4c64926b95ee331bed1f : FAILURE in 29m 16s","accounts_in_message":[],"_revision_number":9},{"id":"ac9073762d59100cb16d04d5f548764b61052c5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-23 22:08:17.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"97b0e9c770306428abc6ccf0f2666ac8fbc95374","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-23 22:08:30.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"5c10b9bb945bee33f6be3c64756577e29c6b115a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 23:26:28.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/64afaa3219934f85a1ef535540ca7569 : FAILURE in 33m 18s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/346cfb58b47642dcad674a4abf732ffb : FAILURE in 1h 11m 40s","accounts_in_message":[],"_revision_number":10},{"id":"600c60dc16563a1adf1e814983513ca2661eb0d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-26 19:45:34.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"36f29a0c47c0056f1c68e17b92effb3f5adca2a8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-26 19:45:57.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"b9014a65dc1481c46b9c91b84934c8496b386faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 20:22:30.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3ce9dcefd25247e085ae6638a7a00c75 : FAILURE in 28m 46s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/fe2961efd8b54fcf8e9722bca9a8b88c : FAILURE in 34m 59s","accounts_in_message":[],"_revision_number":11},{"id":"1ce3b3dcb76c934f512d40203b69fb18c2132309","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-27 22:43:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7dbd36cfed28f11b6cd927011f39a13673d21db2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-27 22:43:35.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"7ca62d6ee9e00dddc146132609a06d2d051eb991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 23:10:53.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e137bd192fdb498391a6cf214ff86815 : FAILURE in 24m 27s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e2d893339eb5426d8f6d39e8adc961b6 : FAILURE in 25m 52s","accounts_in_message":[],"_revision_number":12},{"id":"41629098cab2050ba2c28693e5d6c7dc48bcc7b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-29 18:07:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e06051aab90f4b2a41bb4cd7c200b925a49d8cfe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-29 18:07:37.000000000","message":"Patch Set 13: Workflow-1","accounts_in_message":[],"_revision_number":13},{"id":"b5046026a40b14c3a609615fb9136a8e6e82272f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 19:01:49.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a0e639423d004bac91fe62baca62de1d : FAILURE in 51m 40s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/86bd19c37b5648a0b8d71432d3ba6ba6 : FAILURE in 38m 30s","accounts_in_message":[],"_revision_number":13},{"id":"f6aa8c03f350db49ec17ce08343be9819f792dcf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-30 23:50:44.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"4728c13e68f834823607847ddf8ac2b1bfd01b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-31 00:48:05.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ff2da78d42cf447595ad291788d1a9df : FAILURE in 40m 42s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e9c099d2b6f1415988fa16835302c3dc : FAILURE in 54m 22s","accounts_in_message":[],"_revision_number":13},{"id":"ace6d205df6b7b709168cb5fb8ad75a10d0ff562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-03 23:49:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1ce1ebef99e1a03657d7ce5f40bcbf861a9a97da","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-03 23:50:09.000000000","message":"Patch Set 14: Workflow-1","accounts_in_message":[],"_revision_number":14},{"id":"020688c1f1c5102821fad16b32e187165f52d8be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 00:57:44.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2cd6f9bd3a6a42b5b7e2715907cc6d30 : FAILURE in 35m 38s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b4992707a8054bc3b70a450f9b734f61 : FAILURE in 1h 05m 30s","accounts_in_message":[],"_revision_number":14},{"id":"f03b8a39459192053515c20ff9209c6e871032c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-05 16:22:40.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"72acfda2b86ea74f1106518ee2d98ff466112a23","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-05 16:22:49.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"bfd610ee98ffc7db5ee2ef0cce64519ad2b15543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 16:57:36.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/428966dd42f14853a96ca1beef3a90b1 : SUCCESS in 33m 06s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3132ca9d10cf45d7b47677a29444a789 : FAILURE in 31m 41s","accounts_in_message":[],"_revision_number":15},{"id":"1d985e3e5e83caa5552a5ce26dc65670ddaec853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-05 22:09:16.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"e57726a981506b4646647a3bedf6735f06111dab","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-05 22:09:29.000000000","message":"Patch Set 16: Workflow-1","accounts_in_message":[],"_revision_number":16},{"id":"e2876641733d377a11637d192c70a4ca5f0be0f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 22:50:16.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/af73fc40bffc45e98fd0cd6064abccc6 : SUCCESS in 35m 40s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ceb04ce1f89744bbb598e6590a397116 : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":16},{"id":"ec82e63c53443d41a9bcce07c5bb802488671d95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-06 21:51:59.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"1f84fedab9990c24aa1f6ced4493ad37c00ec689","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-06 21:52:12.000000000","message":"Patch Set 17: Workflow-1","accounts_in_message":[],"_revision_number":17},{"id":"2256812b1b7ffb3a439e5bbfdc4f735708935d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 22:57:57.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/10aab05b8d35412e9f3b8448e6d9c10e : SUCCESS in 38m 02s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3d480a2f1b8647ffa88f281e9fb39cf3 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":17},{"id":"6d4b237d3b3a469ea95196ed4fc5163f195ef3bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-11 23:11:33.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"b0c1db11c4b04ad98ccc1f544f75994419bdf5d2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-11 23:11:50.000000000","message":"Patch Set 18: Workflow-1","accounts_in_message":[],"_revision_number":18},{"id":"7f547488afdaa00bf25827a3795b2ed1355d9d2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 23:49:34.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aa63b399cd3c4145b2fd94c27159dab9 : SUCCESS in 36m 08s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b9587f03bc1b4ae4af6eb5e0254c9597 : SUCCESS in 36m 12s","accounts_in_message":[],"_revision_number":18},{"id":"824efdabb53ced8d7a2ce8a47b70947d1382e35a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-12 20:05:07.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"0a9154bd45be168f0e9f4f4474a03f21e18edc08","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-12 20:05:37.000000000","message":"Patch Set 19: Workflow-1","accounts_in_message":[],"_revision_number":19},{"id":"8cc33f7b5add0991574bd57efa74edd86e4e3548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 20:14:48.000000000","message":"Patch Set 19: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":19},{"id":"1f715be6afb5dca7a1e6d6600d74d8650451a0bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-12 20:29:16.000000000","message":"Patch Set 19:\n\nrecheck\ndependencies updated","accounts_in_message":[],"_revision_number":19},{"id":"45a79235916285ded857da30b756352c9a3ecbe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 21:12:03.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0885aa4f41c34233937c0e73afcf6a4e : SUCCESS in 38m 12s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2de2ad9094c74024b908829981665185 : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":19},{"id":"62a700a36c05e1c340b7ef241bb655416d0fa0c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-16 20:28:12.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"68b42976df07432e8a4bb0236f586dae796f0243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 21:41:59.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5693ddc56acb4a598fc96e16530a1b41 : SUCCESS in 35m 48s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3b94af6ec6bb4ce7a8553fbe41f870d6 : SUCCESS in 46m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3e1a03a5c80b44008cd7b9a79ed80483 : SUCCESS in 29m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d3abcaf8663a40caaefc596d24438916 : SUCCESS in 1h 08m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b8d7fb24ad5d4d27a56d638344822358 : SUCCESS in 33m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3fe2fc7ad8c44e6a99f8cfdb268ee7ec : SUCCESS in 40m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71fe879af3774fb9b810a15ea233996b : FAILURE in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/892019b4898b430e8d6e1fdcd5f38179 : FAILURE in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf565b4ab2740c8af0efa49cf86f17c : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c1edbaca6ac481e95c96a3aed7afaed : FAILURE in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96cba574bcff405db72be35972999bfc : FAILURE in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cf394badea14cafb88d99b52221a25e : FAILURE in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/690920bf5ebf42b090d436578c97ee41 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/397f9cbe24824b5cb83f7c33b5f8bb7b : SUCCESS in 5m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7d2cf1d2495e483e9f72d537f2e275fb : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":20},{"id":"477b1415aade27cf3cdb2d89c1d00a0aec7a17cd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-18 21:39:38.000000000","message":"Patch Set 20: Workflow-1\n\nI have a bug to fix","accounts_in_message":[],"_revision_number":20},{"id":"afbb0a5549c9309fb03d6758522be7748bfe73fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-19 21:55:08.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"f804d797e0c165a70424745af140cc8f30c0b459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 22:57:03.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/df452f26b6974d20a4b03f860980f556 : SUCCESS in 39m 01s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/edf674891997477383e9edce534f5f5f : FAILURE in 37m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6048f15f417a47dcb9ca8080eb3f8658 : SUCCESS in 34m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/87110e9d105742059f27859abb76af32 : SUCCESS in 58m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ba71109212db49df92b989574347d9fd : SUCCESS in 39m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b1345a68e9264e1e8822089c372a366a : SUCCESS in 43m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18b66991eb994d8487f948493da6ca31 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/972ed27841fd44d78e907958dd45837a : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8af5d06df09425d99cba2a068f9a707 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11d230eccde24670be1c6e923f70a312 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1118f52f12449f6b1177863af9a07c7 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b5f458729984a4cbc656bb040748e46 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce4a8b9b0924c0bb705e85d1266ee6d : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31b8b1700997495e84c20a2861ebaf15 : SUCCESS in 5m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dbd296be2d2741e7a9826c4080d165b4 : SUCCESS in 27m 06s","accounts_in_message":[],"_revision_number":21},{"id":"b4ee6b393ff305d4343f51687eb8435493240d29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-20 00:35:04.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"3f5f104dfcfab77d5a9ddd168257ebdd538b7915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 01:35:42.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/baf0ba0ad80641b599670b84fa738833 : SUCCESS in 42m 41s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1ca9321fdd3d48e383d8502809e54ee4 : SUCCESS in 39m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/419882dd56e845a6b1c3e2e11b1fe6dc : SUCCESS in 39m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5bc976b68d6141a99d91f343815c8ef4 : SUCCESS in 59m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bd6e58121511435cac0ff1262ecdb5b8 : SUCCESS in 45m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/26c8de7e87d54c4f96c32c3deb65cad2 : SUCCESS in 41m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/933a2e6da47a40a8b56533369f3c05ca : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15bb05cc9bcc4bee88b630ad0808734a : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b18e222f10c4cbe95409d96b441faf4 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c725f6b9b4b240838663952cc93bb9a0 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3573a0866a134bebb109db1270bffcc0 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/978f88b6f5074e0497c701067bbd6fcd : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/243327c6c2514e6699f55170167fafb0 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e5abb099638425fb16fd598af8f8bbe : SUCCESS in 4m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ffba2bc5ae9549209b3c52eebbb50500 : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":22},{"id":"f77a88c496d4c0c4c26426ab1ed2b08227ab5914","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-08-28 18:03:44.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"bb0cf25a6ec42d260e5563e17ecf5f72b26f4914","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-31 22:47:05.000000000","message":"Patch Set 22: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"383eac517da54b07b81afa6f0dec8e1d8eb01e90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-01 20:55:01.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"a610d844094a537e8f14b491d1c0a0c36bf285a1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-01 20:55:12.000000000","message":"Patch Set 23: Workflow-1","accounts_in_message":[],"_revision_number":23},{"id":"047ee16ae9dd2aa4d3f975f941e6f0a7ae40057d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 21:58:03.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9d55116038404a0cb75d26d173a09af4 : SUCCESS in 41m 25s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2be658c1b64c4b55a6268d184536ec63 : FAILURE in 37m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/42b9ef2236664e3286779c3a0de33f3a : SUCCESS in 38m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9fa4e3fcf36349c4872425a761448542 : SUCCESS in 1h 01m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b9a75402b00642b09fe9026fa2716abe : SUCCESS in 40m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bb98fbf7900f4e26a84bfc4665c434cf : SUCCESS in 42m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/935dfed6b0de494c87740d2b67824560 : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a067d679b8c243019bc1069af5712d86 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7d11575ecb14763b8b501f8e7682420 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6de6b7eae06342e9b245bd7d531590b2 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ed6a2187a8c44ffa07c74da7ccda23b : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb3690f30439474db66fb67d7b5de23e : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6844bfffb9649a1a8d954f8a24a9e4c : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7b575df8a9849fa8a8e85579bb4e108 : SUCCESS in 5m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3bd8857d1f944385baf8981c94a44bc5 : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":23},{"id":"aa5490125a708cd9c88614a3d01cb08aa3e01565","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-02 21:31:38.000000000","message":"Patch Set 23: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"f34e96f3faa72103fd9e84eadfe28a70e917be56","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-02 21:36:38.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"87eae4e3143e2cea3784ea8155beb8e8f48230a4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-02 21:38:31.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"bcc93587611caff54206be342548acf160bfe160","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 22:43:54.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3539561c5dab407e9390d277b9ffcc6f : SUCCESS in 44m 40s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7cf7757e61704ac19f5fea084429ce78 : SUCCESS in 43m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9050b5d54c71417e91ce984af99277ce : SUCCESS in 35m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/47d2b5d51f5b45f5bb1dcdfccdfaa6d3 : SUCCESS in 1h 11m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3b5e763c67894228919065a7cf5257cc : SUCCESS in 39m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6c500962fe054628a802ee58c571ab9b : SUCCESS in 42m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8c23ebe20be4f748022341202b2c2ab : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c9bfd43b1a14340a47df4a5ee522925 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f5e427de5b444f69b83f3d5daee51c0 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73979b94709c441cb80735320b43a33a : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc5d9b8201b34c339d96b7717612ac52 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88b30500410b44939af62079377ab596 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f660304c6937429391cdddbb341308fe : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4918cf0bbfb415d9f0a057e7d296825 : SUCCESS in 5m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/59ec854c4abf467486b2af4ee8ad4cb8 : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":23},{"id":"c0b6e8d270a664f704a005a31e571e90d2e81757","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-09-02 22:55:29.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"16d04590b883c07b9a574eaad20b44a212da122d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-02 23:05:33.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"23cc68a2d43635ad4842338cb04cc4183ffb1f90","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-09-02 23:07:28.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"419135630225a3721618a194053add28b4d70ed2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-03 18:13:34.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"2145158b5d29ebbe9aab7b0e4da55c2582b1fdc0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-03 18:14:09.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"cb5f93a0da3a4398c9cebebcdbfa6a8a76853a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 19:17:40.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c223505fe89745b2abd5916f5c014865 : SUCCESS in 43m 41s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0427f740f2e147268fbb19d7feb35a45 : SUCCESS in 1h 02m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aa40158905654dd18c47dcbf6f12f9f1 : SUCCESS in 58m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/958c5c18904143a789b612c917304f59 : SUCCESS in 59m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1836659599234ecfbd0f87b93e506b2a : SUCCESS in 36m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/03eab5b1d5be44d19e65efbe6459f113 : SUCCESS in 33m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c3b0b8e02874da9afc112ee51b4b94e : SUCCESS in 7m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2e91aabe53d418f8a6047e5cad14eb9 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c5075403c1488c8820cdb25e7d3b97 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/455bb9042f7e47f6b2eb832db25f22c7 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/651bba0b915b46b19d3e0419510a1093 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef4df15539aa4ea6ba7db6ba985758a5 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7745bce5fa014c6c8e7e5cd0a0b0c9d1 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bc2b19d5c714e39b022fd78f7d6e374 : SUCCESS in 4m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f8efc90812154d38800a53665f1f074c : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":24},{"id":"b599f7dc21aefcc14415b274b1a90ff38194b673","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-09-10 22:45:53.000000000","message":"Patch Set 24: Code-Review+2","accounts_in_message":[],"_revision_number":24},{"id":"42283e13d971dff5e590b7e07416a09d8eeb7e42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-22 21:11:09.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"741f02776275234c56f21ce6820f5e8f104a7d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 22:17:04.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/633295f6b837409282d8131751cf2585 : SUCCESS in 37m 27s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a1a9e5118b074530830956e861379a73 : SUCCESS in 42m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aff6ffee89aa4a5c95902c5f5ca20b9b : SUCCESS in 43m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dcac60efd6b34a9b90abd0cbb4096a5b : SUCCESS in 1h 03m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/75c6fc20428b45229c04c6fa15c656ed : SUCCESS in 45m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c53f214e3ca845b782beb7cccb51ba2a : SUCCESS in 40m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fe5fd44aa4a48fda9733cbf0f8af967 : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cb3692e2b0d432a96399ce014fbfa15 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a3055bc39f141f2b1f65a3fc0c6b208 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c7f58e4f601486f823dcae7658ecc00 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a535804233b49e4a3e70b12ef0ab37b : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94edab8f8a0647dda4fdb1044e836242 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c46b37b33571467a8b4bb3094ba67b6b : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e289df62a5bb40c19508b63431f6d138 : SUCCESS in 5m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a51b8bfdedb4b6aa732cb283a7b1a3a : FAILURE in 25m 30s","accounts_in_message":[],"_revision_number":25},{"id":"fd91d35938bca953724c4d07eb76f299f9e5b91e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-09-23 18:32:53.000000000","message":"Patch Set 25:\n\nrecheck unrelated neutron failure","accounts_in_message":[],"_revision_number":25},{"id":"c39931c863167a7c35cf4ebd7c672d57e15e81a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 20:02:11.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fccd091cdae541be9d920f4ce675fb5b : SUCCESS in 58m 05s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/37a51cd85c7042039ca8af73c0d55589 : SUCCESS in 57m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8e92391c11a34189928cdf893e57105a : SUCCESS in 41m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/75ae0ed2c85047d7ab58508c497161d2 : SUCCESS in 1h 22m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b062d8a2d51f4d96b13dccfbb3ae64ca : SUCCESS in 56m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/83ca3472d55b4495afd5149c5d18bb93 : SUCCESS in 38m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57513efed02441da9f41b8fa106df847 : SUCCESS in 6m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a65af97a69e4da691f46f0ab3be16ac : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad86cffa80974d2b8f8b24f54390a8e0 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cefd1bada62a46f2b8964cd13e0f2cf5 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d4c4203e1a74d27bf7912dd815b47d2 : SUCCESS in 14m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67c97bddb4f44ffd81f569558798a36d : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c5a92ad00d47b3ba6bb173eabe0552 : SUCCESS in 17m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/830bfbe5bf614beba6dfe3f79530876a : SUCCESS in 5m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/47781e9b78594e699a4c6e42cca7af6c : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":25},{"id":"bec330dc64180dd6ad94a936252ffba3ca1f1cbd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-19 08:28:52.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"b451a5aa7dc1a2ff688b7cbe1e76a7932820edc9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-20 16:10:28.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"8019cbfd69853821495152d4a8914985dda1c3cd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-21 15:49:29.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"54ebb70c46fe84e3c41dd8f654a49949f1cd62d3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-21 15:51:56.000000000","message":"Patch Set 25: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"ebb652b90203402d7593720117f174e5cb0b3c41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 16:50:40.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bf236bf848b54147bacc9fa704af84ba : SUCCESS in 46m 09s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e7f7ec16579b4dd1a9b8211d3b66c365 : FAILURE in 16m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e9f9f5d84b944504b0ac3326b7a28eef : SUCCESS in 38m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f7658460e8ce473480615bf42f9d1ffb : SUCCESS in 52m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/81abbf458cbb4998947def455268f196 : SUCCESS in 44m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b812dccb949b4abfba3c4bf87e74b753 : SUCCESS in 43m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ea5106c5a5e4951a626868490525298 : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14d2011deb6749f9a378d54a26eccc0a : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc916530bf224665b34ddc0021da996d : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b07e5308999649d7a3c4a99a0e0fae74 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a67f742d335a40cdb601183b35ee97fe : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5e140bea3b0405e849a2d318040d0b3 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7456990cd98f46cba57eabc38feff630 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07a8ac31943f4cc7be71af3ef9afec97 : SUCCESS in 4m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2fed88be92ff4accabce391e5f3b3137 : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":25},{"id":"23dcc597a57909e223fe2a6cadd8b4e1a2784553","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-17 17:28:32.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"c9c92f84341d4b23081f69ebe274ad9166b4d301","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-17 17:34:21.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"95850150f7697328712902b4b04d82e5dd7cd0af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 18:40:02.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e8fca4bc8a17453a93e2d623c22cac3b : SUCCESS in 36m 13s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/76e5ffa503984ecdac00e1622bdc4f7f : FAILURE in 34m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/85d1e7d46b3f4eb6b2dde42b4930cf76 : SUCCESS in 39m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ed00e73ec4d84ba69a57c20d87845e07 : SUCCESS in 54m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/62221482cada4bbe816ccfb4ee19e14b : SUCCESS in 34m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/78554402babc4342b996abb5c14d4357 : SUCCESS in 39m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46898bb87a0347faa3018b34b963ce7c : SUCCESS in 6m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12817b0e1d7747d18cb2c99ec9e937b4 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da0f6da328fb4ccab834bfc64fe84167 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7e6b000830d49b18fd15b5d81ae4179 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b7d7ce29fdf4cf49f3667f5f78688b1 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31c165bc7bdc4a72a1d642d6368bc9f7 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33864cfd70574acda0b309a1f7af23a2 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/815b4d37912142e8a718b9ff55595c34 : SUCCESS in 5m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae9a0952452145df8edf1f48ee3f5ed0 : SUCCESS in 45m 39s","accounts_in_message":[],"_revision_number":26},{"id":"41e1184a59a9f9b0bdf9fe89a993188710ea9fbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-17 21:32:44.000000000","message":"Uploaded patch set 27: Commit message was updated.","accounts_in_message":[],"_revision_number":27},{"id":"6d3053fda0a6d898440705ed61cb3f0c5dacbbe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 22:34:44.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9f57f1c6979d4bee812a92eca0025b73 : SUCCESS in 45m 41s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/21d1ae2fb4274a7ea42a4927643f47bd : FAILURE in 43m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0a53625cd137411e8b2a2995179e5b4e : SUCCESS in 32m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7f0344f5455c4a13834c653aae70e25d : SUCCESS in 56m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/069aecd6347348978c248ec6394708b4 : SUCCESS in 40m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/79124d4ebd1148339bd5072609755468 : SUCCESS in 43m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bd60cf54843426cb1a82e11abe23f3f : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6c6411275ac4d988ee3d3ec5a606306 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ef52df3ab5c45e090d48c52cb2e6630 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bbb4ab4e7d541ecabfd58a3a843278d : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea9d6476f339408aad8743c0b367a2f7 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c3b3765e32c4b17b2946d0f51a3aa0c : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8edb8a932e964fc5b1c4a70d08300fcb : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b89d3c7e7f244ac5b4c900c7cb9c4b79 : SUCCESS in 4m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d23053fb384a4505bbb55eaac49d8327 : SUCCESS in 31m 49s","accounts_in_message":[],"_revision_number":27},{"id":"6f8ee0056f9b761c40dd9a5b6357263549b14bc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-18 22:03:34.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"bb46615acf7bbd1378f35b94560e80fe1743150f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-18 23:11:30.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"8bb028d6469efcc9862225fba0e00586f42622cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 23:15:01.000000000","message":"Patch Set 29: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"01c99091fb828460374934829467427223716ed1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-18 23:15:24.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"df73ab1bbfe6767029099fca090cc8868faf2c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 00:09:47.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b2e8a8a227aa4635916d1fce3e9dcdf9 : SUCCESS in 41m 56s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3770b7f7f2574ecb92a63439c10c3767 : SUCCESS in 33m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1f9305a861c14cd0a34e159dd7db51eb : SUCCESS in 36m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/005fbbf6ef6c493ca9c4918c263e7a56 : SUCCESS in 52m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ec618fb32d2c4ffbb4c9545b398bdaae : SUCCESS in 46m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/21aa1294fd134116af227499677e3372 : SUCCESS in 35m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ccc453f16724ceabb99968bc2c562eb : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da23a6e905664756814aa0f28f70fe3f : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e99047c680444e3bcc3346ae93d56e6 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab01efc51b0845db9d97897168dcf888 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc2fff9772f3446eb00575515320739c : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9400f0f0b7f24db2baa887e574ec5852 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a36a527cb964bc7a0f9833db66bf910 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f3b63e261e54c4faecd78de8db71dea : SUCCESS in 4m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6122b140d67d42dd820630b227406ecc : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":29},{"id":"ffaf26e5664fe58aaaf78e01e1344abc960812a8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-22 13:14:54.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"6cae3dfd9a1d479d89aa06d1679a99aa769b2a69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-23 00:13:50.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"7104a9b1a1146cb73227c789ace9886999a6c002","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-23 00:20:54.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"095d80b145e8a7b85dffe6f06ad64509d7a552be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 01:57:43.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/29629d4679fa40ff9f50a469d27a35b5 : SUCCESS in 39m 36s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/8f1199321dd140b49247c85d2b72e1cc : FAILURE in 40m 33s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1ea67ee426ee4959be75c704c142a7da : SUCCESS in 51m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b00c9f69c4f5435bafb45baddd70640e : SUCCESS in 44m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/40fa0a80c2b54f88b479e022c97f7801 : FAILURE in 1h 38m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ce690e1027bf474c84ab797e7cea865e : SUCCESS in 40m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3831f10a6dff4ecf9dd06348cf51e64f : SUCCESS in 48m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/779cf057b8614897ac18660f4f7205b2 : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8269611e0df46f1959898470fe5fb23 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f58cd593614741865e87b7aa05abca : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e229a97e0dae48df9b4a8ed163574618 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90acc9d188514d2fb7ed533bec1d658f : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44f5315380a8429a963fcae4c5e99db6 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e063f835e7494ae7a7fedfa2e8111070 : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a41f9bf7a104a8a9cc71445e8d19899 : SUCCESS in 4m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/87d9095f5b1342c58ffd2b834a4c367c : SUCCESS in 31m 27s","accounts_in_message":[],"_revision_number":30},{"id":"1d90dc84793c39920ba353199091e8fa4e8561ef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-28 07:27:43.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"36aa1d5807fc9b2dbc18060e558e37cd66993a79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-28 08:25:11.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1873090bcf834a7797a62873d1c28aaf : SUCCESS in 46m 13s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2cdab1f1b6a5444bb0ed68644a13e27d : SUCCESS in 41m 24s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5bfcdff3ed9641e1bcd9110d6e93a685 : SUCCESS in 35m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/48fd6a95ad754fa5a3b5f511d1670d0b : SUCCESS in 33m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f43c6eddd0f0431ba47b037f28d43a96 : SUCCESS in 55m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2d507dacda2046f3aaf0e9e142cee2a2 : SUCCESS in 36m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1ff3c1e9e2d247298dfa8d9997ca62ac : SUCCESS in 34m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36008d8be68242039c0d2673ea494c3d : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb78b3fd94e942d5b9dadd49d70b987f : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74076b22f554465dbf76a1ac7e824e0f : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/334ec72cd57d40d0933de7f64925d9c9 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86e3a06e912149ef877cc532f26db73e : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f75ac3462a1c4d7fb3aeef4f0a6468b3 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1d2f617ed1a4753890db414517d9951 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fc06cbc090c462f8ced703199a57e54 : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":30},{"id":"3b03c93abf9241f8f9258f5d8a30c01ae2fd1f46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-29 01:29:53.000000000","message":"Patch Set 31: Patch Set 30 was rebased","accounts_in_message":[],"_revision_number":31},{"id":"aab331950b35b2562dc725c51709bcb69473cf45","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-29 01:30:31.000000000","message":"Patch Set 31: Code-Review+2","accounts_in_message":[],"_revision_number":31},{"id":"f9d1f8dc7b9e89e09fda8867176fae147871eac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 02:26:51.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/02aba8abdbc247078389004d54467d81 : SUCCESS in 37m 38s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/cd76a067f219498388fd4a79d9055829 : SUCCESS in 46m 48s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/95f6445966f64482b602b77507ec9997 : SUCCESS in 34m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f95d2c4ea52841e0bbca4ca770bb41d3 : SUCCESS in 35m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a0af5bdc4b30466c9f501d4827e19395 : SUCCESS in 54m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2539c5dace7446e993832c97d4ec948c : SUCCESS in 34m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8a5c1253e85442c3994546ac9013ba93 : SUCCESS in 34m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f875f4d53be149ecb560141af1400492 : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a2e9f6d9c8f4f4796a87b3afb611758 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb5c0df61c644989340a634c5257ebd : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1df2209133cd49a8a626e0e932aff498 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00f5ca5b16754359954babba9f1029f1 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc86a1b5fb174b34ba25eb1d45c92bdc : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53f10f58a2d54477993447769d048c8c : SUCCESS in 4m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2fc5a253f5034bccacc48433b7b7c91a : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":31},{"id":"4c85f3db0b7a8e4b6d5ace5e36b6718679aaaf9d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-01 22:47:00.000000000","message":"Patch Set 31: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"2b5c567bb60c99b72e8affa5d387db612b7aa92f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-13 05:55:20.000000000","message":"Patch Set 32: Patch Set 31 was rebased","accounts_in_message":[],"_revision_number":32},{"id":"d99a2adbfd652c680d7c27fd24fb8f33c2da856a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-13 05:55:30.000000000","message":"Patch Set 32: Review-Priority+1","accounts_in_message":[],"_revision_number":32},{"id":"be3fed0554217fa8738fecdb8cf7dcad22d0e057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 07:27:29.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/708474b70e9f47cb991c762233720fce : SUCCESS in 37m 39s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/6d0faba438754e208950f72103cb54d1 : SUCCESS in 51m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/412cb8ed9f3c43f09c1d418dfabaa4b7 : FAILURE in 50m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1c5269dd739446489020cb882e28c083 : SUCCESS in 49m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/73f65048afec457f9df6536f057a8c5f : SUCCESS in 1h 09m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/00b40076fc98412fba211aa747d9d943 : SUCCESS in 51m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c28f2028e1774510bb40d6ec68c44994 : SUCCESS in 40m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3371a480a0b94af1912bf548cf6a3594 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56ba67469b9e43fbb76d290e7d5bc97b : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09942489ab1644d189e6aba94d04765a : SUCCESS in 13m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86ea00800ceb49da90e56618a524c7b2 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e2eb16104e24fdc83bd870fb41f7906 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c004500552174800ae2c1e221352dfeb : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cb06a1e994241acb69ee883b3e21364 : SUCCESS in 3m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/48b765a780c04f4498381a521b6795cf : SUCCESS in 34m 15s","accounts_in_message":[],"_revision_number":32},{"id":"9538a5e8b005c0436e227b27d5b925a0d1349c86","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-13 08:06:53.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"b725149555e378181aabb10dd3ff757c96d2169d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 09:11:41.000000000","message":"Patch Set 32:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1b22a1338cfa437cb3e00b6e9089963f : SUCCESS in 36m 43s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/e9e4b391cf8c4179a254c7bba640a05d : SUCCESS in 43m 07s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/17111064ca4141d189e509b8dc1644fa : FAILURE in 37m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/791ffc034c2449b3a6fb57c62f98b542 : SUCCESS in 36m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3cb5b1d134bf4c68ba9f16a9fb8d325c : SUCCESS in 1h 00m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ea1dbaf4ed1e4a84a581123c88cfe7ca : FAILURE in 44m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/64840efdf90240b4aa4d3064e80b4ca5 : SUCCESS in 54m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e824660de71470fbe0d803f5fbb585f : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9e423fb13c740428781150c90c099e3 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c4c81de132e4df5952cad488f1d8bf0 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2d7d7d163274082a45bcd0257e42a21 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fd880bf076a4582bade51b14e34181b : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87c8d22778cc46fc93e11b09e298b417 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f103253c01c44ff6bce09013f4ea8791 : SUCCESS in 5m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c2bb9ebb72d4d60ae2d0a1f5b3220ad : SUCCESS in 40m 12s","accounts_in_message":[],"_revision_number":32},{"id":"8dc91dee4aba31e555f32f2bbac5b1e09012b8d6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-13 17:26:41.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"b9ab4125dbbeb30ab7450c2ca2a9f615c3856e65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 18:29:38.000000000","message":"Patch Set 32:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2c910204a2c943b5b8708d7af2b451ce : FAILURE in 18m 30s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/17a37cdc8f1a47b3a1452623b8b58a81 : SUCCESS in 47m 44s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f3ee929d5953491f858a6f35b0ff9f29 : FAILURE in 31m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e886796081f44bcbad5dede555a7ecf1 : SUCCESS in 34m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/66357c42cf844c70ad0c5f195370dcce : SUCCESS in 58m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2e2414a4266741fc835fa29aed324811 : SUCCESS in 36m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6ca6f5a194a442308b41189f7a80805b : SUCCESS in 42m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54e2e27e6b7342bfa387be5feaef7a7d : SUCCESS in 7m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73facb3267f949739c646842d121ec9b : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4038a767a15d4cbc9c43f42bc7b3b97e : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db5d22acf9e44b7dbdb29c0ba358f1a2 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f314c179aff84b6185f32791ef1adbd8 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9aff9e2a074fcab860875545ed8c94 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ac4bf0b3e4a45638abdda9045f74ada : SUCCESS in 6m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bbe145b838ce41bab254332752946d72 : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":32},{"id":"c50744c6d2dcabf16637f6df2e73772b2743b8d0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-13 19:32:10.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"baaf59ad5d5a27d9e954594205475275501bc061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 20:24:05.000000000","message":"Patch Set 32:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/904485baa50545f8b191b02fb31bbfc2 : SUCCESS in 34m 54s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/f2b0a97e2ded4cafa3d13f665b11de58 : SUCCESS in 46m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b1bb6ef8c98645a0a3f482a15f81a797 : FAILURE in 35m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cb02bed1a1ca4b50b3443d01cf662f6b : SUCCESS in 32m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/050511efd4fb4350bbebe91694e7fd51 : SUCCESS in 49m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1ab326d433004d23b7ef82930c44066e : SUCCESS in 32m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cca02ccbd5fa4b4e9ca27a02711ffceb : FAILURE in 35m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41706adcb8a54310a35feb994ca1233f : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc64166bcca64a038eaa1fca4450ae04 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8938d1de61234a51947daa701cf1f095 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f786191fee946fab76628efd2a68cbd : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ceb525bf66224a968f9ca6aed80cca15 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec2259240ac742ddaed20a666c111151 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05ea53738edb4291bae16608a80dd8ec : SUCCESS in 4m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b45b6a07d2df42749b2d96df4cc9810f : SUCCESS in 42m 08s","accounts_in_message":[],"_revision_number":32},{"id":"ab680e15ee18102883f8aecc9618ae21f6523bb6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-13 20:33:20.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"dcbb77ee6ecd17c7f638137b45afd601886094f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 21:39:03.000000000","message":"Patch Set 32:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/308f2c5a98c8441980063a3294d73328 : SUCCESS in 36m 32s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/8aca73ecd67949d4affd4408cb757e5d : SUCCESS in 1h 04m 23s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/12e8c4d8551f4c2ea9a45d8d8c2a9acb : FAILURE in 30m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/24f6cf44837641b993825178c2d9698b : SUCCESS in 40m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dd8724b1446d4c59baa7c485bd60b48a : SUCCESS in 55m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/efa2de426f6e4ae0944bfca40ece4817 : SUCCESS in 27m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/164e5a23adf149818f60e639c6dc6b72 : SUCCESS in 41m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43abf0a691b84279bc92023fe4e35937 : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84bf275ab4504555a07a76e9137118e4 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb2e4f4b4ae34cb9ae5a4748af2b1b9d : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e1fe88d46d34482bb8c8cee4bae7d04 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/896e957d2c5844a999262fa96fbc9bc1 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7510a9633caf4543bd6359011885ade5 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caa3b2e2ce2e4303909a7dda3925e16f : SUCCESS in 4m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2d691c002ec04a3cac2747f4cd5eee0c : FAILURE in 19m 50s","accounts_in_message":[],"_revision_number":32},{"id":"b9f52756a2167d76a77b78308b6d50bfe5831bfb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-13 22:08:10.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"71c88e372410601492d95d94f134f7fe479c64f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 23:18:13.000000000","message":"Patch Set 32:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aefeaf081a3940ed9224be651afb4eed : SUCCESS in 46m 48s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/3469eda130504e0fada8f318d772ccc5 : SUCCESS in 1h 02m 42s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ce7fb2d8cab54350ab3e910c5865e0e4 : FAILURE in 34m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/36b1cc3311514401938025c4e0af3755 : SUCCESS in 39m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/25bed48bd17b4b3cbed16eda26cc3426 : SUCCESS in 1h 03m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cba21ad2de22403089ef611f254424c2 : SUCCESS in 42m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/89b8320e23524d3b9c31b14224c2e369 : SUCCESS in 43m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1360410346c4d9aadda54fa65bb20ec : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a76570daed1e4dfe934299d2fb9f3bf3 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0019b6134694cdda6a1e5b17a1e845a : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5f70d09ece849d49a6cb6518fe9052b : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2acdf73f1a1d45e7847c044a14ee009e : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/595e10787de7483e998aee3691b28da1 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a05e26f208a430eab1730fa9b010265 : SUCCESS in 4m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/faf9779ec2074786a6dae653d1bff161 : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":32},{"id":"6a08dadd890ca629961ca209b79aa31a0301ffdc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-13 23:53:44.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"3b9c1800be8deb0009e6f9ad6dfb22d18c0169e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 00:58:07.000000000","message":"Patch Set 32:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1498b57e71fe4119bf0cd0e29c01ae58 : SUCCESS in 37m 02s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/b1393bfac027453bacf5a2f3582290db : SUCCESS in 43m 42s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/eb7391e6c2ff4b9b8c2c36f78241fac3 : FAILURE in 40m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c610eeb4c16a4a06a931f00a19b333ce : SUCCESS in 40m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/408a3570c7ee48088c9164dd279ff5a3 : SUCCESS in 59m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/79503dea8abc474b99869e343609ad47 : SUCCESS in 41m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5801841826e5495bb30548b4a3606492 : SUCCESS in 40m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7766608f56694b24ae5214daec822ff4 : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e2422a325b418ba03cc99b97e12566 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c94f5d5c092f41fe97fe67f9fbead928 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e88acee38592489f85132123d4015a3c : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c8e0d2e96cc45349d6f5c0f2fcb580b : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa6b389ffbdc46e2891869ae0a2aca37 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8465f32ea9d4290a9e5b033892a0cd7 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5374ee4d92ee47428d27bab27c0d6374 : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":32},{"id":"4ec392f6ecc63a55b10d7986c05191a6d14b3568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-14 01:02:50.000000000","message":"Patch Set 33: Commit message was updated.","accounts_in_message":[],"_revision_number":33},{"id":"87fdf74daf75aa9005d67e29346645c8a69b0b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 01:58:29.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9fcdad69a89a4b8d89683523ef9e6a09 : SUCCESS in 44m 24s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/7d1cc4e89974481296a05b16ed456fe1 : SUCCESS in 44m 17s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c18aef9d7fe443b7bfc568790e916cd7 : SUCCESS in 31m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/00594d30a4844a8a9a6285d8acf57059 : FAILURE in 26m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/914d2277b63f4c37b073f9eca2150187 : SUCCESS in 53m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/980c62a037104d9ab09ddb55b25edc2c : SUCCESS in 40m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/730ef7e27c8540438895a69a9dd1866d : SUCCESS in 30m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1794c98fa027457d8c94140ea6fa90f4 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3abfcc9c2cc642eeb27992bcebb3d22e : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52fc35785344ab68e0ea9729e91a43e : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac358ede93048a9b327850d1b4a56a4 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e238ee3dd1ae45f0be94cdb0da66f7d0 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15017d0c1fb649a389b04d04bff05abe : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfee9676cbb14aec89f30ed16c939c13 : SUCCESS in 4m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eaf04f67b6eb4941b58308e564db26c9 : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":33},{"id":"1729395d43ddd749989e0689ddd75be1c0316e1c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-14 01:59:59.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"9b405188b9b1980e3fe65695f3e44cefb8ff598d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 03:05:30.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4c1077f11f9c4f79ab13a630e29ad1ae : SUCCESS in 36m 38s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/74cf15d53e8841cbada3216b0a4d769d : SUCCESS in 55m 28s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7fc1cb510cb04751acc1a36aba6a02d7 : SUCCESS in 39m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/64bc1096116643dfbbf6e504b28cce76 : SUCCESS in 36m 45s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/42779f92f3e9481e9359a668969c8ed0 : SUCCESS in 58m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cb0afee87981480da9104265b1370b57 : SUCCESS in 43m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/be9f754667884141a7ec6557d4f259ca : SUCCESS in 44m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00e8f59ed675461790d75a3eea42be3c : SUCCESS in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/878a6be3de494a5189c460a95e3c803a : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/145538f081a4448ba72c97c6707bb0c8 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/285c856ca4be40359885bd67d8559788 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc1a9fcdd4a84ebda94b82be2f8d1ae1 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd8c86d1564247c5b71968e2cfd6ee40 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6f61dd98cc24bc390df4cff808aa32a : SUCCESS in 4m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7d02acfd59c94167899739d2b29c8c0c : SUCCESS in 44m 33s","accounts_in_message":[],"_revision_number":33},{"id":"aa2b1e211d736165332b256a9ef10feb62134924","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-16 05:49:53.000000000","message":"Patch Set 33: Code-Review+2","accounts_in_message":[],"_revision_number":33},{"id":"d8c3b17033a8b5b0a680690c7df0bfee6f2fb9b3","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2021-12-20 14:53:11.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"4a7da068befa8de06612a8583106cce749296bcd","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2021-12-21 16:39:44.000000000","message":"Patch Set 33: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"79358b03f8c495800152f5a163d1e8dbd491936f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-21 22:39:31.000000000","message":"Patch Set 33: Workflow+1","accounts_in_message":[],"_revision_number":33},{"id":"699bf233490f396a3959ffd50afcd8e94bb27b79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 22:40:29.000000000","message":"Patch Set 33: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":33},{"id":"57cfe6c4ab9f4ddb67580b8faf67bcb10505d9f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 23:29:08.000000000","message":"Patch Set 33: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cb3ff0f7ed5243bfa1eb89651815889e : SUCCESS in 29m 53s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4e52949c008c4663bf3fac0f9348cc20 : SUCCESS in 30m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c034d7923ae0495387e3a7039ca6dfaa : SUCCESS in 30m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d6c46fde853645b58b08f053612540fc : SUCCESS in 48m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/126d15566fd74848b97726d272ef7ce3 : SUCCESS in 33m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/38d27519210044a9b112f3158280ff76 : SUCCESS in 32m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a60da98107e940c7a042b5d516220e49 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ee3127b8934e76b62972083035339d : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b25e09f167849a5aca3534931e01015 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/680e2df27c01435381373a98b3567932 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37a74daa16424b12b9f2c4106ac37330 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4b6674ad173491f9d65faf120243f95 : SUCCESS in 5m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/274bf9bc6b4f482eb7e585ebd5efe575 : SUCCESS in 27m 06s","accounts_in_message":[],"_revision_number":33},{"id":"963e36612ab0d77376cb28828095dcfdd9c7e3a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 23:29:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":33},{"id":"d37c9444ac4ab8a12d4bf74602e5d890e55c3dad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 23:34:03.000000000","message":"Patch Set 33:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb986b1f3074c63bc4187331aae216b : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6338c06da708434091da1db3c69a2c19 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":33}],"current_revision_number":33,"current_revision":"5f87d207b448ec4e24988af5671c3b3593b13804","revisions":{"6e730b2adca4fe612719e129089452da23f9306a":{"kind":"REWORK","_number":1,"created":"2021-07-13 21:25:47.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/1"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:25:46.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e730b2adca4fe612719e129089452da23f9306a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e730b2adca4fe612719e129089452da23f9306a"}]},"branch":"refs/heads/master"},"e9468b31ad80d39787046bb77f21223cf850d769":{"kind":"REWORK","_number":2,"created":"2021-07-13 22:40:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/2"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 22:40:14.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e9468b31ad80d39787046bb77f21223cf850d769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e9468b31ad80d39787046bb77f21223cf850d769"}]},"branch":"refs/heads/master"},"54c107f9d05a25f9c6f93f4c16ef62a66b20fee8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-15 22:09:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/3"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-15 22:09:28.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54c107f9d05a25f9c6f93f4c16ef62a66b20fee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54c107f9d05a25f9c6f93f4c16ef62a66b20fee8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7083759d2dd28b8b3ec32900d329533973fecae6":{"kind":"REWORK","_number":4,"created":"2021-07-15 23:10:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/4"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-15 23:10:14.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7083759d2dd28b8b3ec32900d329533973fecae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7083759d2dd28b8b3ec32900d329533973fecae6"}]},"branch":"refs/heads/master"},"b140a6ed47f770da4a848db739bb7dc27d9e0117":{"kind":"REWORK","_number":5,"created":"2021-07-16 23:25:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/5"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-16 23:25:15.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b140a6ed47f770da4a848db739bb7dc27d9e0117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b140a6ed47f770da4a848db739bb7dc27d9e0117"}]},"branch":"refs/heads/master"},"9e35690816e9be833e3c5a7ee2bd7a335bf60dd7":{"kind":"REWORK","_number":6,"created":"2021-07-19 23:46:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/6"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-19 23:46:13.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e35690816e9be833e3c5a7ee2bd7a335bf60dd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e35690816e9be833e3c5a7ee2bd7a335bf60dd7"}]},"branch":"refs/heads/master"},"33b72d451d8d4dd6507524a3cbadb5eb9be7bd4d":{"kind":"REWORK","_number":7,"created":"2021-07-21 20:57:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/7"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-21 20:57:11.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33b72d451d8d4dd6507524a3cbadb5eb9be7bd4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33b72d451d8d4dd6507524a3cbadb5eb9be7bd4d"}]},"branch":"refs/heads/master"},"d29b497f0413fd8aafe5d33a5790338102a19708":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-07-21 21:58:31.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/8"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-21 21:58:09.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d29b497f0413fd8aafe5d33a5790338102a19708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d29b497f0413fd8aafe5d33a5790338102a19708"}]},"branch":"refs/heads/master"},"a0d8b27c8511c3e6d940c914a030804b8e4ebab0":{"kind":"REWORK","_number":9,"created":"2021-07-21 22:17:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/9"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-21 22:17:41.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0d8b27c8511c3e6d940c914a030804b8e4ebab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0d8b27c8511c3e6d940c914a030804b8e4ebab0"}]},"branch":"refs/heads/master"},"70e06a0da0a4a2a211f7ea16a6930b127d10152c":{"kind":"REWORK","_number":10,"created":"2021-07-23 22:08:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/10"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-23 22:03:45.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70e06a0da0a4a2a211f7ea16a6930b127d10152c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70e06a0da0a4a2a211f7ea16a6930b127d10152c"}]},"branch":"refs/heads/master"},"348545f123122d7a0574503531c2fa82ce3ce5ed":{"kind":"REWORK","_number":11,"created":"2021-07-26 19:45:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/11"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-26 19:41:05.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/348545f123122d7a0574503531c2fa82ce3ce5ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/348545f123122d7a0574503531c2fa82ce3ce5ed"}]},"branch":"refs/heads/master"},"55b346dce43df7bd1c83ec47d82c2152f45ec8e3":{"kind":"REWORK","_number":12,"created":"2021-07-27 22:43:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/12"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-27 22:43:16.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/55b346dce43df7bd1c83ec47d82c2152f45ec8e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/55b346dce43df7bd1c83ec47d82c2152f45ec8e3"}]},"branch":"refs/heads/master"},"58c210a215285033daa8a9028d132f4291da9d41":{"kind":"REWORK","_number":13,"created":"2021-07-29 18:07:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/13"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-29 18:07:19.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58c210a215285033daa8a9028d132f4291da9d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58c210a215285033daa8a9028d132f4291da9d41"}]},"branch":"refs/heads/master"},"e7c30d6239d897e8e0fb97fcc0ea6fcfb32ad11f":{"kind":"REWORK","_number":14,"created":"2021-08-03 23:49:28.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/14"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-08-03 23:49:18.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7c30d6239d897e8e0fb97fcc0ea6fcfb32ad11f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7c30d6239d897e8e0fb97fcc0ea6fcfb32ad11f"}]},"branch":"refs/heads/master"},"99046e331e4d390b00f0b82e42fe2d39578ac2f9":{"kind":"REWORK","_number":15,"created":"2021-08-05 16:22:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/15"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-08-05 16:22:34.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99046e331e4d390b00f0b82e42fe2d39578ac2f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99046e331e4d390b00f0b82e42fe2d39578ac2f9"}]},"branch":"refs/heads/master"},"670f925297e40bce070a8beedb9580ae1dca5d95":{"kind":"REWORK","_number":16,"created":"2021-08-05 22:09:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/16"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-08-05 22:09:08.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/670f925297e40bce070a8beedb9580ae1dca5d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/670f925297e40bce070a8beedb9580ae1dca5d95"}]},"branch":"refs/heads/master"},"5dae272de0434c7bd64fefee7138fe152e5ed3ee":{"kind":"REWORK","_number":17,"created":"2021-08-06 21:51:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/17"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-08-06 21:51:52.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5dae272de0434c7bd64fefee7138fe152e5ed3ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5dae272de0434c7bd64fefee7138fe152e5ed3ee"}]},"branch":"refs/heads/master"},"83302032fabe01fe4b970cbc81facc28ae23b0a8":{"kind":"REWORK","_number":18,"created":"2021-08-11 23:11:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/18"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-08-11 23:11:22.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83302032fabe01fe4b970cbc81facc28ae23b0a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83302032fabe01fe4b970cbc81facc28ae23b0a8"}]},"branch":"refs/heads/master"},"568c51f06f2cf8bd1631a08eb4175e26a1ddbb23":{"kind":"REWORK","_number":19,"created":"2021-08-12 20:05:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/19"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-08-12 20:05:00.000000000","tz":0},"subject":"Add a job to test with keystone scoped tokens","message":"Add a job to test with keystone scoped tokens\n\nThis patch adds a job to test running with keystone scoped tokens\nenabled in the Designate API.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/568c51f06f2cf8bd1631a08eb4175e26a1ddbb23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/568c51f06f2cf8bd1631a08eb4175e26a1ddbb23"}]},"branch":"refs/heads/master"},"3c1a103b320f3f33ce1b01de6a762323852fd25d":{"kind":"REWORK","_number":20,"created":"2021-08-16 20:28:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/20"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-08-16 20:27:12.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c1a103b320f3f33ce1b01de6a762323852fd25d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c1a103b320f3f33ce1b01de6a762323852fd25d"}]},"branch":"refs/heads/master"},"b6f2dddc2a549c5aad32a16e412804fa2ba6ab64":{"kind":"REWORK","_number":21,"created":"2021-08-19 21:55:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/21"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-08-19 21:53:56.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b6f2dddc2a549c5aad32a16e412804fa2ba6ab64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b6f2dddc2a549c5aad32a16e412804fa2ba6ab64"}]},"branch":"refs/heads/master"},"bced5b606759604136d772451e05c628163dcce9":{"kind":"REWORK","_number":22,"created":"2021-08-20 00:35:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/22"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-08-20 00:34:56.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bced5b606759604136d772451e05c628163dcce9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bced5b606759604136d772451e05c628163dcce9"}]},"branch":"refs/heads/master"},"01476840ff5d8e3fce327499750f3e151849b918":{"kind":"REWORK","_number":23,"created":"2021-09-01 20:55:01.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/23"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-09-01 20:54:59.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/01476840ff5d8e3fce327499750f3e151849b918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/01476840ff5d8e3fce327499750f3e151849b918"}]},"branch":"refs/heads/master"},"8287055155e0b670cf8720478c7c4915de1b1484":{"kind":"REWORK","_number":24,"created":"2021-09-03 18:13:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/24"}}},"commit":{"parents":[{"commit":"6fc04e72ec1fd039290173197dc4921df2f3d1c0","subject":"Merge \"Replace md5 for fips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fc04e72ec1fd039290173197dc4921df2f3d1c0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-09-03 18:13:32.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8287055155e0b670cf8720478c7c4915de1b1484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8287055155e0b670cf8720478c7c4915de1b1484"}]},"branch":"refs/heads/master"},"94de6b35a16fe4c37e8137d2e175668a5667368d":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2021-09-22 21:11:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/25"}}},"commit":{"parents":[{"commit":"c3f084751006a2fe7562f137930542c4759d6fd9","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3f084751006a2fe7562f137930542c4759d6fd9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-09-22 21:11:09.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.context/+/800852\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/800998\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/801662\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/94de6b35a16fe4c37e8137d2e175668a5667368d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/94de6b35a16fe4c37e8137d2e175668a5667368d"}]},"branch":"refs/heads/master","description":"Rebase"},"0bacbdd32a19b677f5ccf75e897f210539706cbc":{"kind":"REWORK","_number":26,"created":"2021-11-17 17:34:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/26"}}},"commit":{"parents":[{"commit":"c6d15d3d5e38e4921e4f9f7121306ee6311c34f9","subject":"Merge \"Deprecate parameters of SSLMiddleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6d15d3d5e38e4921e4f9f7121306ee6311c34f9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-17 17:34:19.000000000","tz":0},"subject":"Fix support for scoped tokens and default roles","message":"Fix support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0bacbdd32a19b677f5ccf75e897f210539706cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0bacbdd32a19b677f5ccf75e897f210539706cbc"}]},"branch":"refs/heads/master"},"f5d94075fc9f9a9cc8130af1b74ea959829c6bcf":{"kind":"NO_CODE_CHANGE","_number":27,"created":"2021-11-17 21:32:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/27"}}},"commit":{"parents":[{"commit":"c6d15d3d5e38e4921e4f9f7121306ee6311c34f9","subject":"Merge \"Deprecate parameters of SSLMiddleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6d15d3d5e38e4921e4f9f7121306ee6311c34f9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-17 21:32:05.000000000","tz":0},"subject":"Fix support for scoped tokens and default roles","message":"Fix support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/818293\n\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5d94075fc9f9a9cc8130af1b74ea959829c6bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5d94075fc9f9a9cc8130af1b74ea959829c6bcf"}]},"branch":"refs/heads/master"},"9c74df44cacc02df77c0de95e04d49f0c5d3359a":{"kind":"REWORK","_number":28,"created":"2021-11-18 22:03:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/28"}}},"commit":{"parents":[{"commit":"0e16095a80c28c86dcb0a4fa2be4370b8ee9116b","subject":"Enable H306 check and fix import ordering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e16095a80c28c86dcb0a4fa2be4370b8ee9116b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-18 22:03:32.000000000","tz":0},"subject":"Fix support for scoped tokens and default roles","message":"Fix support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/818293\n\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9c74df44cacc02df77c0de95e04d49f0c5d3359a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9c74df44cacc02df77c0de95e04d49f0c5d3359a"}]},"branch":"refs/heads/master"},"c1bdcbcfeab75f7c205b4e34e515dc256971bf08":{"kind":"REWORK","_number":29,"created":"2021-11-18 23:11:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/29"}}},"commit":{"parents":[{"commit":"0e16095a80c28c86dcb0a4fa2be4370b8ee9116b","subject":"Enable H306 check and fix import ordering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e16095a80c28c86dcb0a4fa2be4370b8ee9116b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-18 23:11:24.000000000","tz":0},"subject":"Fix support for scoped tokens and default roles","message":"Fix support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/818293\n\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1bdcbcfeab75f7c205b4e34e515dc256971bf08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1bdcbcfeab75f7c205b4e34e515dc256971bf08"}]},"branch":"refs/heads/master"},"7c42a2a0be159c5e23adaa06f05557b2f800ff79":{"kind":"REWORK","_number":30,"created":"2021-11-23 00:13:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/30"}}},"commit":{"parents":[{"commit":"bee8dc032e32536d0c7a24e65e9883a7a196b068","subject":"Merge \"Add fips jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bee8dc032e32536d0c7a24e65e9883a7a196b068"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-23 00:13:48.000000000","tz":0},"subject":"Fix support for scoped tokens and default roles","message":"Fix support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/818293\n\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c42a2a0be159c5e23adaa06f05557b2f800ff79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c42a2a0be159c5e23adaa06f05557b2f800ff79"}]},"branch":"refs/heads/master"},"a5383f8dafcdd082a3bf44ffaa96cea38867f317":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2021-11-29 01:29:53.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/800727/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/31"}}},"commit":{"parents":[{"commit":"acd930d3429aa30bea449f5fc7246581b0d2eb25","subject":"Fix dns.query.tcp/udp not always handling ipv6 properly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/acd930d3429aa30bea449f5fc7246581b0d2eb25"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-11-29 01:29:53.000000000","tz":0},"subject":"Fix support for scoped tokens and default roles","message":"Fix support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/818293\n\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a5383f8dafcdd082a3bf44ffaa96cea38867f317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a5383f8dafcdd082a3bf44ffaa96cea38867f317"}]},"branch":"refs/heads/master","description":"Rebase"},"b4ba2f5c80b4f1da6322d8c93329c0672ffbac1d":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2021-12-13 05:55:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/800727/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/32"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-13 05:55:20.000000000","tz":0},"subject":"Fix support for scoped tokens and default roles","message":"Fix support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/818293\n\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4ba2f5c80b4f1da6322d8c93329c0672ffbac1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4ba2f5c80b4f1da6322d8c93329c0672ffbac1d"}]},"branch":"refs/heads/master","description":"Rebase"},"5f87d207b448ec4e24988af5671c3b3593b13804":{"kind":"NO_CODE_CHANGE","_number":33,"created":"2021-12-14 01:02:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/800727/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/800727/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/800727/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/800727/33"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-13 21:24:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-12-14 01:02:50.000000000","tz":0},"subject":"Fix support for scoped tokens and default roles","message":"Fix support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Designate API.\n\nIt also migrates to using project_id in the context objects instead of\nthe deprecated tenant_id.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/821632\n\nChange-Id: I43bb76dc4dc1d167d86fd5ea139a50f95f3b0b4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
