)]}'
{"id":"openstack%2Fdesignate~802815","triplet_id":"openstack%2Fdesignate~master~Idbfec1adca2024cd5f352017a7c9319dcec65d42","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"Idbfec1adca2024cd5f352017a7c9319dcec65d42","subject":"Fix grenade upgrade API check URL path","status":"MERGED","created":"2021-07-28 23:45:37.000000000","updated":"2021-08-02 11:29:59.000000000","submitted":"2021-08-02 11:28:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802815","meta_rev_id":"55fff9a5a7d854d466f14e315b7dc5aaa1e18466","_number":802815,"virtual_id_number":802815,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-02 11:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-29 05:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-29 05:37:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2021-07-29 03:09:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 01:06:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 05:37:19.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"c9cdc3869883994684b317e11e8db1ef473fcf9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-28 23:45:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ff4493935e221423833a48449856acd342eead5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 01:06:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/93bf5864b0ee460b8db7c03299180faa : SUCCESS in 36m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3054977b234b4ef8a546ef689e45f32e : SUCCESS in 39m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/549953bb9d4e4c3abb5434977b4eb728 : SUCCESS in 1h 17m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ca53c80c8dfa4a65980c91ad49469952 : SUCCESS in 40m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/898936cedae347dd922f0ff22c8a07aa : SUCCESS in 44m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddbe13c7d1a140f0b883b16b09645cb0 : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9e66343aacb4b0ba7c3bef6bea42b74 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f165ba3a4e40411f83b0ec0606a279d9 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60b79e2d909746519876b2a0c74ebb1f : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cfd7a6c248b42d79a30638ac0351181 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e34385fe4c1546579b82c6b2cc05c6ad : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22145d2923c94d5294419df430a7932c : SUCCESS in 7m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a5af499e2f22476bbf4a846764dc7d9a : SUCCESS in 26m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a8faa26fbc3d2bb4a3625e4a9bf35ab4a5348b81","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-29 03:09:25.000000000","message":"Patch Set 1: Review-Priority+2\n\nGrenade gate fix for cinder v2 issue in horizon.","accounts_in_message":[],"_revision_number":1},{"id":"22968b8b095e58c16b3bc9e0d37044756f18a023","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-07-29 05:37:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83f5becfd758144457f3c90d63f8ac052373b041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 10:16:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"afb59068b70fe6667f5dcabd340e4e90468601d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 11:28:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1dc13b529296491aa7e1354d2abdd1c1 : SUCCESS in 44m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7472aebfc9c749c9907193e4b5417df2 : SUCCESS in 38m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d94dfd0bcaf9489bb2da8e600d30e922 : SUCCESS in 1h 06m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4cc7f9515e5e487fb6bbdec2d8dc93f6 : SUCCESS in 38m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8c33205dade64bb5a21f4fcd6fddf805 : SUCCESS in 40m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a6781de7b2a47dc91b370eb1a7664bb : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35abce73dfc14a27831d318d70b4a605 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24a610a6496a4fa2a088e541ed9afcaf : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6cff05dce58461690f4fab7a424bb32 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946a88f862074e95bfc0a45f8cd093b0 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"475619cad4ddc6be72abb7b63266d57bcc5e29cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 11:28:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"55fff9a5a7d854d466f14e315b7dc5aaa1e18466","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 11:29:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4120dc806844f75a24560e845f60bb4 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0e83084c2b67112cd0dc9d56e233c98e8538d15","revisions":{"b0e83084c2b67112cd0dc9d56e233c98e8538d15":{"kind":"REWORK","_number":1,"created":"2021-07-28 23:45:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/15/802815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/15/802815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/802815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/802815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/802815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/15/802815/1"}}},"commit":{"parents":[{"commit":"beb75cc529771fb6a1859abb3f19ad1ad138bc41","subject":"Fix doc building for sphinx 4.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/beb75cc529771fb6a1859abb3f19ad1ad138bc41"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-28 23:43:23.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-28 23:45:28.000000000","tz":0},"subject":"Fix grenade upgrade API check URL path","message":"Fix grenade upgrade API check URL path\n\nPreviously, the grenade API started check during the upgrade was\nchecking if the root of the web server was functioning. In this job\nthat is testing the horizon endpoint and not the designate API.\nThis patch fixes this to have the check run against the designate\nendpoint under apache. This prevents false failures when horizon\nhas an issue.\n\nChange-Id: Idbfec1adca2024cd5f352017a7c9319dcec65d42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b0e83084c2b67112cd0dc9d56e233c98e8538d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b0e83084c2b67112cd0dc9d56e233c98e8538d15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
